)]}'
{"id":"openstack%2Fmistral~674824","triplet_id":"openstack%2Fmistral~master~I2a6b99aeb3db3b45e0a4c76f1ea4af3b798de73d","project":"openstack/mistral","branch":"master","topic":"eyalb/raw_string","hashtags":[],"change_id":"I2a6b99aeb3db3b45e0a4c76f1ea4af3b798de73d","subject":"Use raw strings in all pattern matching strings","status":"MERGED","created":"2019-08-06 14:58:27.000000000","updated":"2019-08-07 09:01:34.000000000","submitted":"2019-08-07 09:01:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674824-1565168494737-c5a61b3b","meta_rev_id":"ecad2c45ae0560fa336db71e87d3fbfaafce702e","_number":674824,"virtual_id_number":674824,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-08-07 09:01:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-08-07 03:47:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-08-07 06:45:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 06:45:20.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-08-07 09:01:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85332bdbb9fc2060f193474d78e67a6158a00b9a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-06 14:58:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"804c68dca794ea0c7ac591615b7b911784862993","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 17:47:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/24/674824/1/check/openstack-tox-pep8/e1ad6bc/ : SUCCESS in 5m 00s\n- openstack-tox-py27 https://logs.opendev.org/24/674824/1/check/openstack-tox-py27/71edd5c/ : SUCCESS in 8m 37s\n- openstack-tox-py36 https://logs.opendev.org/24/674824/1/check/openstack-tox-py36/e519fa9/ : SUCCESS in 8m 33s\n- openstack-tox-py37 https://logs.opendev.org/24/674824/1/check/openstack-tox-py37/06bdf53/ : SUCCESS in 8m 10s\n- openstack-tox-docs https://logs.opendev.org/24/674824/1/check/openstack-tox-docs/7318397/html/ : SUCCESS in 4m 31s\n- openstack-tox-cover https://logs.opendev.org/24/674824/1/check/openstack-tox-cover/927bb02/cover/ : SUCCESS in 8m 24s (non-voting)\n- mistral-devstack https://logs.opendev.org/24/674824/1/check/mistral-devstack/a135384/ : SUCCESS in 43m 56s\n- mistral-devstack-non-apache https://logs.opendev.org/24/674824/1/check/mistral-devstack-non-apache/8f15e5c/ : SUCCESS in 50m 00s\n- mistral-devstack-kombu https://logs.opendev.org/24/674824/1/check/mistral-devstack-kombu/874c7ea/ : SUCCESS in 53m 29s\n- mistral-tox-unit-mysql https://logs.opendev.org/24/674824/1/check/mistral-tox-unit-mysql/8adca08/ : SUCCESS in 27m 54s\n- mistral-tox-unit-postgresql https://logs.opendev.org/24/674824/1/check/mistral-tox-unit-postgresql/3bea532/ : SUCCESS in 26m 39s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/24/674824/1/check/tripleo-ci-centos-7-scenario003-standalone/7de2ed9/ : SUCCESS in 1h 20m 42s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/24/674824/1/check/tripleo-ci-centos-7-containers-multinode/d069a6c/ : SUCCESS in 2h 38m 49s\n- mistral-rally-task https://logs.opendev.org/24/674824/1/check/mistral-rally-task/b7ec0b0/ : SUCCESS in 1h 05m 32s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/24/674824/1/check/openstack-tox-lower-constraints/4bb0b0d/ : SUCCESS in 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6deea6b32719643dd99d734d7466a7248ffe4341","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-07 03:47:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f66777602ec5d83f826a47fc668d3bd230d1d090","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-07 06:45:20.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a3f95327ca9164307ec6867d5d1f9c87d4bc8a89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 06:45:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42c97016443ec21b88585564d4d9fb3427370f90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 09:01:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/24/674824/1/gate/openstack-tox-pep8/8157c31/ : SUCCESS in 5m 08s\n- openstack-tox-py27 https://logs.opendev.org/24/674824/1/gate/openstack-tox-py27/bde3899/ : SUCCESS in 7m 57s\n- openstack-tox-py36 https://logs.opendev.org/24/674824/1/gate/openstack-tox-py36/f1b0825/ : SUCCESS in 8m 12s\n- openstack-tox-py37 https://logs.opendev.org/24/674824/1/gate/openstack-tox-py37/293b620/ : SUCCESS in 8m 55s\n- openstack-tox-docs https://logs.opendev.org/24/674824/1/gate/openstack-tox-docs/febf0fe/html/ : SUCCESS in 3m 53s\n- mistral-devstack https://logs.opendev.org/24/674824/1/gate/mistral-devstack/99e1c42/ : SUCCESS in 44m 51s\n- mistral-devstack-non-apache https://logs.opendev.org/24/674824/1/gate/mistral-devstack-non-apache/6da407f/ : SUCCESS in 42m 13s\n- mistral-tox-unit-mysql https://logs.opendev.org/24/674824/1/gate/mistral-tox-unit-mysql/59422d6/ : SUCCESS in 29m 31s\n- mistral-tox-unit-postgresql https://logs.opendev.org/24/674824/1/gate/mistral-tox-unit-postgresql/7e97ca0/ : SUCCESS in 28m 03s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/24/674824/1/gate/tripleo-ci-centos-7-scenario003-standalone/88cc4ca/ : SUCCESS in 1h 26m 39s\n- mistral-devstack-kombu https://logs.opendev.org/24/674824/1/gate/mistral-devstack-kombu/3b8dd85/ : SUCCESS in 41m 49s\n- openstack-tox-lower-constraints https://logs.opendev.org/24/674824/1/gate/openstack-tox-lower-constraints/bb3d8d7/ : SUCCESS in 11m 29s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/24/674824/1/gate/tripleo-ci-centos-7-undercloud-containers/ca9ffe6/ : SUCCESS in 1h 41m 44s","accounts_in_message":[],"_revision_number":1},{"id":"dd1dcb8f06a48222f0a74d4ef260860b8e643743","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 09:01:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b576aec7c2ecc56d7c105b80cee3e2ad3dc66e4","revisions":{"9b576aec7c2ecc56d7c105b80cee3e2ad3dc66e4":{"kind":"REWORK","_number":1,"created":"2019-08-06 14:58:27.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/24/674824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/674824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/674824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/674824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/674824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/674824/1"}}},"commit":{"parents":[{"commit":"f949901b8c63bc7ebb69237436ed48ec288f477e","subject":"Merge \"Improve error message when sync actions timeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f949901b8c63bc7ebb69237436ed48ec288f477e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-06 14:12:49.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-06 14:57:13.000000000","tz":180},"subject":"Use raw strings in all pattern matching strings","message":"Use raw strings in all pattern matching strings\n\nThis removes some of the deprecation warnings on\nunsupported escape sequence for python3.6 and up\n\nChange-Id: I2a6b99aeb3db3b45e0a4c76f1ea4af3b798de73d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9b576aec7c2ecc56d7c105b80cee3e2ad3dc66e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9b576aec7c2ecc56d7c105b80cee3e2ad3dc66e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
