)]}'
{"id":"openstack%2Fmistral~676628","triplet_id":"openstack%2Fmistral~master~Ife48d9e464114fd60a08707d8f32f847a6f623c9","project":"openstack/mistral","branch":"master","topic":"bp/mistral-redesign-scheduler","hashtags":[],"change_id":"Ife48d9e464114fd60a08707d8f32f847a6f623c9","subject":"Improve new scheduler","status":"MERGED","created":"2019-08-15 08:11:07.000000000","updated":"2019-08-28 20:25:28.000000000","submitted":"2019-08-28 20:24:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676628-1567023847212-91f773fe","meta_rev_id":"e2780433864c12ed9e876771c5d1d0882f13000e","_number":676628,"virtual_id_number":676628,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-08-28 20:24:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-08-26 08:40:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":1,"date":"2019-08-25 06:50:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":1,"date":"2019-08-19 13:24:05.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-28 09:24:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-08-19 07:46:11.000000000","_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"date":"2019-08-19 07:46:11.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2019-08-19 07:46:11.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2019-08-19 07:46:11.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-08-28 08:57:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 07:46:11.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-08-19 07:46:11.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-08-19 07:46:11.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-08-19 07:46:11.000000000","updated_by":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"reviewer":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"state":"REVIEWER"},{"updated":"2019-08-19 13:24:05.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-08-25 06:50:34.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2019-08-26 08:40:25.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-08-28 09:24:52.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-08-28 20:24:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"799e3dc3575af23e219bce4676029f4939c3ed48","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-15 08:11:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4169f4cc7e7a4fc6cd6fab3289e8b9071ac66e3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-15 10:18:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3ac0e089e1ca62f2d7bcb688e9e8c310712ad2a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 13:02:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bd4f352fa5541c383cdccbae274ca65 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2c181db7fde4c55a82943a74d0c95f5 : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55f48b49e55e48a789ebdc1ecd491a3f : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6b487d9ff58474d80de7212f2c5dd74 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aea531a4c945417894fb2245f5a8fbd0 : SUCCESS in 4m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f345537726b84dbd99bd221761873c5f : SUCCESS in 9m 13s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/81205a6215c0426ca26780b0657de3b5 : SUCCESS in 44m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/73bf2f6388174e919ba24d9ce9af0b4e : SUCCESS in 45m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ff40de33c289406593b2a77a767be37c : SUCCESS in 45m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/053b585a5127433285daec5044cf5cc8 : SUCCESS in 28m 05s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/02823d1882294840b733a8bec3c6df62 : SUCCESS in 26m 18s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/99a9df92cb51443c9b9e97ae004b743a : SUCCESS in 1h 13m 00s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b5e46e02978f46d5b6b8267865b383e3 : SUCCESS in 2h 22m 53s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c241ef802b2c41e88baa5f3f12fa0278 : SUCCESS in 1h 07m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ca25a2b265b4b9fa561d403aa12ed5d : SUCCESS in 9m 46s","accounts_in_message":[],"_revision_number":2},{"id":"20d4d7f3531c0af326562b5d28c4caf2c20d0bc8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-16 05:15:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ae46ca8483065c059d1c0bb88017005477fa3c75","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-16 05:16:04.000000000","message":"Patch Set 3: Workflow-1\n\nSwitched to the default scheduler temporarily to see if all tests pass with it.","accounts_in_message":[],"_revision_number":3},{"id":"6373b77d9edf35958878a3bfa4b5bb3f238e4850","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-16 06:39:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"23e6eea8f1ddf786b1092f2cc90f355fded5eb44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:18:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad4416842ad477bb30fe8daf8860eb5 : SUCCESS in 6m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/807a2f21894d488887ea04f0665a1500 : SUCCESS in 10m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d86e4e3758a048618d122bcb8b18dc8e : SUCCESS in 9m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d3b14935593496fb0aebf07ff2a615e : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14d1479ddfdc4ee7857cd480a1cb0d12 : SUCCESS in 5m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52b5704bb36544f894c79ed6e685de38 : SUCCESS in 9m 09s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/78d1c9c21a324ded8e7138ce57f59a71 : SUCCESS in 47m 55s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c01cc2662e18473eb6350abcab903f5a : SUCCESS in 44m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a7eb0d2a79e147db87df6ce0b95dc269 : SUCCESS in 54m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9421f7740a7f461498198b45851a11a5 : SUCCESS in 29m 01s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/38941bd021d14c6a983719574c2f5a67 : SUCCESS in 26m 49s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/3c11e0000528499081f3495bdfcbb92b : SUCCESS in 1h 17m 24s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0a0a315e05b740edbfc4747133f33d5b : SUCCESS in 2h 31m 04s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d564e4b9ec5b4e7ead4a2d32b0c281cc : SUCCESS in 1h 01m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f60d64d011c448efbefcb757ce38864e : SUCCESS in 10m 11s","accounts_in_message":[],"_revision_number":4},{"id":"eccfb59552a0298eb42fbd8bfc09d5fe4a472b62","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-08-16 19:40:30.000000000","message":"Patch Set 4:\n\nCan you explain what do you mean in \"avoid DB calls in some cases\"? I can\u0027t see any logical fixes, just signature change","accounts_in_message":[],"_revision_number":4},{"id":"99dc1b1c0c254f295b0b0409ba8fa8d4311da4f3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-19 08:30:08.000000000","message":"Patch Set 4:\n\n\u003e Can you explain what do you mean in \"avoid DB calls in some cases\"?\n \u003e I can\u0027t see any logical fixes, just signature change\n\nYes, sure. If you look at the default_scheduler.py you\u0027ll see that we now don\u0027t have to sum the result given from the in-memory jobs count and the jobs stored in the DB every time. Now, if we see that there are jobs in memory that satisfy the given criteria then we can return quicker. Practically, on some workflow, this reduced number of DB queries gives a serious different in performance (like I mentioned in the commit message). With the previous method signature we would have to count all jobs (in memory and in DB) every time.","accounts_in_message":[],"_revision_number":4},{"id":"f80f98debb0c689c9cd628f53d44a3ba0ce49e11","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-08-19 13:24:05.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"27896ab555b8912601eb362a1a510eef994a97b8","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-08-25 06:50:34.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"7dc68464696845ad680f9e8422ef5a3344834a9d","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-08-26 08:40:25.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f6a0e8c31402fafee02902d518c24d9316ef4ae9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-28 08:57:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f7614b6eeb4dc9e0020028d629b22252f5c68efc","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-28 08:57:48.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2bb759c718d476604d4efba6b5074d1384d9ec3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 08:58:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bac7dad820951f09495a0e9d5718495e88b24b26","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-28 09:24:52.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4ff5583c5cf4998f0b570287c9dbc414ace5b2d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 10:54:02.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c17594ca6b94382929cc67c95ffa82e : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d63d428b79a429fb894a860bfa694a9 : SUCCESS in 9m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e81fc0e765ac48b0ae7a57fff9100553 : SUCCESS in 8m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e5b36ed06bc4a16875d29f6ee169f0e : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96465ea265764213adf4cd652840102b : SUCCESS in 6m 33s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6f8400d3178c4d268e150692dde0c00d : SUCCESS in 54m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3626a864498b46588f58cc5c668ebebe : SUCCESS in 47m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/dd6246270fd5497e97187c60c0857851 : FAILURE in 31m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fb09beb00c414f7c86add20e2b1d9e2b : SUCCESS in 28m 38s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/0ae23451378d44ada43f10584f64a1d3 : SUCCESS in 1h 36m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/35f762ac1d7a4e86b49b2b1ae3cb1b08 : SUCCESS in 48m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7679a310b1d94e2f8fb143006e17fabf : SUCCESS in 9m 46s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ba7baf5daff74790ae5d29fc311e1626 : SUCCESS in 1h 47m 06s","accounts_in_message":[],"_revision_number":4},{"id":"254ec31c022dd12b4a16cdea6050ae8bd7b529b5","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-28 10:55:04.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"867bea04ac7a9246413134a75b614c11e485a1c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 13:46:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15175333fc674586b79a8bb15f973c64 : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c37436e14974ec9b9ecaebaaf63c308 : SUCCESS in 9m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1157edbc976947528047ac4bb00f7cee : SUCCESS in 9m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e39ee252632c4791a39a82568a0d4e8e : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9554cf54a0fb4554bbf3f54662c74364 : SUCCESS in 5m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6160858b53264958977ade28d87beca7 : SUCCESS in 9m 13s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5345baa7805649dfa95244fd0cb2857f : SUCCESS in 45m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9568acae6e3a45ca938979bd0643fdd9 : SUCCESS in 47m 38s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/881e13ec69364216b65713a60864f2a5 : SUCCESS in 43m 20s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d4fc5c77d41b441f89bef203eab3f6ce : FAILURE in 29m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/033a9ac5515b446e9f9ba9731b05a239 : SUCCESS in 30m 26s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c6bcad197aa940f9938cfa704158a5df : SUCCESS in 1h 24m 58s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8adfb1a51bcb435a8e112c25feedd6f4 : SUCCESS in 2h 48m 09s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e2967981878f46aab608946c57d392e5 : SUCCESS in 1h 08m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15661421a0a642788743170b113f918b : SUCCESS in 10m 53s","accounts_in_message":[],"_revision_number":4},{"id":"88cba10be3e077dcb4de14c5460c1f76639a9917","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-28 13:52:04.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"467e34eb0cb5f072b9384f7c2054ae04b6ab122b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 17:10:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82e7df07e71343b89fc5fe5e8b65dcf8 : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7dfdeb67439443f0a30b818e630e4bcf : SUCCESS in 8m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/551ddf430028463baf58bfcd3af700c3 : SUCCESS in 8m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7eae93b542a4f7f85afeaa5850b4886 : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2f8651f734847a49c966a026b029db2 : SUCCESS in 5m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c852b78959f473fb292b846b5397ae6 : SUCCESS in 8m 35s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0d4b862a66d34eb19f6881c753dd71c7 : SUCCESS in 46m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6be4e1d00e9e46b19202fc8304a5fcc8 : SUCCESS in 44m 26s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/99b96ff98df444c6b948efe8317b2800 : SUCCESS in 45m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ce4cb977f6284e49b784f2a83501ee9a : SUCCESS in 30m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c8dd1d50c65e43b894a858bdca528825 : SUCCESS in 28m 15s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b02420aa90d04b77b9dcf779dadf57b6 : SUCCESS in 1h 32m 44s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4f046c678ab8406f8243e5af3913b833 : SUCCESS in 2h 56m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2735906e44e14778b5b96007d1b22478 : SUCCESS in 1h 06m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16d6e4240e964055993375dbea477e6c : SUCCESS in 9m 19s","accounts_in_message":[],"_revision_number":4},{"id":"5436dd07c29b84f3abab43796ac9394c446273e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 17:10:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d84bf18cfd56a00a247c036d0a2922f2be41f7f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 20:24:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76be62964adc4dc493f35f579a590850 : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9187c3eb8b9c4c3f9cee81182407b15d : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa4055f1b79c4580abde27d320ceae3b : SUCCESS in 7m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9449310b3248499daa399d85c9847798 : SUCCESS in 10m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84043f90e5dd4c9c816a66e62f87369d : SUCCESS in 6m 40s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c853bb2b416b45a19b3c62bcd6cdfe4b : SUCCESS in 45m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/96981529862b4acfbf08df85d904bd25 : SUCCESS in 56m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/180877465b844351aff626e3684afbe2 : SUCCESS in 28m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d2968996ab0240b4b093698f353b5f78 : SUCCESS in 30m 46s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1dadfde6a30440ed9a009586e93ab997 : SUCCESS in 1h 42m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c48d9f358d554b2480c43cd286713640 : SUCCESS in 44m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e97dc832b9fa4e769add60979a6c2f6f : SUCCESS in 10m 18s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/cb3e0d2bc9fa40848666435f7c886a5a : SUCCESS in 1h 48m 26s","accounts_in_message":[],"_revision_number":4},{"id":"ffb29293d884854a54f1e321341f24c6de75358c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 20:24:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ea249849babe9c6a51af2796cd7803e446a06904","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 20:25:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/515f60fca8f74263b4528a1c58bf5027 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0f6bc1897fdc5809e8999c01509d9f4c5feb5eff","revisions":{"be39195aca5c664a26665583b0d19ca585b81884":{"kind":"REWORK","_number":1,"created":"2019-08-15 08:11:07.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/28/676628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/28/676628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/676628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/676628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/676628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/28/676628/1"}}},"commit":{"parents":[{"commit":"ee94087c8972579ce658cf2d0bb59540d8f974e0","subject":"Merge \"Create docs for the workflow namespaces feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee94087c8972579ce658cf2d0bb59540d8f974e0"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-12 04:55:09.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-15 08:10:35.000000000","tz":420},"subject":"WIP: Improve new scheduler","message":"WIP: Improve new scheduler\n\nChange-Id: Ife48d9e464114fd60a08707d8f32f847a6f623c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/be39195aca5c664a26665583b0d19ca585b81884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/be39195aca5c664a26665583b0d19ca585b81884"}]},"branch":"refs/heads/master"},"d85955bf2badd00a16556de69c8991d6f5dbc037":{"kind":"REWORK","_number":2,"created":"2019-08-15 10:18:44.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/28/676628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/28/676628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/676628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/676628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/676628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/28/676628/2"}}},"commit":{"parents":[{"commit":"ee94087c8972579ce658cf2d0bb59540d8f974e0","subject":"Merge \"Create docs for the workflow namespaces feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee94087c8972579ce658cf2d0bb59540d8f974e0"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-12 04:55:09.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-15 10:11:24.000000000","tz":420},"subject":"Improve new scheduler","message":"Improve new scheduler\n\n* Changed method get_scheduled_jobs_count() in the Scheduler\n  interface to has_scheduled_jobs(). In fact, the callers\n  always need to compare the number of jobs with 0, i.e.\n  to see if there are any jobs or not. But more importantly,\n  this semantics (returning just boolean) allows to make a\n  good optimisation for DefaultScheduler and avoid DB calls\n  in a number of cases. Practically, for example, it saves\n  several seconds (5-6) if we run a workflow with 500 parallel\n  no-op tasks that are all merged with one \"join\" task. Tested\n  on 1 and 2 engines.\n* Added test assertions for has_scheduled_jobs()\n* Other minor chagnes\n\nChange-Id: Ife48d9e464114fd60a08707d8f32f847a6f623c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d85955bf2badd00a16556de69c8991d6f5dbc037"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d85955bf2badd00a16556de69c8991d6f5dbc037"}]},"branch":"refs/heads/master"},"fd42ba08a3e563147b2703e101aed2be6942397c":{"kind":"REWORK","_number":3,"created":"2019-08-16 05:15:03.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/28/676628/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/28/676628/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/676628/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/676628/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/676628/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/28/676628/3"}}},"commit":{"parents":[{"commit":"ee94087c8972579ce658cf2d0bb59540d8f974e0","subject":"Merge \"Create docs for the workflow namespaces feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee94087c8972579ce658cf2d0bb59540d8f974e0"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-12 04:55:09.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-16 05:14:44.000000000","tz":420},"subject":"Improve new scheduler","message":"Improve new scheduler\n\n* Changed method get_scheduled_jobs_count() in the Scheduler\n  interface to has_scheduled_jobs(). In fact, the callers\n  always need to compare the number of jobs with 0, i.e.\n  to see if there are any jobs or not. But more importantly,\n  this semantics (returning just boolean) allows to make a\n  good optimisation for DefaultScheduler and avoid DB calls\n  in a number of cases. Practically, for example, it saves\n  several seconds (5-6) if we run a workflow with 500 parallel\n  no-op tasks that are all merged with one \"join\" task. Tested\n  on 1 and 2 engines.\n* Added test assertions for has_scheduled_jobs()\n* Other minor chagnes\n\nChange-Id: Ife48d9e464114fd60a08707d8f32f847a6f623c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fd42ba08a3e563147b2703e101aed2be6942397c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fd42ba08a3e563147b2703e101aed2be6942397c"}]},"branch":"refs/heads/master"},"0f6bc1897fdc5809e8999c01509d9f4c5feb5eff":{"kind":"REWORK","_number":4,"created":"2019-08-16 06:39:58.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/28/676628/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/28/676628/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/676628/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/676628/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/676628/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/28/676628/4"}}},"commit":{"parents":[{"commit":"ee94087c8972579ce658cf2d0bb59540d8f974e0","subject":"Merge \"Create docs for the workflow namespaces feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee94087c8972579ce658cf2d0bb59540d8f974e0"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-12 04:55:09.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-16 06:39:39.000000000","tz":420},"subject":"Improve new scheduler","message":"Improve new scheduler\n\n* Changed method get_scheduled_jobs_count() in the Scheduler\n  interface to has_scheduled_jobs(). In fact, the callers\n  always need to compare the number of jobs with 0, i.e.\n  to see if there are any jobs or not. But more importantly,\n  this semantics (returning just boolean) allows to make a\n  good optimisation for DefaultScheduler and avoid DB calls\n  in a number of cases. Practically, for example, it saves\n  several seconds (5-6) if we run a workflow with 500 parallel\n  no-op tasks that are all merged with one \"join\" task. Tested\n  on 1 and 2 engines.\n* Added test assertions for has_scheduled_jobs()\n* Other minor chagnes\n\nChange-Id: Ife48d9e464114fd60a08707d8f32f847a6f623c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0f6bc1897fdc5809e8999c01509d9f4c5feb5eff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0f6bc1897fdc5809e8999c01509d9f4c5feb5eff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
