)]}'
{"id":"openstack%2Fmistral~676897","triplet_id":"openstack%2Fmistral~master~I1c937ff05c4a1f3e28010427c03e3e4f113f6fba","project":"openstack/mistral","branch":"master","topic":"bug/1814850","hashtags":[],"change_id":"I1c937ff05c4a1f3e28010427c03e3e4f113f6fba","subject":"Add \"retry_count\" field into workflow execution report","status":"MERGED","created":"2019-08-16 10:02:59.000000000","updated":"2019-08-28 13:16:51.000000000","submitted":"2019-08-28 13:15:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676897-1566998130585-30b2ecef","meta_rev_id":"7d858856697fa39bac60b95b26b8bb34209e8cfe","_number":676897,"virtual_id_number":676897,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"date":"2019-08-28 08:15:54.000000000","_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-08-28 13:15:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-08-26 08:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":1,"date":"2019-08-20 11:24:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2019-08-16 19:46:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":1,"date":"2019-08-21 09:13:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-28 05:27:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"date":"2019-08-19 08:30:27.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-08-26 08:40:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2019-08-19 08:30:27.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"date":"2019-08-19 08:30:27.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 19:46:21.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-08-19 08:30:27.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-08-19 08:30:27.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-08-19 08:30:27.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-08-20 11:24:44.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2019-08-21 09:13:38.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-08-26 08:40:56.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-08-28 05:27:03.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-08-28 13:15:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a60586eeee775e865b222b79f72ad20b0087b7a1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-16 10:02:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a2b60b544d6ea79dc9bd706fe46ab48f0a8d9c8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-16 10:05:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ab8dea6b1d28c1cf5f307d7adcb5eeaa3896fe2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 12:46:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e59f349705e4174ac57e50735b1e4e5 : SUCCESS in 10m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa06c473d770446a8bee1ad56a77b0df : SUCCESS in 15m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52c6c9ae816c4b8cb49efb05e074d91c : SUCCESS in 8m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1e8316ff35e4dfd9fed4e552af8d438 : SUCCESS in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4806fabf72df43b2a95240e6305ded7c : SUCCESS in 9m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fc9245c2c2140c091905ea4de290aec : SUCCESS in 10m 19s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/429859089c5d4b6fb061486590f0634f : SUCCESS in 45m 56s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a401919fdb6a49dd8c40628ac6427a8f : SUCCESS in 59m 27s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d20dbb21e70d4295918087292487394c : SUCCESS in 53m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fc9d1021be1a48afbf10cd430fcde05e : SUCCESS in 28m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/989b18422bd74b56a7774aad0b998eaa : SUCCESS in 30m 46s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/acb4d5aba4a4464e908eb476a10fe1e8 : SUCCESS in 1h 23m 11s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0aa42ecf2e7e4799bdeb6cd09c9f9050 : SUCCESS in 2h 36m 09s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2e6dcb83ac2e42f9a0f0214cef45cf3d : SUCCESS in 56m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2189a9f193084916b6c7cb4f1ffdde5f : SUCCESS in 9m 46s","accounts_in_message":[],"_revision_number":2},{"id":"e368547febdae2c287276c7fd8024953594a0835","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-08-16 19:46:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3d37b3cf87739175e7ed985b89a2a674ccfed28e","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-08-20 11:24:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"51accd40be38b24ddc1529694a03e31bbd129757","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-08-21 09:13:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a0e6de237ac6bd8edebd9285740b38bb9ed0c693","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-08-26 08:40:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8e69e71c4a845797c565feaf6de953ffdfbe43ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 08:41:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fcf7c00705a45ad62e986384e94d5aabc820abc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 10:17:18.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf84fd294e8499a9a677f69678ef96d : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7cc3e03847dc4ec8989aab8e79ce31e7 : SUCCESS in 9m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc90d7a7fa0e4a0db7dc35a4427e94bc : SUCCESS in 32m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6245dbb89ce1439680c0c1f21178543c : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce93087eafaa4a78bd3db0ebac3e13cc : SUCCESS in 5m 47s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/69b039a0c5994918877039f0812accb7 : SUCCESS in 51m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8fd59d8036fc46df9e5e5135d531ef9a : POST_FAILURE in 1h 19m 22s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b2b27f9f925c43be9d9d8b51f341e56b : SUCCESS in 30m 16s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5cfc4c18d97b48bf8a979fd252f0d283 : POST_FAILURE in 56m 49s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/be6ac0fdec8b4d6aa95ba30e0b1aa3a4 : SUCCESS in 1h 08m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/81b8d970737243dab098e0567ebc22b7 : SUCCESS in 48m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57dcbb5fae5343b98bc9d55a151823d7 : SUCCESS in 10m 17s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c819a0f9ab1f463bac2c1f4dcda972c2 : SUCCESS in 1h 29m 13s","accounts_in_message":[],"_revision_number":2},{"id":"295bd7fe2ea0061706521b81acb810289449fa11","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-28 05:00:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4f6d9c8a9fdff62d15890bef97eccd22f41e2b1f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-28 05:27:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2eca406981e7bc8e28fc6ec1b92036d854511402","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 07:56:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a69e5ce8407743f6ae89c30af783d862 : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3eb9d6cc39a3496cbc0a0f41fb4f7307 : SUCCESS in 9m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6ac8b234a1245a4a4fccd90cd222985 : SUCCESS in 10m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9499f46d8d694678ab7b6c931a9f9556 : SUCCESS in 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b939e87651704d9c9f69828c1b29a1d5 : SUCCESS in 6m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ecea1de49b94620b7db292275daca0b : SUCCESS in 9m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3634468e73354fafaa6f2523fe60bc16 : SUCCESS in 51m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f2abc9d058574572830f3888a559cf12 : SUCCESS in 51m 27s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4f47ce107d714fd6ab834fdc371ec771 : SUCCESS in 49m 06s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a8c91bf75c974de18f0d81b8a06b749d : SUCCESS in 29m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b47bc1688e7f467faf55c0fa0fdd5ace : FAILURE in 28m 06s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a74661beb9fc4ab9ab23a16177d60a03 : SUCCESS in 1h 31m 38s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a2625c70dfdc4141b9e260972418bec7 : SUCCESS in 2h 55m 11s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6abe0e5be2bc4bbd847dde8c7262ff42 : SUCCESS in 1h 06m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d80ffa8898142b2b330c3ea899222e8 : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":2},{"id":"9611cfec8198ab72ddd73ed7c8e59c93fe8c96ca","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-28 08:15:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b5e3aebe05a82a89f0d004a6b46360a47a75a45d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 11:15:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df7e2741512a4ad6993eea0340333279 : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fcb15b8b97bf430e99b8ee9a19bf0cd4 : SUCCESS in 9m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/665638a68bc44903896f31c4bab81ede : SUCCESS in 8m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0a6aa6923b64dc98c274e51c3fc2132 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de1ceb5fd77343bda2f46282021b93cc : SUCCESS in 5m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfa4dc9755f74f24ba00ab9467b62b28 : SUCCESS in 9m 35s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0d00764b13224b46854b27d9d5d4de82 : SUCCESS in 49m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/941d82d1ff8a48cd8287b4a761a9c3da : SUCCESS in 42m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ce9884907a8d43fb849da248cf5a7805 : SUCCESS in 41m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1b394be6da0241e9a09a969f1c56f8a4 : SUCCESS in 27m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e8b8c02ed5864fc697d4f7a22b725d02 : SUCCESS in 27m 28s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/818722adc6744c20a61926671e5e1510 : SUCCESS in 1h 41m 12s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/29e2e657c1814fa9a710f1ab4ea5a5e4 : SUCCESS in 2h 52m 06s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e6ab8506f72d44ec8cc33a17ae54ffe9 : SUCCESS in 1h 10m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/090c060c032e4fb18c6a99a72289898f : SUCCESS in 10m 17s","accounts_in_message":[],"_revision_number":2},{"id":"095dd401d6258e9c7db9df59e6108d7a0add3b4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 11:15:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"19feddc5e55e40ceb7c190e592f3ef310e7f4b85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 13:15:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/946648d34ebd4a34a3e850d0a71002f4 : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/412fb1f4d2244d9fa33beb822f96c652 : SUCCESS in 10m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/961509be6d97476ba213e1cabd655d4d : SUCCESS in 8m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4f7bb69623b4b08b5035842d285aa4d : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca3b8438700c4af98f457e40aae74fd9 : SUCCESS in 13m 14s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ece4d5fd931042cba80797027bb6cd3a : SUCCESS in 49m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1f889c4ef28a4591bed475735afdf822 : SUCCESS in 43m 18s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/693de7e08329441795126afece6c44f8 : SUCCESS in 28m 01s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bcfa76e0f70a4b16a9dfd1d7382011e3 : SUCCESS in 26m 39s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/9114e80ba23f402c8e6232146830db1b : SUCCESS in 1h 28m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0ff7d563b2f145f58d9b837d2bb05ba3 : SUCCESS in 42m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/503c0d2708bf4ab89ab799389194576d : SUCCESS in 11m 19s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/cdc8b1584831474aba2edf6e8eb6c49e : SUCCESS in 1h 52m 46s","accounts_in_message":[],"_revision_number":2},{"id":"fc16dc1c953c2fe6a8842fa2712faf1a01809d52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 13:15:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a61f786a0e797b2df508137d52753c0da6d79f78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 13:16:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/687081570ebe4ab19d3d83a06b550e6d : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6c0bd2a26844b215f16252af7c28a88513451c90","revisions":{"ba74dd177955ef39cc15f18441717a731491865c":{"kind":"REWORK","_number":1,"created":"2019-08-16 10:02:59.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/97/676897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/97/676897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/676897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/676897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/676897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/97/676897/1"}}},"commit":{"parents":[{"commit":"4e926a1f136a54ddf83e9eae43c33d9636d48aaf","subject":"Fail-on policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e926a1f136a54ddf83e9eae43c33d9636d48aaf"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-16 10:01:39.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-16 10:01:39.000000000","tz":420},"subject":"Add \"retry_count\" field into workflow execution report","message":"Add \"retry_count\" field into workflow execution report\n\nChange-Id: I1c937ff05c4a1f3e28010427c03e3e4f113f6fba\nCloses-Bug: #1814850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ba74dd177955ef39cc15f18441717a731491865c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ba74dd177955ef39cc15f18441717a731491865c"}]},"branch":"refs/heads/master"},"6c0bd2a26844b215f16252af7c28a88513451c90":{"kind":"REWORK","_number":2,"created":"2019-08-16 10:05:59.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/97/676897/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/97/676897/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/676897/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/676897/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/676897/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/97/676897/2"}}},"commit":{"parents":[{"commit":"4e926a1f136a54ddf83e9eae43c33d9636d48aaf","subject":"Fail-on policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e926a1f136a54ddf83e9eae43c33d9636d48aaf"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-16 10:01:39.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-16 10:05:43.000000000","tz":420},"subject":"Add \"retry_count\" field into workflow execution report","message":"Add \"retry_count\" field into workflow execution report\n\nChange-Id: I1c937ff05c4a1f3e28010427c03e3e4f113f6fba\nCloses-Bug: #1814850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6c0bd2a26844b215f16252af7c28a88513451c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6c0bd2a26844b215f16252af7c28a88513451c90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
