)]}'
{"id":"openstack%2Fmistral~678416","triplet_id":"openstack%2Fmistral~master~I16aa8fa90412b125f4940821ef8f2c0060475fca","project":"openstack/mistral","branch":"master","topic":"eyalb/py3","hashtags":[],"change_id":"I16aa8fa90412b125f4940821ef8f2c0060475fca","subject":"DNM: use python3 for rally","status":"ABANDONED","created":"2019-08-25 11:45:01.000000000","updated":"2019-08-25 15:01:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eca643f55af59a871ceb06b2c43f1272be930158","_number":678416,"virtual_id_number":678416,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-25 14:02:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f1a6d95941655e8504e0ec9f6099a43be29ec82","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-25 11:45:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25593d3d458f67eb66f26a567bf4da2eafbda44c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-25 11:45:36.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"e597137b29e10431b095d01aae8e83868dea3abb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 14:02:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/097a550c13094265801f14bed7922800 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1103986a146c4420bd38aa7040968364 : SUCCESS in 9m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcc03065bafa42e3ba86526a581beaf9 : SUCCESS in 8m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b522137b527c4e7fb1005ab60aa0f03f : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb2330c67a8544a6bed5bc3565b6e2c5 : SUCCESS in 4m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0aa0dc5f45c346ba8851dbb2db19e035 : SUCCESS in 9m 01s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3cbae6126651457f9d231c3af576d817 : SUCCESS in 38m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6d0e1e3c60904a02a0ffcab4c27908f0 : SUCCESS in 44m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/479cea1a24f14a6b955664bebd81f3ed : SUCCESS in 38m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1a5eadaa9dd64a2a8902f0aa8dba285e : SUCCESS in 28m 32s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/34a648b55c374571b29e67c4a2e71f0e : SUCCESS in 27m 42s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/023c03ad1acb4fb8a6dcd509b47f7c5d : SUCCESS in 1h 07m 50s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/79fb554e0f6c44acb9835e4d14b14b43 : SUCCESS in 2h 15m 51s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/97f3b9bc1847466789419e445b238765 : SUCCESS in 1h 06m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee7808c0865744cbbbf8f95f2e356a74 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6d05f50dccfcffa3ba9334c759bf390458e3ddc0","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-25 14:05:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac67e406c93756cd55cefe24d74cf7d4584230ff","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-25 15:01:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dbfdf78859cebefa4aed6736bffa8abf6f6db14f","revisions":{"d9d441818991263244dd9e494dde656ec811bb73":{"kind":"REWORK","_number":1,"created":"2019-08-25 11:45:01.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/16/678416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/16/678416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/678416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/678416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/678416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/16/678416/1"}}},"commit":{"parents":[{"commit":"61b20e084c74f934803d32aabda6dfe9ee3ca95f","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/61b20e084c74f934803d32aabda6dfe9ee3ca95f"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-25 11:44:45.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-25 11:44:45.000000000","tz":180},"subject":"DNM: use python3 for gate","message":"DNM: use python3 for gate\n\nChange-Id: I16aa8fa90412b125f4940821ef8f2c0060475fca\nDepends-On: Idc39d6f6e367d3dea673d48a160b8314a51394e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d9d441818991263244dd9e494dde656ec811bb73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d9d441818991263244dd9e494dde656ec811bb73"}]},"branch":"refs/heads/master"},"dbfdf78859cebefa4aed6736bffa8abf6f6db14f":{"kind":"REWORK","_number":2,"created":"2019-08-25 14:05:42.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/16/678416/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/16/678416/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/678416/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/678416/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/678416/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/16/678416/2"}}},"commit":{"parents":[{"commit":"61b20e084c74f934803d32aabda6dfe9ee3ca95f","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/61b20e084c74f934803d32aabda6dfe9ee3ca95f"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-25 11:44:45.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-25 14:05:06.000000000","tz":180},"subject":"DNM: use python3 for rally","message":"DNM: use python3 for rally\n\nChange-Id: I16aa8fa90412b125f4940821ef8f2c0060475fca\nDepends-On: Idc39d6f6e367d3dea673d48a160b8314a51394e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dbfdf78859cebefa4aed6736bffa8abf6f6db14f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dbfdf78859cebefa4aed6736bffa8abf6f6db14f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
