)]}'
{"id":"openstack%2Fmistral~679203","triplet_id":"openstack%2Fmistral~master~I3df711ef2484374418085fe0117fe8b37ce5ba3f","project":"openstack/mistral","branch":"master","topic":"bug/1839840","hashtags":[],"change_id":"I3df711ef2484374418085fe0117fe8b37ce5ba3f","subject":"Fix \u0027with-items\u0027 expression evaluation","status":"MERGED","created":"2019-08-29 08:38:01.000000000","updated":"2019-09-04 23:21:44.000000000","submitted":"2019-09-04 23:19:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679203-1567639194001-321edceb","meta_rev_id":"80af6b68d1102b57e0ed3272a2dc47008730a36a","_number":679203,"virtual_id_number":679203,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-09-04 23:19:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"recommended":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":1,"date":"2019-09-04 03:57:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2019-09-04 03:57:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2019-09-04 03:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-04 03:57:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-09-04 03:57:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-30 09:01:13.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-09-01 05:10:44.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-09-01 10:53:40.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2019-09-02 07:32:09.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-09-04 23:19:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"00d1de1e39c9b7391e2402f4ccfc084fa5d5d990","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-29 08:38:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bba54bc98605c6898e57e03081d0a08aa99d8ab7","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-29 08:39:04.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5ceb6fa1d1d1344f22d645a1b0daf232bdd91e63","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-29 09:44:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7e9c2ae88f2e4c9e014826f3cb2fc60ca8b2898b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 12:40:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/908de707272c4df2ae5c23aa25e80292 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2dd7131d32ed48638ee9979110d47375 : SUCCESS in 8m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76c543d719d744038402019f4139957f : SUCCESS in 9m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c11fc077fba466f8dd1aed3210bf2ae : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/003f5dc908c04d1fbfdc2eebb4fbc567 : SUCCESS in 5m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/715d807f9d684e318e00820449e73265 : SUCCESS in 9m 19s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c7e196f8e54140018ed3776eef3b86c2 : SUCCESS in 41m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/496ebaade1cf4501ae2104f5c1f80997 : SUCCESS in 42m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/13f19eb65d274bc19e0d4c72a542ece0 : SUCCESS in 41m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/277b29a264e94e9aa7fd439b0d550b53 : FAILURE in 29m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/432a67a6cce541cbb8c946eafbe8e545 : SUCCESS in 27m 41s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/d4e491c4a2eb4a1096e94a1618534bd4 : SUCCESS in 1h 22m 42s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b7841a8000f54c12878f2cf8360e3b3a : SUCCESS in 2h 48m 11s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4f3206debb474fdabcacdef5f538bd0b : SUCCESS in 1h 05m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e072696f83d42c3bca30025364f1793 : SUCCESS in 11m 35s","accounts_in_message":[],"_revision_number":3},{"id":"70206f1c03b5f1ad5175dee57ae4923413c4afeb","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-29 12:44:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"dfa68423519e213f2620ab9da8dbb92454487fcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 15:54:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/453c3ed200944a208f2fdd3a2f4216a4 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d38289c65b3443a9684cc1b57d4b3b6 : SUCCESS in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb65656ef86b40069996db7989c7243e : SUCCESS in 8m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/93280f6758eb42c0832071eb3806f7d1 : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7736dafb10144148bfaebf1d6ca53fd5 : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a27f76ebe1e643eb99d7a2f120691448 : SUCCESS in 10m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7be49d22e3264bf6a5a44405e8833130 : SUCCESS in 53m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/503fc22b455f4ecfb2bf0d97c3073151 : SUCCESS in 51m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ebc407ebbb62494588aa739c6d2ac31c : SUCCESS in 1h 01m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/55e4a8e9d3a5473f81e45e2c8616938f : SUCCESS in 33m 00s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e4a82837988e40a5a4e0ced847c1be07 : SUCCESS in 27m 15s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/eab86c417d8e4f53aff356c03bf27a8a : SUCCESS in 1h 29m 55s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4f27dfc1ddff4391b62c2ae6c5c7da5c : SUCCESS in 3h 03m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d70db0016aa74969a10a4ef81f2dd2bd : SUCCESS in 1h 07m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/384184617dcb42fbbe87b5f89c771a24 : SUCCESS in 11m 13s","accounts_in_message":[],"_revision_number":3},{"id":"d582f004132d52f2f42c4431348d67b44af4e8ed","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-08-30 09:01:13.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"886703c12e4adb77bdae778a17ce668de8344278","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-01 05:10:44.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e64ab33edcc025d8b5177852146b808214c5834b","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-09-01 10:53:40.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a953a81fda1fb4a882c865f3ba56d3f6c31dbb76","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-09-02 07:32:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"82e1206f732b211cc9da14faf24a5a37eae1807a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-02 07:48:33.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c5d3fc24eadc1285f1057b1ab9b090ea44f8509a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 07:48:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fa34b50829a16745054e54275ba722b9f73d6b7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 08:48:27.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/614c2791de794f57b53727c2cd3daa76 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5e861cd41be4586b2011c558e7e58b7 : SUCCESS in 9m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da67c7070bb649338fe8623a12ab57b2 : SUCCESS in 8m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da20a71f066b43cc807562a380f79b81 : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a8c9c482354063973c02430330b7b2 : SUCCESS in 5m 30s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3cd387c86f5b46d997ebd6e1a85f101e : SUCCESS in 42m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b472a0007fce4112b022d38d3f183436 : SUCCESS in 45m 17s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1f93f37aeb4845b6b2ceeea28b7c3272 : SUCCESS in 28m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/625274730b1e4da493aa872f9b431fb2 : SUCCESS in 27m 44s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a87c03397890452cabf9d2af917c0e0f : FAILURE in 22m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4fc342b5cacd48f38955c2bc6be57a9a : SUCCESS in 43m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/094a9741cbb8489995ff3a6c2c48d893 : SUCCESS in 11m 05s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/20e4904d708f422596ece33eb2befb60 : FAILURE in 22m 13s","accounts_in_message":[],"_revision_number":3},{"id":"77aa0086daa18d9fc6f570dc4d9842c3b430b2af","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-03 08:21:50.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"357cbd22795a865d5a64e3bc3499f9c8f014d868","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 11:19:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/263bfda960f74464b8208f46e4aedf01 : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e0596587a37443985d65a2dc9e0ddb6 : SUCCESS in 10m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/965a277b0169452eb6b06896964dce9c : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72d387f45be94dc49c40c9c2a19fb132 : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7125d48d3bc466c98cbe3f1d85a3a3c : SUCCESS in 5m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18cebb2656a94870a0b418d81ad79652 : SUCCESS in 9m 33s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7d218a169d414d85b5c8b7a8561e9e1e : SUCCESS in 41m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0687a898230a4cc2a9b5efbde8a083c7 : SUCCESS in 42m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f1d5516590dd470eb9ed72787f3cf260 : SUCCESS in 43m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/908317f3235a4e74aa7ef969fc62a8fd : SUCCESS in 29m 20s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0231e90f148748f7a18f7f44e287904d : SUCCESS in 26m 57s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c994b1e3a121400bbf8e51f85933e4be : FAILURE in 26m 55s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/91efffa29c4a476e8f928b28185dfef6 : SUCCESS in 2h 51m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6eacff81eb4848a2a6fb2e589e1ef847 : SUCCESS in 1h 13m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef085127afa04eb0937d79340f3145ef : SUCCESS in 12m 29s","accounts_in_message":[],"_revision_number":3},{"id":"d308a928b9fe8b534654a87094c5440197ef2daa","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-04 03:57:03.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"9f99e05148a5caac5b22bf614cbe2b3da421ba72","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-04 03:57:07.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"22959f2099157a744cd8b2e4e59d6c52f43460c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 07:29:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e009880969ab4275832cc3b572924527 : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/05b60ed51b4342e78d3344e64a2ec26c : SUCCESS in 9m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1973d33cea54ca1bf597f04b02bfa6c : SUCCESS in 8m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25060e492040455e85d937fbe66f788c : SUCCESS in 10m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3e1d90873d04b2ca75524c32d02dec8 : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e01ede6b49e46fdbc473633871eb9b2 : SUCCESS in 10m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9e88bc7082d049beaee244028713e9c3 : SUCCESS in 42m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d584a6c4e40e4701b792bec4e9b1ccfb : SUCCESS in 46m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cb6401109b43457aa838949b65d5dab7 : SUCCESS in 43m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/08044311d12b43579c95b6d5d52d4ffa : SUCCESS in 29m 57s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c7741c9f8fcf41eebb15dc3c1d2007ea : SUCCESS in 28m 41s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/00cd4705dd7f46e7bfc609c80ce8a5c2 : SUCCESS in 1h 29m 59s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/622d6b9a188946ecb357dd2a8b764434 : TIMED_OUT in 3h 19m 18s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/fa9189d1bca34056aff7cdd29cc8c48a : SUCCESS in 1h 04m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/027ddc9d6d554bedb0386b2edd6e7350 : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":4},{"id":"7b8625e86c330fffad1d32f421c70065fa4354b9","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-04 08:31:27.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9855ddae20e546f85eb3035a9e9c5ba5c776052d","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-09-04 09:42:18.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5f0e99d35477d16afdf052f2c5528fe11673555b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 11:50:57.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9ebc3e9741b46d4a187223aac5ad959 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2793417078e14affae2a11863ae40a4a : SUCCESS in 10m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e51649c314624aa580efc81c2f3f16da : SUCCESS in 8m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aff149db4871464ab3570b2b106bc51d : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02d0fc5f0f034367b9634bde7129f300 : SUCCESS in 5m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a48158f7637c4c6b97c0faf81059eb2a : SUCCESS in 9m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0a55c233b90c4ac98ce55937a12e4e97 : SUCCESS in 46m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b78d7c76cd21468391517475a7c45313 : SUCCESS in 1h 03m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/97a018ee75f94f00b0d5e89846189372 : SUCCESS in 59m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e30f51ed11db43cabcbe23d72957f35f : FAILURE in 31m 01s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bb04bf35c3a548f79b0349bc3113a92f : SUCCESS in 28m 13s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/2634a5ea3e9646c3ab9633b81016e884 : SUCCESS in 1h 31m 37s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/db9fc5be0a4f4e0dbecdd64fce01143e : SUCCESS in 3h 13m 19s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4efbef6da27f4e75afea6edb37c8ea77 : SUCCESS in 1h 03m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1e18fb05d4f444d8db8c0e4087ef4d5 : SUCCESS in 11m 57s","accounts_in_message":[],"_revision_number":4},{"id":"451007c332447e34355a6c7d3199137c05559aa3","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-04 11:52:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"59f3e317ff1b58013f3218646e742924e2f75a3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 14:47:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b29b96f41884a20b1fd5485d6e1c849 : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cca4225e49e044af975edd746012263a : SUCCESS in 9m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f84e61e2d2842b99b83ed55000be197 : SUCCESS in 9m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e9a77da8e1a4fd4a397d873439b637b : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd1c8ed983c8450f980be02cd6af5002 : SUCCESS in 5m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01424739b2cb47c0921c0c7464aedb2b : SUCCESS in 10m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/bf1760304bc84f56af72e3e61ce2252c : SUCCESS in 38m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d629fb16f63f44d18e55a8cde1b318f4 : SUCCESS in 40m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7da07ace868b49e7b810a50cc09eb5d5 : SUCCESS in 41m 32s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/31b3399edd684bf48ca150a8545c9ca8 : SUCCESS in 27m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/be9f9d8e32a945db9f3bc4ab092a4585 : SUCCESS in 25m 38s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/af173873b3ae48048b65a33fe1fc90ce : SUCCESS in 1h 38m 38s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1a198041e4d64bdd94fef0061bd8d792 : SUCCESS in 2h 49m 41s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ca328de39de34c80b32b4e2c674578e3 : SUCCESS in 1h 04m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9470c05a9ba6403ca5430046dc6d620a : SUCCESS in 9m 39s","accounts_in_message":[],"_revision_number":4},{"id":"786b5d309d26554bf1aa032d9dfb2f55d88c6aae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 14:47:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4d0322dbeb71bdb621f768c412051b5561efe056","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 18:11:44.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc6cd3e400f048378f8352d325efc958 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/107fc50b78fe4063ac415ed30c406435 : SUCCESS in 12m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d63b898035c4962aa5773958ef710f7 : SUCCESS in 9m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59d3256a5fd34eb3b8a0f943a07c9077 : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e45dfe95b5354e00865a4d0dd519a865 : SUCCESS in 6m 58s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c1b94d2fba06474aa608812d4a237d97 : SUCCESS in 45m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/eefc015977ac4c3d96a6900d15cb6495 : SUCCESS in 46m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/24b562de25e3431c83c78569d811786e : SUCCESS in 29m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0884f2c2ec13435d9f3152529de44d96 : SUCCESS in 30m 34s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/7b185dd6441b4b3e9293acb290ab1e63 : POST_FAILURE in 1h 52m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f79eaf95c7c8411495a2229c9e4b5e16 : SUCCESS in 40m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8835d9ffc801490d98019dca7d2be505 : SUCCESS in 9m 50s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d375dbf63afe46968b743d5a09264cd2 : POST_FAILURE in 1h 47m 03s","accounts_in_message":[],"_revision_number":4},{"id":"51682bb2209f849da37991fc0ab993cd42d4ed6e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-04 18:13:05.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a179b52d44310c88b5045894a5d5c46377d6a470","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 21:07:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab34f96339c3453ab82c31a6776674c8 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9938abf2bb25425aa8dbed710a53fa7c : SUCCESS in 10m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/274d64eb956e4ed397feb8fde5ce351e : SUCCESS in 9m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf872f03bcd94a1690ae46ce94f5de9f : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/220fbd4d363944fcbd6c930ee39c7788 : SUCCESS in 7m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e514d630b31459e8722e5bd0f3e6925 : SUCCESS in 10m 48s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/53b3653d989542c48545922d484b1030 : SUCCESS in 42m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c1203213fbcf4b889065c8a2e2cf2d5b : SUCCESS in 46m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/41fd445bca1a4e84bbb250e431483728 : SUCCESS in 43m 54s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7b23c86060f944d78a291f9be7302f75 : SUCCESS in 28m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2f7b809575ea43228c64faf253b2be93 : SUCCESS in 28m 31s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1cb8059e4d1746e49e1e2fdc5833e2a6 : SUCCESS in 1h 51m 40s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ab98efe530414f65a86aff56be02fd15 : SUCCESS in 2h 51m 25s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bccda0dc463740ada860f680c0c9f237 : SUCCESS in 1h 11m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d79c774378a14aacaf0e0927903fb3ce : SUCCESS in 11m 26s","accounts_in_message":[],"_revision_number":4},{"id":"b8c9245061525fb74e574391b12a8b1927a517b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 21:09:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"367e9768af0ebf78094f094c0f53550f3381867f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 23:19:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/109cf03f0cf845b698d2f4684f613683 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c0db524ce9684131972057a5924e6751 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55cad597854f432484d41ed3e92ca587 : SUCCESS in 10m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f99b8c1a99643a7a7c4ccd498ae3a0e : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20e8f9271df040e38086e3a07ec8f4f8 : SUCCESS in 5m 36s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/84cf66f895c14ef4b577d88583612d27 : SUCCESS in 39m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0dc29777e67040b59140c778b6cad053 : SUCCESS in 42m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a17899a59f8f472b80157f598cbf6ed0 : SUCCESS in 28m 33s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/92aad548f07540329afdf13a2b1c1951 : SUCCESS in 28m 19s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/04b8e860dada4d86b8ff1cad6f372c09 : SUCCESS in 1h 40m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/083c4fe1cadf41b18a0d6295401122b3 : SUCCESS in 52m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1527a0fe08e4ab7ac8b7ae2d076ef3e : SUCCESS in 12m 05s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/95e5b47c0cc24194a7559ccfa4985b1e : SUCCESS in 2h 04m 22s","accounts_in_message":[],"_revision_number":4},{"id":"444e573f183cf055f15195fcc400608a200cd01e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 23:19:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"0df6e7b5df7ef5b98dad30f2ffae5e12b0b3e1fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 23:21:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e250509208de41d08caf89dc37bf3f21 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f92a5c8f445f132c230f20429bda0d1889996a84","revisions":{"f839c57b42229dac33cd5e5a7267b7be4c029e9a":{"kind":"REWORK","_number":1,"created":"2019-08-29 08:38:01.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/03/679203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/679203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/679203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/679203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/679203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/679203/1"}}},"commit":{"parents":[{"commit":"2cdcb5415ead23a129d2e6698833ce10e6a6cf4a","subject":"Merge \"Improve new scheduler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2cdcb5415ead23a129d2e6698833ce10e6a6cf4a"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-29 08:34:23.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-29 08:34:23.000000000","tz":420},"subject":"Fix \u0027with-items\u0027 expression evaluation","message":"Fix \u0027with-items\u0027 expression evaluation\n\n* There was a bug left after the recent refactoring. While\n  evaluating \u0027with-items\u0027 expression we didn\u0027t construct a context\n  view properly, it didn\u0027t include a workflow environment. This\n  patch fixes it.\n\nChange-Id: I3df711ef2484374418085fe0117fe8b37ce5ba3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f839c57b42229dac33cd5e5a7267b7be4c029e9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f839c57b42229dac33cd5e5a7267b7be4c029e9a"}]},"branch":"refs/heads/master"},"6dfc6bb9968cf83526e8f945e293077618dd44d3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-29 08:39:04.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/03/679203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/679203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/679203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/679203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/679203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/679203/2"}}},"commit":{"parents":[{"commit":"2cdcb5415ead23a129d2e6698833ce10e6a6cf4a","subject":"Merge \"Improve new scheduler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2cdcb5415ead23a129d2e6698833ce10e6a6cf4a"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-29 08:34:23.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-29 08:38:34.000000000","tz":420},"subject":"Fix \u0027with-items\u0027 expression evaluation","message":"Fix \u0027with-items\u0027 expression evaluation\n\n* There was a bug left after the recent refactoring. While\n  evaluating \u0027with-items\u0027 expression we didn\u0027t construct a context\n  view properly, it didn\u0027t include a workflow environment. This\n  patch fixes it.\n\nCloses-Bug: #1839840\nChange-Id: I3df711ef2484374418085fe0117fe8b37ce5ba3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6dfc6bb9968cf83526e8f945e293077618dd44d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6dfc6bb9968cf83526e8f945e293077618dd44d3"}]},"branch":"refs/heads/master"},"0be34da43bb439693cf141db22b2911abff8bdf8":{"kind":"REWORK","_number":3,"created":"2019-08-29 09:44:20.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/03/679203/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/679203/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/679203/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/679203/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/679203/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/679203/3"}}},"commit":{"parents":[{"commit":"2cdcb5415ead23a129d2e6698833ce10e6a6cf4a","subject":"Merge \"Improve new scheduler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2cdcb5415ead23a129d2e6698833ce10e6a6cf4a"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-29 08:34:23.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-29 09:44:05.000000000","tz":420},"subject":"Fix \u0027with-items\u0027 expression evaluation","message":"Fix \u0027with-items\u0027 expression evaluation\n\n* There was a bug left after the recent refactoring. While\n  evaluating \u0027with-items\u0027 expression we didn\u0027t construct a context\n  view properly, it didn\u0027t include a workflow environment. This\n  patch fixes it.\n\nCloses-Bug: #1839840\nChange-Id: I3df711ef2484374418085fe0117fe8b37ce5ba3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0be34da43bb439693cf141db22b2911abff8bdf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0be34da43bb439693cf141db22b2911abff8bdf8"}]},"branch":"refs/heads/master"},"f92a5c8f445f132c230f20429bda0d1889996a84":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-09-04 03:57:03.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/03/679203/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/679203/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/679203/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/679203/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/679203/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/679203/4"}}},"commit":{"parents":[{"commit":"99194bddc8933e8b0915cfc1a7828746781c6775","subject":"Merge \"remove unused gate code * post_test_hook. used in legacy gate * run_functional_test probably not used and also doean\u0027t work   since mistral_tempest was moved\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99194bddc8933e8b0915cfc1a7828746781c6775"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-08-29 08:34:23.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-04 03:57:03.000000000","tz":0},"subject":"Fix \u0027with-items\u0027 expression evaluation","message":"Fix \u0027with-items\u0027 expression evaluation\n\n* There was a bug left after the recent refactoring. While\n  evaluating \u0027with-items\u0027 expression we didn\u0027t construct a context\n  view properly, it didn\u0027t include a workflow environment. This\n  patch fixes it.\n\nCloses-Bug: #1839840\nChange-Id: I3df711ef2484374418085fe0117fe8b37ce5ba3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f92a5c8f445f132c230f20429bda0d1889996a84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f92a5c8f445f132c230f20429bda0d1889996a84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
