)]}'
{"id":"openstack%2Fmistral~679560","triplet_id":"openstack%2Fmistral~master~Ib54940fdf3db7afff945f7c13d158cc04f5ac250","project":"openstack/mistral","branch":"master","topic":"eyalb/functional","hashtags":[],"change_id":"Ib54940fdf3db7afff945f7c13d158cc04f5ac250","subject":"remove unused gate code * post_test_hook. used in legacy gate * run_functional_test probably not used and also doean\u0027t work   since mistral_tempest was moved","status":"MERGED","created":"2019-09-01 08:21:46.000000000","updated":"2019-09-03 12:33:13.000000000","submitted":"2019-09-03 12:30:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679560-1567513835739-05248e82","meta_rev_id":"75080bb2e8ba40d831d41cae053defd0a7882907","_number":679560,"virtual_id_number":679560,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-09-03 12:30:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-03 06:23:48.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-09-03 05:06:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-09-03 10:38:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 10:38:16.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-09-03 12:30:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e67267b49826b16b47fcd124a44ce5d95bd14934","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-01 08:21:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60a89067b320490b3bd3f0a8e1a75b8d709f8fc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 09:26:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6857a1891de74ff4a48161154cc59fb7 : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/91ccfab4f84649719824052ecf245585 : SUCCESS in 9m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fac4b1598a44e0181e49047ca4c3ff7 : SUCCESS in 8m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a14382dd13da4f41a574ff32354cb8de : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a06db8a01074672bbb3ead18e8e6957 : SUCCESS in 5m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc87c72c9d49446c97e87038e705bd3e : SUCCESS in 9m 28s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/650a83475d3b4c70858d2218db3b7c17 : SUCCESS in 41m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/746f38c85749438b95194a7432e8ff09 : SUCCESS in 39m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ec5abe5ea77249beb44559a80e7b0938 : SUCCESS in 42m 31s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d748683071f841febf53fdc1c405398e : SUCCESS in 29m 37s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ceca3f9a03da462db1a8a932f38dc8bc : SUCCESS in 26m 47s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a424a87b902a445a8d87b2dd1fe2901b : FAILURE in 17m 03s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/66899ee9600b4c02ab5b13468fcff15a : FAILURE in 12m 12s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7accca47c3354602881135316e9a02df : SUCCESS in 1h 03m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7ed7adb03744f3b9a5385c2b7b461fd : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":1},{"id":"79ba8e3a373becd022d475be5cca902df82320ef","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-03 05:06:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e9ecce0e143f1d547459525c175959b33273f42c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-03 06:23:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8042bed75687c16c8238f5c9cd5db3216004f9fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 09:32:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7ba213d4377489c9ff88f3b5e4728b8 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c50be36c009d4dd9abb1aaaffa83377c : SUCCESS in 9m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03e5c151b3d448858072c74bae6b105a : SUCCESS in 10m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1630ae4f3d624236894b20478ab1155d : SUCCESS in 10m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e5af4e1a3b403c842e8b02234e2463 : SUCCESS in 6m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/986d5b0b2ffe4daf830499c3311b845c : SUCCESS in 11m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7036c99030df44aea9382c153b047893 : SUCCESS in 46m 55s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/073551ee1e0a407a8c7122f59301a717 : SUCCESS in 47m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0f548c7fe6724669aadcd8d3aa0a9f7e : SUCCESS in 38m 15s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2045fbdd40394c81a5f74e4bc12ccbaa : SUCCESS in 28m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f7417a03a8b6401b96cf68ca37d0f763 : SUCCESS in 26m 26s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/329930d641f74ec984bd3a7c98bfe0bc : SUCCESS in 1h 33m 56s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8e1f7abb463c40b2a0803e1b85976bbb : SUCCESS in 2h 45m 13s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ab34b599758444549bc3dab5806951e6 : SUCCESS in 1h 09m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d459d55b641a48da888d1e8acaef1b16 : SUCCESS in 11m 11s","accounts_in_message":[],"_revision_number":1},{"id":"51fed341d9934da662b665b2f439761fd2b1f631","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-03 10:38:16.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"79fed04a28ee91d6e0d7bbd14433bf60dc0f3ce2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 10:38:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2eb9fab6bc6e33b139d50b497447812979a1891d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 12:30:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8483b2e2cb5846dca83eb7ca786cb7cc : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc7136e0e5694242b1a965b9f20570cc : SUCCESS in 11m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3c426acdd4c463db56ab53691ddab2e : SUCCESS in 8m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc06a0315e70418198dea7f7b9e1151d : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bca5257c1ae24e2e8564e9e0defc4a97 : SUCCESS in 9m 03s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/476de8e8b9de45d38bb31f607f9e261c : SUCCESS in 45m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6dc8876b2ccb4a0f89a4445bd611acc4 : SUCCESS in 43m 13s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f930888b8c7a40c49e6d3c0095bfe3cf : SUCCESS in 27m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b4ec1f1b60e44430ab0fa12ff0aec9cf : SUCCESS in 28m 10s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/80557ff981de4e4292ef8a0068acccc7 : SUCCESS in 1h 38m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4290877279fd47b3a8687c5035d1d617 : SUCCESS in 57m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/802d42c2991942219754668506a6df38 : SUCCESS in 10m 40s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/5d1c1c9c44bd4c47adb92ec672d83793 : SUCCESS in 1h 43m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7d07797bb007657aa7f2510d05ea7d155cc288dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 12:30:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d952af7a37c92d3d94e49850838124b7c6100f0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 12:33:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/015cd2e263d94b1b995c80a4f8591b3e : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d53992116e268badfbd266761ceb1869b41f175","revisions":{"8d53992116e268badfbd266761ceb1869b41f175":{"kind":"REWORK","_number":1,"created":"2019-09-01 08:21:46.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/60/679560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/679560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/679560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/679560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/679560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/679560/1"}}},"commit":{"parents":[{"commit":"2cdcb5415ead23a129d2e6698833ce10e6a6cf4a","subject":"Merge \"Improve new scheduler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2cdcb5415ead23a129d2e6698833ce10e6a6cf4a"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-09-01 08:21:21.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-09-01 08:21:21.000000000","tz":180},"subject":"remove unused gate code * post_test_hook. used in legacy gate * run_functional_test probably not used and also doean\u0027t work   since mistral_tempest was moved","message":"remove unused gate code\n* post_test_hook. used in legacy gate\n* run_functional_test probably not used and also doean\u0027t work\n  since mistral_tempest was moved\n\nChange-Id: Ib54940fdf3db7afff945f7c13d158cc04f5ac250\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8d53992116e268badfbd266761ceb1869b41f175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8d53992116e268badfbd266761ceb1869b41f175"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
