)]}'
{"id":"openstack%2Fmistral~679579","triplet_id":"openstack%2Fmistral~master~I561736d522033dad841fbcbb03645fff880c9df2","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I561736d522033dad841fbcbb03645fff880c9df2","subject":"Bump lower constraint of python-zunclient","status":"MERGED","created":"2019-09-01 22:38:05.000000000","updated":"2019-09-03 08:31:00.000000000","submitted":"2019-09-03 08:29:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679579-1567499361141-610ae198","meta_rev_id":"37ca50afe284fc69537f2f24df3025ec64e88e07","_number":679579,"virtual_id_number":679579,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-09-03 08:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":1,"date":"2019-09-03 08:00:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2019-09-03 05:03:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-03 08:28:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2019-09-03 07:35:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 07:35:58.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-09-03 08:00:19.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-09-03 08:28:03.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-09-03 08:29:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"95057d1e0a26bf465f5ea026024c0cfa8bff50dc","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-01 22:38:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3e9d826a97babe51eff1136529767ca67cd870b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 23:53:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0c97c5012b84364844d9f4073180ca4 : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/598d4d95c1c54b249bd9849e70fc5ac6 : SUCCESS in 8m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e86676b787d43fc8319ddcd68565b1e : SUCCESS in 8m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92d729f6258a4315803c30fad9c08d14 : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b67dc81bb4641b69597f1274befd3c7 : SUCCESS in 4m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a02f88ceaef44b8895f699d330fe36c1 : SUCCESS in 4m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b9046b9166c4ec4894458cb93bc3201 : SUCCESS in 9m 20s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f802c769fafd44a3b1f5b5c51646e8f5 : SUCCESS in 43m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/13448374404041e7b8217c70a43696f4 : SUCCESS in 41m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b43566e433734c8f9adeff748afa235f : SUCCESS in 42m 52s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c41dc9dfce274f77903a5e38fb22c869 : FAILURE in 27m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4a1cd3de61c94b3a980bcdc3fb3cef33 : SUCCESS in 27m 04s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/21ebe42afff7464184dda5eeed7c5644 : SUCCESS in 1h 09m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bb99168952146a5bc056085bacda594 : SUCCESS in 9m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ce8a207688e586db7e5656e050401ee4e4904da4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-03 05:03:57.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ac0857ea9a5d3c853f73dac07e25f766f367f127","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 06:16:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89f832844b8d40149bcef5e268b642df : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/03bb1b363b18497296a87942223b4380 : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b16429b988e445f18dc54cf63daba2bb : SUCCESS in 10m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eac16a11426f4de08b69295bbbeab7cf : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57a78c4a537e4e2497a21be8e75b0123 : SUCCESS in 5m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3da9cae816f54ed6802d1f7bbad0bd3a : SUCCESS in 4m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33e3e7c35d9e4e0f89c6e83e075426c8 : SUCCESS in 9m 33s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/916c760d8ae84083a059bd46a8e31a78 : SUCCESS in 40m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2fe4af4afabe4994a4e8ee0201d34a23 : SUCCESS in 47m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0ed13ee31e06498bb5337df35f1c68b4 : SUCCESS in 41m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c91b9bf4d75e41c5871d5779615ab608 : SUCCESS in 31m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/92449b13aba54bf68debf97f9498977b : SUCCESS in 26m 38s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9310706631ac46d1a0e86b87f5681662 : SUCCESS in 1h 07m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d967160b8ac4db6adf086e841fcb8a5 : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"845eacd5e0297aab30355e73b25744b255a83c19","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-03 07:35:58.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4132b6a2d8c7014f1e90554309c373754220a0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 07:36:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12671934403d064a08032c177df4c94b324368bf","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-09-03 08:00:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a9d4676ce235ac17b8f68e17c9fce37bea7e40a5","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-03 08:28:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a1ef2c49d934ba67c07f9536463d957a74a9b012","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 08:29:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfb8b7f6261c4603a17f34e2e721bcd7 : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46429161d9814b559590bc1b1852fb47 : SUCCESS in 9m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/858049b7e8614032b101e9fe439f617c : SUCCESS in 8m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25e4755beb2548c4ad0e87c23a2601c2 : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fa72c38a70d43128a3fe9e8a99b724b : SUCCESS in 5m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b4f33bdace5f4e588337c8ab23650d16 : SUCCESS in 3m 59s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/65470778a22f4291bd0fbcb769404478 : SUCCESS in 47m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/440d1c9a64ee4479a065b4c3b980d797 : SUCCESS in 45m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/aa764f96dc594fafb9f47b51f94ab173 : SUCCESS in 27m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3080ec3648f44712b4050b091ac9c764 : SUCCESS in 28m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/66db7a84c3ca45fe83b6b87a8aa43491 : SUCCESS in 47m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f52edefcc90450a967a486e2787d978 : SUCCESS in 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d1019e32bd60df0bf7aac5af5a5eb988c7f6b881","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 08:29:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"37ca50afe284fc69537f2f24df3025ec64e88e07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 08:31:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81126e0adb7a452ba1bc004a75674f67 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60536f6912de7ba1542d02405e472c733c75c8e5","revisions":{"60536f6912de7ba1542d02405e472c733c75c8e5":{"kind":"REWORK","_number":1,"created":"2019-09-01 22:38:05.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/79/679579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/79/679579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/79/679579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/79/679579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/79/679579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/79/679579/1"}}},"commit":{"parents":[{"commit":"2cdcb5415ead23a129d2e6698833ce10e6a6cf4a","subject":"Merge \"Improve new scheduler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2cdcb5415ead23a129d2e6698833ce10e6a6cf4a"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-01 22:37:40.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-01 22:38:04.000000000","tz":0},"subject":"Bump lower constraint of python-zunclient","message":"Bump lower constraint of python-zunclient\n\nProjects that depends on python-zunclient should use the latest\nversion as lower constraint.\n\nChange-Id: I561736d522033dad841fbcbb03645fff880c9df2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/60536f6912de7ba1542d02405e472c733c75c8e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/60536f6912de7ba1542d02405e472c733c75c8e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
