)]}'
{"id":"openstack%2Fmistral~680330","triplet_id":"openstack%2Fmistral~master~If68d125154e13390d512a8f0ef29caf121bf412f","project":"openstack/mistral","branch":"master","topic":"eventlet","hashtags":[],"change_id":"If68d125154e13390d512a8f0ef29caf121bf412f","subject":"Blacklist eventlet 0.21.0,0.23.0,0.25.0","status":"MERGED","created":"2019-09-05 08:38:20.000000000","updated":"2019-09-06 06:27:30.000000000","submitted":"2019-09-06 06:26:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680330-1567751175172-707fe744","meta_rev_id":"89ee5799b70d330fdb7867b334df42eccd6101bf","_number":680330,"virtual_id_number":680330,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-09-06 06:26:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"all":[{"value":1,"date":"2019-09-05 09:59:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-09-05 10:15:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-05 10:15:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":1,"date":"2019-09-06 05:04:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-05 09:59:34.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-09-05 10:15:33.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-09-06 05:04:02.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-09-06 06:26:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"480ab7381db2f5aa4989d27a57b7b14f82225c66","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-05 08:38:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46fea0801a93722dfef539ff254e290a905f3043","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-09-05 09:16:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5b445680367b72f8861ac219d19617e57cd2e50","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-05 09:38:27.000000000","message":"Patch Set 1: Code-Review-1\n\nagree","accounts_in_message":[],"_revision_number":1},{"id":"8f685b75fae765feec04709fbc8d80a5bbc9d1cc","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-05 09:47:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"56d8b3c3907bd1663644ed3688efd39ded81259f","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-09-05 09:59:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0b46d1b6a3ffc71dd0d5bcf95c8b3126e510e3e0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-05 10:15:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"460bb97c9f4538a3f479944ee315b16e0c5e6e2a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-05 10:15:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"165ae3c8886a6f4663310c761680db1ff35a964c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 11:56:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0e8ed8ef688457fa1759d4c943da396 : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/accaf345195642a3b76db560d839d52f : SUCCESS in 10m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4b31185f45e47228bcbda8072c01af7 : SUCCESS in 9m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20a644d7b31345daa1bdf45457ff3e6a : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c4857138c04749bce1f562bcb89625 : SUCCESS in 6m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5d7df32c98614cc2a31465d1708a16e1 : SUCCESS in 3m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c546323a83af4ef0b8fecace1550e3fc : SUCCESS in 12m 40s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/712f722177e14a5baaddd6dcc4f93cfe : SUCCESS in 48m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/91d5f1b73ccf405b91f2720fd882b9bd : SUCCESS in 38m 06s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/78235e6188064375b7ac79a5da1d2f4e : SUCCESS in 49m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9235797aa01b48a29db7400e0aac641a : SUCCESS in 28m 33s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/aadb01c58f734b35a4b399530ed035ef : SUCCESS in 27m 51s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1375aa3f4ddf41eeb40f29a9239c0c96 : SUCCESS in 56m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47cc69431d7040cd83367fcec7a6ee78 : SUCCESS in 12m 05s","accounts_in_message":[],"_revision_number":2},{"id":"05e5abd115f65642c6bc0ef19c3b4d856921edf3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-06 05:04:02.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"81af97cd73a3c2faa74bfe44754356804055363e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 05:04:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5e463f742cbdfc02968f12f06db5612158f81738","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 06:26:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6952a14b343243fab83cb92f192fcb69 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82b535ea69f544aea3f219ecee9ad136 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f852f5f433a24f42bc683a9fb6b04910 : SUCCESS in 7m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/11673bd7a3be43749d3d83a1c0f3629b : SUCCESS in 12m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bc0677d2c8a4178b05a6c5266d85bbb : SUCCESS in 7m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aaa40e9aad4e4fae8cfd8826fa766c1e : SUCCESS in 2m 31s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d2a6f3be034e49c4aefe63d723c1ee3b : SUCCESS in 41m 38s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/537ed0a0212d4c2290f8acf79657eecc : SUCCESS in 45m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/42ea2fad40eb412b826885ae4f331262 : SUCCESS in 27m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c924572263694d2f9b9b1a146cdaaf5b : SUCCESS in 26m 01s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/807728c6c3e94e4785f9d622686f9738 : SUCCESS in 42m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8c974e6bf23431a8701f4112f9d373b : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":2},{"id":"26f3568bec1c40658d7d72653e28698e0bb8e032","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 06:26:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5afcf3399f087d456afb0db0d692f7de33e18820","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 06:27:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b25543ebeeae47e9ae8df1ac28d78343 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6a5842990e4d72dd728630552f14ec8b6a65c944","revisions":{"045dffd484cc9fe55457aaea765c6c7ed63fd8c5":{"kind":"REWORK","_number":1,"created":"2019-09-05 08:38:20.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/30/680330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/30/680330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/680330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/680330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/680330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/30/680330/1"}}},"commit":{"parents":[{"commit":"a0c26e3054f9515ff4928c5a908e320574d1bec3","subject":"Merge \"Mistral don\u0027t work with cron trigger\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a0c26e3054f9515ff4928c5a908e320574d1bec3"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-05 08:37:50.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-05 08:37:53.000000000","tz":480},"subject":"Blacklist eventlet 0.25.0","message":"Blacklist eventlet 0.25.0\n\nmore details https://review.opendev.org/#/c/678078/\n\nChange-Id: If68d125154e13390d512a8f0ef29caf121bf412f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/045dffd484cc9fe55457aaea765c6c7ed63fd8c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/045dffd484cc9fe55457aaea765c6c7ed63fd8c5"}]},"branch":"refs/heads/master"},"6a5842990e4d72dd728630552f14ec8b6a65c944":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-05 09:47:45.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/30/680330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/30/680330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/680330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/680330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/680330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/30/680330/2"}}},"commit":{"parents":[{"commit":"a0c26e3054f9515ff4928c5a908e320574d1bec3","subject":"Merge \"Mistral don\u0027t work with cron trigger\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a0c26e3054f9515ff4928c5a908e320574d1bec3"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-05 08:37:50.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-05 09:43:28.000000000","tz":480},"subject":"Blacklist eventlet 0.21.0,0.23.0,0.25.0","message":"Blacklist eventlet 0.21.0,0.23.0,0.25.0\n\nmore details https://review.opendev.org/#/c/678078/\n\nChange-Id: If68d125154e13390d512a8f0ef29caf121bf412f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a5842990e4d72dd728630552f14ec8b6a65c944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a5842990e4d72dd728630552f14ec8b6a65c944"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
