)]}'
{"id":"openstack%2Fmistral~680860","triplet_id":"openstack%2Fmistral~master~I294d18b341a3c7dd0df9c24588540f9c94dd4562","project":"openstack/mistral","branch":"master","topic":"bug/1843178","hashtags":[],"change_id":"I294d18b341a3c7dd0df9c24588540f9c94dd4562","subject":"Fix don\u0027t work with senlin actions","status":"MERGED","created":"2019-09-08 14:28:37.000000000","updated":"2019-09-12 22:38:45.000000000","submitted":"2019-09-12 22:37:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680860-1568327847782-4596321f","meta_rev_id":"3db69a7c940e950b87490b17853a894791e0805e","_number":680860,"virtual_id_number":680860,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-12 16:39:21.000000000","_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-09-12 22:37:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":1,"date":"2019-09-12 06:22:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2019-09-12 06:22:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2019-09-12 06:22:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-09 12:35:05.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-09-10 14:09:54.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-09-12 06:22:58.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-09-12 22:37:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d9a39030256cc4e531f9b28d485297a7245eba30","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-09-08 14:28:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8db64a1963422d7190acf5600554ef63ba4088b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-08 17:35:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f4733cb6d64fc4a6bd92b36f035d70 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08aa0508f6e14057a5a237b14972d42c : FAILURE in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b43072ad96041ac9a1258e2e60ee7ef : FAILURE in 7m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0fdadf19de9a43faa9b8a9cb9143d749 : FAILURE in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/636b00b48e164b3aa4e0c47dbf12b8c8 : SUCCESS in 4m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/309700f80e01447c9e03e5bbd370325b : FAILURE in 8m 31s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fb2db2c60c504bef999c9aa9725ceba8 : SUCCESS in 42m 09s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d12125a84e1346c0bb7622029d3bc0d3 : SUCCESS in 41m 59s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e62dd5a45f2040a4bb788464151dbdae : SUCCESS in 41m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a79e9c89d76945dabbaf447f0da61cc0 : FAILURE in 27m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/edf7215598c6429dbe6a49591e9f1951 : FAILURE in 26m 17s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1f8940aa2cf84c3387afab4244887b08 : SUCCESS in 1h 38m 48s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/853fcd3226904cbca539e8c149a73a61 : SUCCESS in 3h 01m 55s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e6bf1cf8dac64c2aba9db4d3d7100b88 : FAILURE in 34m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f0b33b81b00450eb1eb66c7604175b2 : SUCCESS in 9m 44s","accounts_in_message":[],"_revision_number":1},{"id":"cee67aa347ac8370b95d332280e72c691e3080ad","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-09-09 03:31:51.000000000","message":"Patch Set 1:\n\nplease help me check error in zuul. Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"c0a796cf141569c7658184dabd92173a9a37cb81","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-09 12:35:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5874305a1028e17b9bba85bbadd4ce176d4f293c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 15:43:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fc950ebb64f4f498d65748c39da0369 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3e988ae3a11416382d4009e11483442 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92c16e8b59ab4570a76f554321bb65b1 : SUCCESS in 8m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e22f02277e04cef85cace44a4687560 : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cbac4e0a659434b891ac9b384aff8be : SUCCESS in 4m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3735733df4dc4303a23dedb5ba2d15cb : FAILURE in 8m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/486563cc9cde448cb78698ea684aa83f : SUCCESS in 44m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/26ffb246ce5f4c59a791f4d2522523a8 : SUCCESS in 42m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/80a06b9f3dcc48feb73b8c9cf5143dd5 : SUCCESS in 44m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8a39096ec13946a5a848a63857060e0b : SUCCESS in 26m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/98c5f67041684281bf31bc3125d0e3e4 : SUCCESS in 27m 23s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c2bcf41a3beb496da7d31dee26cff622 : SUCCESS in 1h 37m 40s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/cf824185d3804f3d8522d7a4b7b3d68c : SUCCESS in 2h 52m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7c21a9fd6a73423eb3b04576ad6a626d : FAILURE in 34m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/378c4f9fde234d88b2342eb6fd5857e1 : SUCCESS in 10m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e2a01d97c8c76240c65108eca5be6b34a680e752","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-10 04:50:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nLooks ok, just one style comment but I\u0027d like you to address it.","accounts_in_message":[],"_revision_number":1},{"id":"e7c421beaf52b962640aaebd43418e446206ec76","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-09-10 05:26:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eb5f99453fb912bae6f82e5f5949fe855736c5b5","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-10 07:22:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fe5bf435b277b9bdbfd541963ecc4271a6b098cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 08:44:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6d1437c0c80401f87517ad0c074caf9 : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/efc4fe2b880245ee8317f7613d5222e7 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8aa3a05a4634d6f84b6ddd9773ad090 : SUCCESS in 7m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87a383625e9b40fea0a7235991917aa4 : SUCCESS in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b935785a52e4a4f8be8623d666bc301 : SUCCESS in 5m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a6009a480a64d96ae48d2ac4975defc : SUCCESS in 8m 26s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b5b01f57ee364db3a14056fc986d29fc : SUCCESS in 40m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8d29e6a78215418da6b8f78411eab74d : SUCCESS in 40m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8d837b6157614bb7b19728ada1b22542 : SUCCESS in 41m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7317b6371798417f9c04f3ea4ab47846 : SUCCESS in 28m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/685358f077a64c5ea4669c28f5c84803 : SUCCESS in 30m 01s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b4078d2ffd4b4d6288eebfb3f1bedb33 : SUCCESS in 1h 28m 04s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6ecddc8e07784faaa8b91aac82444b89 : SUCCESS in 2h 52m 25s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c710d3103c8f4737ad86b71c733d25fe : FAILURE in 38m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b920a7715bc469ea96ec2f3229a8d3e : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a0f50244e21bb2d6b1c206f068acea89634546bd","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-09-10 14:09:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0e088592ad8aefa64ace56bc2e185f1d00f8ad50","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-12 04:18:31.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bcd19c556493187777f4d2796fe8b1975f98ed27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 04:18:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"76f3990b4d52a8ee0c748df41da3f9108310810a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 06:18:21.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8b3b2327f5141e182a7f6035c8ebe75 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1e264d1a84a499caa7eb40520ec43b0 : FAILURE in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1976752994745868dc585e2436a0f75 : SUCCESS in 8m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49adead437314c788cf2652eea987211 : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a02c8fe9046443c8a5b4862c273f8d53 : SUCCESS in 5m 34s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/84dc9847417b4ee686ffc204c4fd07f7 : SUCCESS in 37m 51s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c9958b7b94bb412c9ea6dfe6d1527917 : SUCCESS in 42m 31s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/72ff50257c20467d8d321356c479679a : SUCCESS in 30m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4e530558eb77429782bae223aadd11c8 : SUCCESS in 28m 20s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/9d3fa08d29584d30a5074a41ab804d6a : SUCCESS in 1h 39m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/07d5b2e25c71434786750cb2b70df900 : SUCCESS in 36m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3613c9d7269246f493a7108e26c9c80f : SUCCESS in 9m 15s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4b0bf572222848589785d85d652a8e48 : FAILURE in 1h 43m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8ccf8d2b29b26e994e53111c7c0f70437b857956","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-12 06:22:51.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e5ab69b3a77d4431c807f63e493e88b1665ddd7e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-12 06:22:58.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"87e41c6168531adb8e16276128acd24b388631ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 09:27:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b6a8873d8b14d23af2b023ad530258d : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7da7045bbc9147fc94fb80afa4cee2c0 : SUCCESS in 9m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5b14d27c54c425197ed12fb4c1e07c5 : SUCCESS in 9m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9335eb99df24aeeb07ccd16538bfcd6 : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0231f33004b44f699a8d5ece0ca4db5 : SUCCESS in 4m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77cc35902b024d4a85092f33778adec6 : SUCCESS in 13m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/74d4a63fadf44ac79155fd215ec8a6e4 : SUCCESS in 41m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/96b7274f26be45d7b7b26dd70b925ca5 : SUCCESS in 42m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/907658edcd7b4fd1aab26777f72f5b5d : SUCCESS in 42m 39s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/03c8283674744485acba29705a46deb1 : SUCCESS in 28m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1dcb85725cd94ae68965e5bfcdb47e29 : SUCCESS in 28m 28s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a43e0fc03f4343449e37fc312aa0e03a : SUCCESS in 1h 26m 45s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/269f00b0f4e848568ee28f83f946189a : FAILURE in 2h 45m 31s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2c75bd8b2ec246f18ddfd405f268a34f : FAILURE in 36m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebf01a568bbe4d8284aa0acf425abaa6 : SUCCESS in 12m 18s","accounts_in_message":[],"_revision_number":3},{"id":"885baa31e42586ebd3e5af2a6c407fca8a73484a","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-09-12 16:39:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"51d13537587de30a1b39a52a9ed6dbe3c0d20f5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 19:48:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad6496d825a64d2dbc4c6247df628331 : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bbca7d38eb8f40f1a0022a0216dfc6b6 : SUCCESS in 9m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d2d75f5738c40a59b9440a91a19a915 : SUCCESS in 8m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f4365cc87fb49de81af5954764dc8c2 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e585068919f34883a41ea9d3b15dcc7f : SUCCESS in 5m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f06d511b5b04d959a03310c01ff27ad : SUCCESS in 9m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/38af1c6eec62473199b380b2cee3f378 : SUCCESS in 41m 42s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ed56333355dc4f42b11c245204a514e8 : SUCCESS in 45m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a8ee6dd3c6fd4e3a9de44c1f5c8634c3 : SUCCESS in 42m 38s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f5610f41f1064d64956e974cb6fc8ff6 : SUCCESS in 28m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/016b32b3566f40a4a3db28abae4893c9 : SUCCESS in 31m 41s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/74853b22eb844269995d81c59faf768e : SUCCESS in 2h 03m 49s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/660c52a14f944eb3aedeba4fea014903 : SUCCESS in 2h 55m 03s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7ebd5a496c904eceaba431cc29f17326 : FAILURE in 31m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/329234555e024e6eaabb766812e24d69 : SUCCESS in 10m 01s","accounts_in_message":[],"_revision_number":3},{"id":"04fe9cdf4e168eab4e800c05a3157e06637a487d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 19:48:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"617088bbe087cae080fd3b737b4eefe366926af3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 22:37:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c654a17cce1babf25b07cb992845805a6415207c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 22:37:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d67d8233e7ac47ae85fac3a2307e169c : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9fe86ac6c95b4cdba3afd73f522de1d4 : SUCCESS in 9m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce49237fa0f6430fa3cd4cb4ce706b67 : SUCCESS in 10m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea13c2032f254e3d80dcfa67977f0df5 : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03bb354677114a9382536ae078d1201d : SUCCESS in 3m 57s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/01a4565a65fd4c628dc26272a5dc3f10 : SUCCESS in 43m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/36acd465c6b74985ab38c4d76dc87277 : SUCCESS in 57m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/db2ecf2daa2546d7ac50f879429db970 : SUCCESS in 29m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d725a3d5906144d99d344935c18da9cd : SUCCESS in 26m 54s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/cfeeafef172746cc9377ddf16c92cf54 : SUCCESS in 1h 37m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ce115670509549558970819bc51d2381 : SUCCESS in 42m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1801697688240fe9dd160652dbc075f : SUCCESS in 9m 27s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0448bbaf144d4974b7e4a5b782036f09 : SUCCESS in 1h 48m 42s","accounts_in_message":[],"_revision_number":3},{"id":"9937b2c7622fdfe324554c8e5e62fb69a0de4d3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 22:38:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8922755df2c14eaa9ae6ac292b59d659 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9585a63847034de2117db0ea08ae3c4c0da40a60","revisions":{"76d03550fd7e17137352c81c970c1398ca9e0f22":{"kind":"REWORK","_number":1,"created":"2019-09-08 14:28:37.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/60/680860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/680860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/680860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/680860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/680860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/680860/1"}}},"commit":{"parents":[{"commit":"b8a8aad6452f236aec37de04fbbaf93bb3744bbf","subject":"Merge \"Fix misspell word\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b8a8aad6452f236aec37de04fbbaf93bb3744bbf"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-09-08 14:25:17.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-09-08 14:25:17.000000000","tz":420},"subject":"Fix don\u0027t work with senlin actions","message":"Fix don\u0027t work with senlin actions\n\nWe can\u0027t do run senlin actions because have an error when\ninit client senlin. We need an other way to init client to\nrun client with cron trigger and manual.\n\nChange-Id: I294d18b341a3c7dd0df9c24588540f9c94dd4562\nCloses-Bug: #1843178\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/76d03550fd7e17137352c81c970c1398ca9e0f22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/76d03550fd7e17137352c81c970c1398ca9e0f22"}]},"branch":"refs/heads/master"},"5e7386fc60f2b5364ea0d14fb845ccfb834d14bf":{"kind":"REWORK","_number":2,"created":"2019-09-10 05:26:24.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/60/680860/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/680860/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/680860/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/680860/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/680860/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/680860/2"}}},"commit":{"parents":[{"commit":"b8a8aad6452f236aec37de04fbbaf93bb3744bbf","subject":"Merge \"Fix misspell word\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b8a8aad6452f236aec37de04fbbaf93bb3744bbf"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-09-08 14:25:17.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-09-10 05:26:05.000000000","tz":420},"subject":"Fix don\u0027t work with senlin actions","message":"Fix don\u0027t work with senlin actions\n\nWe can\u0027t do run senlin actions because have an error when\ninit client senlin. We need an other way to init client to\nrun client with cron trigger and manual.\n\nChange-Id: I294d18b341a3c7dd0df9c24588540f9c94dd4562\nCloses-Bug: #1843178\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5e7386fc60f2b5364ea0d14fb845ccfb834d14bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5e7386fc60f2b5364ea0d14fb845ccfb834d14bf"}]},"branch":"refs/heads/master"},"9585a63847034de2117db0ea08ae3c4c0da40a60":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-12 06:22:51.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/60/680860/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/680860/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/680860/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/680860/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/680860/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/680860/3"}}},"commit":{"parents":[{"commit":"7698c3def052f78d67ac9276fc0567a9291d5510","subject":"Merge \"Add db api tests for scheduled jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7698c3def052f78d67ac9276fc0567a9291d5510"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-09-08 14:25:17.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-12 06:22:51.000000000","tz":0},"subject":"Fix don\u0027t work with senlin actions","message":"Fix don\u0027t work with senlin actions\n\nWe can\u0027t do run senlin actions because have an error when\ninit client senlin. We need an other way to init client to\nrun client with cron trigger and manual.\n\nChange-Id: I294d18b341a3c7dd0df9c24588540f9c94dd4562\nCloses-Bug: #1843178\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9585a63847034de2117db0ea08ae3c4c0da40a60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9585a63847034de2117db0ea08ae3c4c0da40a60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
