)]}'
{"id":"openstack%2Fmistral~680930","triplet_id":"openstack%2Fmistral~master~Ifc36f4a1f6126d6cbde2424fd88967c04d84185a","project":"openstack/mistral","branch":"master","topic":"fix_lang_tests","hashtags":[],"change_id":"Ifc36f4a1f6126d6cbde2424fd88967c04d84185a","subject":"Fix workflow language tests after updating jsonschema to 3.0.2","status":"MERGED","created":"2019-09-09 07:36:36.000000000","updated":"2019-09-09 12:33:43.000000000","submitted":"2019-09-09 12:32:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680930-1568032344755-61dee02a","meta_rev_id":"5b41935a9f68131d47d36e4272b70acd7dbf71b7","_number":680930,"virtual_id_number":680930,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-09-09 12:32:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":2,"date":"2019-09-09 07:59:47.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-09-09 08:04:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":1,"date":"2019-09-09 09:24:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2019-09-09 08:29:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-09 07:37:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2019-09-09 08:00:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-09 07:37:44.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-09-09 07:59:47.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-09-09 08:04:17.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2019-09-09 08:29:43.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-09-09 09:24:20.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2019-09-09 12:32:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b99cdb32a63477760db17b704788e5094eef0d8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-09 07:36:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ae0e069b31b8fb544ac771c2092c7011c05f5b5","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-09 07:37:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a025fbaafcc86cf5885e77b4cd4e461e36e2bdaf","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-09-09 07:59:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"da420c5ae89ab19dae61f33071c681210f398eac","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-09 08:00:20.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"21b2009a8ae00c0ab59218bb593d7d48cf0b362b","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-09-09 08:04:17.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef0577169df374a8cd4ec9cb9d4f9061d678e42c","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-09-09 08:29:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f0e50f4e341eb103801a52ab719f81371e7ec80e","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-09-09 09:24:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a4682f3ddb334e8a41cd90075ceeffae4cc40662","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 10:38:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2e1b76a5123497da3a3d4f6a0e9182d : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d6ce1a57cc1a4d3ab9a02d5f1afb5920 : SUCCESS in 9m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3a1deecbe6e4797ad290df402f742b2 : SUCCESS in 11m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3b096fec7614ec28f8bbe2ec6f40f80 : SUCCESS in 13m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c2b939a3c2744c992d1b5fcbf8d3c13 : SUCCESS in 7m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a3ad57ff5cb45d6a9cfd02ac34e9df1 : SUCCESS in 8m 21s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ed53e67c48ea429495de32ae15e4f429 : SUCCESS in 46m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7883a465ea75463da520289e47fa53da : SUCCESS in 43m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3e8ec96ce6944d598949003a4dd8c377 : SUCCESS in 42m 01s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/22a5886060b946a0ad4307753cc9067e : SUCCESS in 28m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4e9e83f676944493bc89931cb8a5ab50 : SUCCESS in 31m 40s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/19167ef8c7e54d1f825ca2aebfd1ddfd : SUCCESS in 1h 32m 41s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/eb6fa1e9965c4effb56d61aa79fa0dd7 : SUCCESS in 2h 55m 28s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6ab7a4dc601e49aba096425f3d8b0d35 : FAILURE in 33m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/878a3b418c2146f993175c7c25fc0c09 : SUCCESS in 12m 50s","accounts_in_message":[],"_revision_number":1},{"id":"dec1d8433cc5080547596f54eda9d6da7f0b7500","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 10:38:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"222dafe9a7940decf9090247bb14021a13f5870e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 12:32:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dde501687d03e89c6bf577aa2e95c32e9832d4e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 12:32:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61fdd13d341a4d9eb3f26c675582aa91 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84616d0742d544ec935e3c7f72052afe : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b238eb885f154061badb1ce049dc21e8 : SUCCESS in 8m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/133b75203fdd40dda93516a2ea95381a : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fff8bf5e8ce41548dbe70e430bafe0e : SUCCESS in 4m 34s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/23c543cc5f364f34a555ade3995b87c8 : SUCCESS in 50m 54s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cf87fbb2550f496987ac66af97520e4f : SUCCESS in 46m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8a6732902b60408dafa768e5b770d4db : SUCCESS in 28m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/55d96ebf80d0417d8ff0a526348ffe24 : SUCCESS in 29m 38s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b744f8b2e3b74b1a918481c729178108 : SUCCESS in 1h 32m 03s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bf3fe8eadc1c43608d07cf398d4acc5d : SUCCESS in 47m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/686263372b4a409682b535123662bb1a : SUCCESS in 10m 12s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8550cd21b3704b32965b805f9c55367a : SUCCESS in 1h 48m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a8b058acc9cc2588082489eaee2d890859eb83b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 12:33:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed4d73ed34db4500a48b2c2e042db900 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fdfb65a8ffd2fb0a539a880961e2daa197be7dc7","revisions":{"fdfb65a8ffd2fb0a539a880961e2daa197be7dc7":{"kind":"REWORK","_number":1,"created":"2019-09-09 07:36:36.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/30/680930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/30/680930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/680930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/680930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/680930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/30/680930/1"}}},"commit":{"parents":[{"commit":"b8a8aad6452f236aec37de04fbbaf93bb3744bbf","subject":"Merge \"Fix misspell word\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b8a8aad6452f236aec37de04fbbaf93bb3744bbf"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-09 07:31:51.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-09 07:31:51.000000000","tz":420},"subject":"Fix workflow language tests after updating jsonschema to 3.0.2","message":"Fix workflow language tests after updating jsonschema to 3.0.2\n\n* The new version of jsonschema (3.0.2) considers numbers written\n  as XXX.0 (an integer but written using the form of float numbers)\n  integers. Whereas the previous versions didn\u0027t. So we have to\n  fix the tests that expected an error in this case when validating\n  the syntax of the policies.\n\nChange-Id: Ifc36f4a1f6126d6cbde2424fd88967c04d84185a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fdfb65a8ffd2fb0a539a880961e2daa197be7dc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fdfb65a8ffd2fb0a539a880961e2daa197be7dc7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
