)]}'
{"id":"openstack%2Fmistral~681011","triplet_id":"openstack%2Fmistral~master~I8201e3f89cf62a76e88521901b5b808f33bf3699","project":"openstack/mistral","branch":"master","topic":"eyalb/gate","hashtags":[],"change_id":"I8201e3f89cf62a76e88521901b5b808f33bf3699","subject":"Need to run only doc related jobs on doc change","status":"MERGED","created":"2019-09-09 13:46:01.000000000","updated":"2019-09-10 07:16:17.000000000","submitted":"2019-09-10 07:12:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681011-1568099569543-87eeb112","meta_rev_id":"774bcd539b2b8d30f1569b7a3dc823253968731f","_number":681011,"virtual_id_number":681011,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-09-10 07:12:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"all":[{"value":1,"date":"2019-09-10 06:06:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-09-10 04:56:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-09-10 04:56:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 04:56:44.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-09-10 06:06:39.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-09-10 07:12:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a820ea04e2440ab759bc4bf16749570ea494bdc9","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-09 13:46:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a543ba824d5ec512822edbcd57872d489a985f62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 16:56:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f365f75975e404b909c60993a237d07 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b2bddfababd44daaa50bef038e2fdca : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea46995f844443939bb40ee8a1d3325e : SUCCESS in 8m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f74b14e075974a5a9c34479046b000ee : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3189c7b3745543798f7b9b584bda02df : SUCCESS in 4m 56s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5e86166858fa4d158260f1c18e66ca38 : SUCCESS in 43m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/875aeff6dbe946f3a040b06b24e562ad : SUCCESS in 45m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b410d38142a043839cf8438df23e523e : SUCCESS in 46m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9425cf917f0c4831a90c384cae0a5eef : SUCCESS in 30m 32s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ff37c49992ee435499a8aeb8717b6967 : SUCCESS in 26m 40s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/7ea11ca99ee049a3b9cbf630f80ed3c1 : SUCCESS in 1h 52m 38s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d93614d5f1e84cbba3519d088fbfce0f : SUCCESS in 2h 56m 35s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/92279159a0d045bcbe704498eb91d24c : FAILURE in 33m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eaab14517be3456e8f8809b34ad2004f : SUCCESS in 13m 09s","accounts_in_message":[],"_revision_number":1},{"id":"771d5055bfc87c0b24a85d2575bb08dc76f4d27a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-10 04:56:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1f782a6bc5928aab2f23f186891140c546aa638a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 04:56:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cceff80d72b2a8b036e85fabe842283d23a7ed5e","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-09-10 06:06:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5999d20b7f2430b456fda1273d9c00e37c9e77e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 07:12:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40a3b2af57a3425e834884a3e311609a : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/98cf767044c34c61a155bb119c95d04b : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdc627a8b10a4700ac074a315e21199b : SUCCESS in 8m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6098a56020544f6ba21f8a8ffe606b2d : SUCCESS in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf09d49d6ce46519c0680fe368d98d5 : SUCCESS in 4m 42s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/69265685bfb74263bec739e5928b3e90 : SUCCESS in 41m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cb65ec3ff46f45c4a03dd712da9bf020 : SUCCESS in 41m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3829bbf882a744d9b7b868c9b24548b6 : SUCCESS in 27m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a18211440ae04933b144da76135742bb : SUCCESS in 25m 59s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/9820f44640834a4d97a3fa3ba2e4ba04 : SUCCESS in 1h 25m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7b4a7a23340640138ca693e3ca5930bd : SUCCESS in 41m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f9a0f26167b47b6a7ca24e78d93f9d3 : SUCCESS in 10m 29s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/492d898fe4ba49358627118909bd4cc0 : SUCCESS in 2h 10m 50s","accounts_in_message":[],"_revision_number":1},{"id":"77498270022374495da33f4e901f18035d9110c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 07:12:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0e4e4594460e256935767ad0152bb532541ef00c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 07:16:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f0c0a1d0e7547f7858d6944d5fa485b : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24f50dab2ff24c719978558d8838d9d12b6b9e7b","revisions":{"24f50dab2ff24c719978558d8838d9d12b6b9e7b":{"kind":"REWORK","_number":1,"created":"2019-09-09 13:46:01.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/11/681011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/681011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/681011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/681011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/681011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/681011/1"}}},"commit":{"parents":[{"commit":"fdfb65a8ffd2fb0a539a880961e2daa197be7dc7","subject":"Fix workflow language tests after updating jsonschema to 3.0.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fdfb65a8ffd2fb0a539a880961e2daa197be7dc7"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-09-09 13:44:54.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-09-09 13:44:54.000000000","tz":180},"subject":"Need to run only doc related jobs on doc change","message":"Need to run only doc related jobs on doc change\n\nChange-Id: I8201e3f89cf62a76e88521901b5b808f33bf3699\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/24f50dab2ff24c719978558d8838d9d12b6b9e7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/24f50dab2ff24c719978558d8838d9d12b6b9e7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
