)]}'
{"id":"openstack%2Fmistral~682024","triplet_id":"openstack%2Fmistral~master~Ia700e25752d9d35ece18609f2977e6568062e4bf","project":"openstack/mistral","branch":"master","topic":"bug/1844242","hashtags":[],"change_id":"Ia700e25752d9d35ece18609f2977e6568062e4bf","subject":"Optimize creation of language specs","status":"MERGED","created":"2019-09-13 11:11:49.000000000","updated":"2019-09-19 14:45:51.000000000","submitted":"2019-09-19 14:43:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682024-1568904220756-0d6172b9","meta_rev_id":"2b1c8c7791299b8e9fd1b3d84d4f700402c03bb1","_number":682024,"virtual_id_number":682024,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-09-19 14:43:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-19 08:23:09.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-09-19 05:03:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-09-19 05:03:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 04:35:51.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-09-16 04:35:51.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-09-16 04:35:51.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-09-16 08:28:02.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-09-17 11:02:24.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-09-19 08:23:09.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-09-19 14:43:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab7fb6c8c450f1a16babcd5e879ee077b33f47d2","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-13 11:11:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"facb84d89e73b9ad8a1c44673ec2d0cb1d3e3ffc","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-13 11:12:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"997a3024095d571de5c3e55a48ce7825606310c8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-13 11:13:04.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"009c006a337ae265d7386ad2622d64af5fb8b273","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-13 11:18:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f0ff682f164fdd3718ae384f1482bef5c726726a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-13 11:20:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b8590cb26eb51b1e92dcb5a5e0c0ea7c11625cdd","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-13 11:22:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"039d74cb4fc738dcd85d1522435f3c69fdec6968","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 14:31:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1872106f62df41a18d96df7017885462 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6af8f263972b4f2cb6411d86c9f46432 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fa6da3f57724797b4263f743646396d : SUCCESS in 8m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d67b1f78c3bf46dc8f9c9357fa2d2f39 : SUCCESS in 9m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df40dbd0b9ff4d7b9af784dd50b7739f : SUCCESS in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6ac419d03474548bdd80bde156946db : SUCCESS in 10m 26s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3d7fa3d6d4624535bc219a7ad013cdf8 : SUCCESS in 45m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b24f12492b9042f786f7a46317fdcb14 : SUCCESS in 48m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1068155f465f415d8b84d2467c6bc32a : SUCCESS in 46m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/29ada8a72c4a4a9fbc13e74a9d3433f6 : FAILURE in 30m 21s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/599d1ecd161549cf8df4b6f8dce646d5 : SUCCESS in 26m 48s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/24f20ed22e1c4512b27c0499fe098493 : SUCCESS in 1h 36m 18s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3c0ee26c602e426ab312aad506372201 : SUCCESS in 2h 59m 43s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/db1c916cc2784a2b970c1748faa2424b : FAILURE in 34m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8eee2e60e26b41ceb6a47ae5c158f649 : SUCCESS in 11m 29s","accounts_in_message":[],"_revision_number":6},{"id":"9206699b0f40f1f8e60a18a59a65fb88bf6c9a01","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-13 14:39:36.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"9bef63617b9cfd57b67d59fdd23a13c04726f780","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 18:02:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b059082dc63b49c2b0ebba764b7a01f8 : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e89e701e44f04140a4c6650522022d12 : SUCCESS in 8m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47e9cb510272437fbc2b01cc13fa23ce : SUCCESS in 7m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83215b24684043fb80e55cf1a8eed553 : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dc8874228104f1b8afb0aad746cebd4 : SUCCESS in 9m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd82e526570e4e79b275b4f1ddaf1fe2 : SUCCESS in 11m 32s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9f0f51e876064c15a586e65d5ed86938 : SUCCESS in 42m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/41a995bb9f8e4ec98d047d7883d378fe : SUCCESS in 1h 00m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2643951c0bb74f288406c2d40a7a1673 : SUCCESS in 41m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7b05c5f4784c4f728613e14eba83350c : SUCCESS in 27m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1d64e549b7e845349a4baae4d53a9a45 : SUCCESS in 34m 14s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5a1c4cc3f7054bfd9fef2d78e30c7264 : SUCCESS in 1h 34m 44s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7272f6549f524f5bb10409100c051ac6 : SUCCESS in 3h 11m 57s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/42268074255740828ed156cf2c1f62ea : FAILURE in 33m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e44468872ab5491da74509331dce6df5 : SUCCESS in 11m 31s","accounts_in_message":[],"_revision_number":6},{"id":"f3319fe68259f6f3ca886cb0bec8adcf27872af7","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-15 06:53:05.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"bd7b27dbe39b3b0565b023fa7bc41d838482a967","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-09-16 07:56:52.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"c579bd5a0cfd95d2785f377e448d15034ecff3e4","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-09-16 08:28:02.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"bae9d7cb43eb7b363655ab06de8d7ac38f7d8a86","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-09-16 08:58:09.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"aedb666a0626bd66b2a085062a94e3062021b726","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-17 07:11:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"45247946b66bc5ede6240c1de17a7dac19daa5ea","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-09-17 07:25:12.000000000","message":"Patch Set 6:\n\n\u003e (1 comment)\n\nRegarding the caching. I\u0027d incline towards using a generic solution (memoization) instead of brewing a custom locally, even if it looks simple. But I can let this go :)","accounts_in_message":[],"_revision_number":6},{"id":"b539dadd8d63fea15743af3727157ed1e0413524","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-17 07:33:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"84a516bdd5104208c08bc7876d790899022db43d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 10:57:40.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67eafe69697f4d32abb731f97efd9ca9 : SUCCESS in 5m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d97667b9a66f4b99a20d69ed652146c4 : FAILURE in 9m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41f486d0552d4ac0814e733224fda312 : FAILURE in 11m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f35629de7fe4bd88e4462c89abb6ed5 : FAILURE in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3693ec39da704cb5a3332096640d6b52 : SUCCESS in 5m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d65da0215094f60bb0abe591c72cc79 : FAILURE in 8m 46s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4e78915a4cb440daa8a2dfed2699b78c : SUCCESS in 54m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bf50188c763346888da7415d5078f6d8 : SUCCESS in 45m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ae442b57980147ac89cc46b581bf81bf : SUCCESS in 45m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/951d422eece04df58f10b8f6e89ea967 : FAILURE in 28m 00s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/580552e50d4a48e6b5a18e4359a00d71 : FAILURE in 27m 22s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/20bc016a58064f42b0eba4120e4df183 : SUCCESS in 1h 46m 00s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ed011bc9cbaf45a9850d3fbebf3bae7a : SUCCESS in 3h 18m 59s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/30114517b5ce4137aeb97373406b094e : SUCCESS in 1h 09m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ee38c40539d471bbeb7e4a3969a69ad : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":7},{"id":"42c4eb65a99fa298f1266974bf39e8ef22d3d812","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-09-17 11:02:24.000000000","message":"Patch Set 7: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"3340210841d5a9c8bd218b2ad1d96926c1117f13","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-17 11:18:33.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"78a6fba7c54695e003a87081336c78e2dfdeac64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 14:09:35.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32f31e0a184946db8d7f73537dcb9c41 : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9820f4d5e98045c29e4110f212e6473e : FAILURE in 10m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99cb4c08b0be43e38d42acbabafe7308 : FAILURE in 7m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42154df3c1ad49a89861b466d4a7f96d : FAILURE in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58dd61e0abb340019ab6544b49a2a8cd : SUCCESS in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44b1dde6f70b4c3eb62c89df50d55cd6 : FAILURE in 10m 44s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ee1490e0687e43c380c83232b8259ab8 : SUCCESS in 51m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4cbdfbd97ba4473ab3054c2ec95758e3 : SUCCESS in 44m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1a226cd32a7c4b8f8ed307983fc4ea38 : SUCCESS in 50m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e4c21c7b2b1b4d8683d040e739620b62 : FAILURE in 31m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a61f9f5dcc634e089dec5f040cd5340e : FAILURE in 27m 34s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/2c0159c967dd40c8b19341907832985e : SUCCESS in 1h 43m 30s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/80515db2d29e4710aa8afbf2104e6d7e : SUCCESS in 3h 04m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b69f7740e9814545adb2e019c421c0c2 : SUCCESS in 1h 09m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eae481dccc624ee7ba25672bad562172 : SUCCESS in 10m 15s","accounts_in_message":[],"_revision_number":7},{"id":"e90cd41650ed58ccd3a3d9669bdcf8d66ba09ab3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-18 03:24:12.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"9db09ef9a8ddc5b09cfa34dadcb44eab429f546b","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-09-18 04:37:08.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"58f3bda18631584cfcacd234567ca0b49f20d398","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 07:26:50.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d41e0fe2eaec4916a8cec87aae0f06b5 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e611917724fd46828876f1d666460919 : FAILURE in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9f5023da8d84505a903739570cc9641 : FAILURE in 8m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb0b42af918d4206a157372962c321a3 : FAILURE in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e4bc01acd864bad953be926a56042ca : SUCCESS in 6m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cf155d67b624187a24459b6f539e6f7 : FAILURE in 9m 00s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/58d4e907c96c447a8b7f622e90ea095b : SUCCESS in 43m 56s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ca4267a97a654bb5bca85f7c015e7328 : SUCCESS in 44m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ca0f0da9bc624076a1cafe47be4e88b8 : SUCCESS in 42m 48s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/03da7e4cc00b49689fce5bf300c775a9 : FAILURE in 28m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fc9dbfe7d86840bb95888f1e65e84dce : FAILURE in 27m 05s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1e7442edc9be4313bf3ebacc93d76f91 : SUCCESS in 1h 49m 13s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a66cbae40f2442e89b8c88e3500faffa : SUCCESS in 3h 29m 01s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0cc925e2424b459b9bbd672e18855fde : SUCCESS in 1h 07m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ab0c78f05ed42e1a5cc9d181a54573f : SUCCESS in 10m 01s","accounts_in_message":[],"_revision_number":8},{"id":"f887a5bacec1bc86220aa71000b4092addafe1c0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-19 05:03:13.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"776b476f10a32e18b088c260db04b4ce7b2fef1b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-19 05:03:27.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"708f7cf01c2187459279756c000166b03ef6b6d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 08:20:18.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdd7cf50d6c54a42aec326ca0cb139b9 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b9e9bf5e6064294a08d2b04631f69fe : SUCCESS in 9m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44792b6b67eb4493b17fd164824a5fe5 : SUCCESS in 8m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f64ea41bf4a547b5aa2cd71105fb6521 : SUCCESS in 9m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e18337ca7d141268c1abd93abc0ade0 : SUCCESS in 4m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa555b96fdee4d5184efaeb5a1380cb2 : SUCCESS in 8m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/139e5a48cbcc4dec91d68fbd5681768e : SUCCESS in 42m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f5d67e36091e4363af3d42d775145262 : SUCCESS in 42m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/da41dbc56eff4dd7a3af484dfdc99d15 : SUCCESS in 50m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a026c6e9f5134e019bcc1c357bf5c06d : SUCCESS in 29m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c51e62c3196249638959c82155385e56 : SUCCESS in 26m 44s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/2400b113dabb4fdf9b83be51d0732a9d : FAILURE in 36m 37s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/48bb492a507b482481cc0bd32303c49f : SUCCESS in 3h 09m 31s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8ac88f6fc83a4ef89df96b6e85518d4c : SUCCESS in 1h 01m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17d341ef0c9041c6a1cf46c4947498e7 : SUCCESS in 9m 55s","accounts_in_message":[],"_revision_number":9},{"id":"7d7226b63a222370bbe6a427cf75297eef8caf6a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-19 08:23:09.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"0224401e2c2fb8b65aea149491da7ff2523a49f3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-19 09:48:22.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"e92b2706dad038d43f6f042c78fffac2aeccf937","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 12:13:47.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/568cdd2f73d04da1a12f99baaae0fd31 : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6212e5441e684ddabac46ba389555f4c : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25cfe1a280b149398c85af1e8fc0333f : SUCCESS in 9m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50312a1913724af3946f59cc389763c6 : SUCCESS in 9m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7b735c3d3194e71b5303b27cc92c44a : SUCCESS in 5m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99d9a51ebc8740c6a8926d9bd49bc276 : FAILURE in 9m 48s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ccd154745cd24706a5ab54bc6b7b5548 : SUCCESS in 46m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/40b61d048f5f4df9865398fecca88434 : SUCCESS in 44m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5e42b583467b4376aa94798ae535f3cf : SUCCESS in 47m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3741d7b14af44c0d8740ba82571cbc60 : SUCCESS in 29m 26s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5eb65131c9954e74a972bcf3681e8723 : SUCCESS in 31m 35s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/cf0d583ef8d04e2898a6d3d2bdbd5998 : SUCCESS in 1h 55m 09s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3b405458ba0f4b2c9bf947b1a409a022 : SUCCESS in 3h 39m 12s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/862c6febb0844b42a647debd80535ddb : SUCCESS in 1h 12m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0599a2542d34c218a95fd25afc0d3c7 : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":9},{"id":"4c249a5d55627a829db0338d67f59c65ac6f2975","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 12:14:00.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"8664c8fbe1147ed92ad4d9ea9912258d022bd091","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 14:43:40.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0ad8f62de9a41838ed0aaa09fa5495f : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b09392117db74bd794b70a763e0635a2 : SUCCESS in 8m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e9be8c5f90344e2ad4d87cb4d1618f7 : SUCCESS in 7m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4894603e6a9d4915ab6ff75f794b3280 : SUCCESS in 9m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6355f94382374d72affcccd9b442f3c6 : SUCCESS in 4m 44s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0862a15bec4a41cfa5eb135cef97c569 : SUCCESS in 44m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/778a26abe50c4016a9ef89fcd8786beb : SUCCESS in 51m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ff446703d22b49999ab8a58c97090035 : SUCCESS in 28m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2e3890c238ba42aa92449c249c0e8e65 : SUCCESS in 27m 38s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/6ad273bb7e214feca105175a2a549a62 : SUCCESS in 1h 42m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/17a3b24ea1614f58b00ad356ff615f28 : SUCCESS in 45m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ddf04d7acdc49679a2bd8981acd387c : SUCCESS in 10m 14s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d5d199c37153494bbe6fd70ed0dac1e5 : SUCCESS in 2h 23m 27s","accounts_in_message":[],"_revision_number":9},{"id":"2c8ade832d2beb189265c5b2dd601791c080d3cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 14:43:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"ad5e1d649ef4cbfb0d33ae2b23366ee65bfeb6fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 14:45:51.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a408eaa820f4a5c87530ca434cc352f : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"64c4038b2c11170a942c909c314e681f73cf895e","revisions":{"3270afcf0e849b212e6f4179faa0a32bc6441f16":{"kind":"REWORK","_number":1,"created":"2019-09-13 11:11:49.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/24/682024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/682024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/682024/1"}}},"commit":{"parents":[{"commit":"6418132a9ba14d77e7516bc3d5ed1db4ea062155","subject":"Fix missing in workflow documents","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6418132a9ba14d77e7516bc3d5ed1db4ea062155"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-13 11:11:07.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-13 11:11:07.000000000","tz":420},"subject":"Optimize creation of language specs","message":"Optimize creation of language specs\n\nChange-Id: Ia700e25752d9d35ece18609f2977e6568062e4bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3270afcf0e849b212e6f4179faa0a32bc6441f16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3270afcf0e849b212e6f4179faa0a32bc6441f16"}]},"branch":"refs/heads/master"},"d22cedc560b7fbd0397d8220239cc9acefacd21b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-13 11:12:17.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/24/682024/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/682024/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/682024/2"}}},"commit":{"parents":[{"commit":"8eb0951a3ff11e376b6ef64bab8bbd1b205cf6f4","subject":"Merge \"moved generic util functions from mistral to mistral-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8eb0951a3ff11e376b6ef64bab8bbd1b205cf6f4"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-13 11:11:07.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-13 11:12:17.000000000","tz":0},"subject":"Optimize creation of language specs","message":"Optimize creation of language specs\n\nChange-Id: Ia700e25752d9d35ece18609f2977e6568062e4bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d22cedc560b7fbd0397d8220239cc9acefacd21b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d22cedc560b7fbd0397d8220239cc9acefacd21b"}]},"branch":"refs/heads/master"},"3491a475f78ac6a1116a368b0b6594ca08a83950":{"kind":"NO_CHANGE","_number":3,"created":"2019-09-13 11:13:04.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/24/682024/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/682024/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/682024/3"}}},"commit":{"parents":[{"commit":"8eb0951a3ff11e376b6ef64bab8bbd1b205cf6f4","subject":"Merge \"moved generic util functions from mistral to mistral-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8eb0951a3ff11e376b6ef64bab8bbd1b205cf6f4"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-13 11:11:07.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-13 11:12:45.000000000","tz":420},"subject":"Optimize creation of language specs","message":"Optimize creation of language specs\n\nChange-Id: Ia700e25752d9d35ece18609f2977e6568062e4bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3491a475f78ac6a1116a368b0b6594ca08a83950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3491a475f78ac6a1116a368b0b6594ca08a83950"}]},"branch":"refs/heads/master"},"0b4c211c38736b03607853d1d672a86e082fd009":{"kind":"REWORK","_number":4,"created":"2019-09-13 11:18:39.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/24/682024/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/682024/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/682024/4"}}},"commit":{"parents":[{"commit":"8eb0951a3ff11e376b6ef64bab8bbd1b205cf6f4","subject":"Merge \"moved generic util functions from mistral to mistral-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8eb0951a3ff11e376b6ef64bab8bbd1b205cf6f4"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-13 11:11:07.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-13 11:15:48.000000000","tz":420},"subject":"Optimize creation of language specs","message":"Optimize creation of language specs\n\n* Various simple optimizations for creation of specs. Mostly\n  additional caching of non-growing data related to the\n  structure of the language spec classes. For huge workflows\n  with thousands of tasks these changes reduce execution time\n  by dozens of seconds.\n* Minor style changes.\n\nChange-Id: Ia700e25752d9d35ece18609f2977e6568062e4bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b4c211c38736b03607853d1d672a86e082fd009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b4c211c38736b03607853d1d672a86e082fd009"}]},"branch":"refs/heads/master"},"e446cc3b4ddc1e0f7d2229f8a8a246f06499b1f3":{"kind":"REWORK","_number":5,"created":"2019-09-13 11:20:35.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/24/682024/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/682024/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/682024/5"}}},"commit":{"parents":[{"commit":"8eb0951a3ff11e376b6ef64bab8bbd1b205cf6f4","subject":"Merge \"moved generic util functions from mistral to mistral-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8eb0951a3ff11e376b6ef64bab8bbd1b205cf6f4"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-13 11:11:07.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-13 11:20:19.000000000","tz":420},"subject":"Optimize creation of language specs","message":"Optimize creation of language specs\n\n* Various simple optimizations for creation of specs. Mostly\n  additional caching of non-growing data related to the\n  structure of the language spec classes. For huge workflows\n  with thousands of tasks these changes reduce execution time\n  by dozens of seconds.\n* Minor style changes.\n\nChange-Id: Ia700e25752d9d35ece18609f2977e6568062e4bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e446cc3b4ddc1e0f7d2229f8a8a246f06499b1f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e446cc3b4ddc1e0f7d2229f8a8a246f06499b1f3"}]},"branch":"refs/heads/master"},"9562a58150797e14a17fe0cf34478bd114404f47":{"kind":"REWORK","_number":6,"created":"2019-09-13 11:22:26.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/24/682024/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/682024/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/682024/6"}}},"commit":{"parents":[{"commit":"8eb0951a3ff11e376b6ef64bab8bbd1b205cf6f4","subject":"Merge \"moved generic util functions from mistral to mistral-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8eb0951a3ff11e376b6ef64bab8bbd1b205cf6f4"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-13 11:11:07.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-13 11:22:10.000000000","tz":420},"subject":"Optimize creation of language specs","message":"Optimize creation of language specs\n\n* Various simple optimizations for creation of specs. Mostly\n  additional caching of non-growing data related to the\n  structure of the language spec classes. For huge workflows\n  with thousands of tasks these changes reduce execution time\n  by dozens of seconds.\n* Minor style changes.\n\nChange-Id: Ia700e25752d9d35ece18609f2977e6568062e4bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9562a58150797e14a17fe0cf34478bd114404f47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9562a58150797e14a17fe0cf34478bd114404f47"}]},"branch":"refs/heads/master"},"d119d6f12c4bbcd88a4323b92a04f357d26ad364":{"kind":"REWORK","_number":7,"created":"2019-09-17 07:33:50.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/24/682024/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/682024/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/682024/7"}}},"commit":{"parents":[{"commit":"8eb0951a3ff11e376b6ef64bab8bbd1b205cf6f4","subject":"Merge \"moved generic util functions from mistral to mistral-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8eb0951a3ff11e376b6ef64bab8bbd1b205cf6f4"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-13 11:11:07.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-17 07:33:11.000000000","tz":420},"subject":"Optimize creation of language specs","message":"Optimize creation of language specs\n\n* Various simple optimizations for creation of specs. Mostly\n  additional caching of non-growing data related to the\n  structure of the language spec classes. For huge workflows\n  with thousands of tasks these changes reduce execution time\n  by dozens of seconds.\n* Minor style changes.\n\nPartial-bug: #1844242\nChange-Id: Ia700e25752d9d35ece18609f2977e6568062e4bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d119d6f12c4bbcd88a4323b92a04f357d26ad364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d119d6f12c4bbcd88a4323b92a04f357d26ad364"}]},"branch":"refs/heads/master"},"d05442f57fadf525a3c2b9a1db3a4d7ee657f1c4":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-09-18 03:24:12.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/24/682024/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/682024/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/682024/8"}}},"commit":{"parents":[{"commit":"86ae8a7e20010aef753e7441e48765a69389d54c","subject":"Change 403 exception message to something more informative and accurate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/86ae8a7e20010aef753e7441e48765a69389d54c"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-13 11:11:07.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-18 03:24:12.000000000","tz":0},"subject":"Optimize creation of language specs","message":"Optimize creation of language specs\n\n* Various simple optimizations for creation of specs. Mostly\n  additional caching of non-growing data related to the\n  structure of the language spec classes. For huge workflows\n  with thousands of tasks these changes reduce execution time\n  by dozens of seconds.\n* Minor style changes.\n\nPartial-bug: #1844242\nChange-Id: Ia700e25752d9d35ece18609f2977e6568062e4bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d05442f57fadf525a3c2b9a1db3a4d7ee657f1c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d05442f57fadf525a3c2b9a1db3a4d7ee657f1c4"}]},"branch":"refs/heads/master"},"64c4038b2c11170a942c909c314e681f73cf895e":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-09-19 05:03:13.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/24/682024/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/682024/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/682024/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/682024/9"}}},"commit":{"parents":[{"commit":"efc4cf078ec7012dec634db061b454625479161e","subject":"Use v2 designate client instead of v1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/efc4cf078ec7012dec634db061b454625479161e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-13 11:11:07.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-19 05:03:13.000000000","tz":0},"subject":"Optimize creation of language specs","message":"Optimize creation of language specs\n\n* Various simple optimizations for creation of specs. Mostly\n  additional caching of non-growing data related to the\n  structure of the language spec classes. For huge workflows\n  with thousands of tasks these changes reduce execution time\n  by dozens of seconds.\n* Minor style changes.\n\nPartial-bug: #1844242\nChange-Id: Ia700e25752d9d35ece18609f2977e6568062e4bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
