)]}'
{"id":"openstack%2Fmistral~683344","triplet_id":"openstack%2Fmistral~master~Ib509653d38254954f8449be3031457e5f636ccf2","project":"openstack/mistral","branch":"master","topic":"bug/1844242","hashtags":[],"change_id":"Ib509653d38254954f8449be3031457e5f636ccf2","subject":"Add an ability to disable workflow text validation","status":"MERGED","created":"2019-09-20 09:47:03.000000000","updated":"2019-09-27 18:41:28.000000000","submitted":"2019-09-27 18:39:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683344-1569609596714-d890bd4f","meta_rev_id":"d40bdf228217b519b0aff6424b2ba76340b0ed4e","_number":683344,"virtual_id_number":683344,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 18:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-09-27 08:10:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":2,"date":"2019-09-27 08:00:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-09-27 16:09:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 11:09:37.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-09-24 11:09:37.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-09-24 11:09:37.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-09-25 12:42:52.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-09-27 08:00:31.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-09-27 08:10:12.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-09-27 18:39:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe66c231313e00bc168b80fa9d59aa01ae065f89","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-20 09:47:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e36db093fbf162395f92084a14363864022a3d5c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-20 09:47:22.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"00bde4aa0112d102e68c7fe67a14a906fd5efc1d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-20 10:08:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0be9d689882a93648ce4574e90c1d2e0244ca3a1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-20 10:39:05.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"0654200e30bc68bf0fed2f67bf52542f477d63ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 13:30:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11d037d9e08342789f0e628edccc2ed9 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a3d3cd901a74fbebaf4a360d1884393 : FAILURE in 9m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b47e2357144421e904406d6e59b8cf4 : FAILURE in 8m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e81e8591d6d547dc8a2379c43d098944 : FAILURE in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf48b40b75644a4b9263336f285ef1b2 : SUCCESS in 5m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c1b8acc99bd447180ee331f16a41016 : FAILURE in 8m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f47759c39c2a479795e1b234f94d385a : SUCCESS in 40m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3d23b6378f794c078eefebf052177df7 : SUCCESS in 39m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b77b17fa37884f16a6c3bcc9269e2d29 : SUCCESS in 58m 32s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4dd49584a7644cfab1eb35b29986cac4 : FAILURE in 27m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/85a4bf5e34454516ac8ad51ae4988a94 : FAILURE in 26m 04s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/4703af1bc8d241ea8261fefdb2a845c5 : SUCCESS in 1h 47m 30s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9d99c5f3a8064bf59bf6089115185937 : SUCCESS in 3h 14m 45s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/abbaadaa22264522aa7216aebbc4cfb7 : SUCCESS in 1h 08m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d82482b6c44f4e269cb9d9f6131eb511 : FAILURE in 10m 14s","accounts_in_message":[],"_revision_number":2},{"id":"fe784cd03cd71b80eb0cbf9dfa18c196f5ba1611","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-24 08:48:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"582e489cd19d28a0920792b7b76fe93ca79a39cf","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-24 09:53:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a5a801fe8b4301fbf37866792e0786c2f853dd4e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-24 09:56:13.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"465ba313eba7d2c8fb2b687e8c74e1b74a504410","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 13:50:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a7f0c64646640a68401e2a4676d0c6e : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9791577c6a1f45fe8d41db6f426c538a : SUCCESS in 9m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acd986facff148d38e812247bce706c1 : SUCCESS in 8m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd91858fcdbc4078af4f69944fea54bf : SUCCESS in 11m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be916459bf444e33aa602cc8b4917943 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9feec88aa66a498fbab1d09eed6b9d5c : SUCCESS in 4m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4506a736340744dc834ef3362fa75527 : SUCCESS in 9m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/173f1ce08a04467594790926747107c4 : SUCCESS in 40m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5907690c5d0e4bb69cef4a03525b3719 : SUCCESS in 47m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cabf9feefa31440985c0031c4bf9ff93 : SUCCESS in 46m 32s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f4349fcdf9ec468eac04d1a7564d7beb : SUCCESS in 32m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7a56625848294873a6a1f5707410ccaa : SUCCESS in 27m 02s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/ea41a7500f9e47a784b21905fd9bba99 : SUCCESS in 2h 16m 47s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f72e1bfd5db04e918364e5c42a6bb348 : TIMED_OUT in 3h 47m 06s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f54479beb3d54b01993e8bb91e8afe67 : SUCCESS in 1h 22m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5f028ed6195445b8312c6c4dac59df3 : SUCCESS in 10m 15s","accounts_in_message":[],"_revision_number":5},{"id":"e74349e077e4dc4e4b023660de71ca7455bd50ad","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-25 06:32:59.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"dbae6ba2d31284e62fe90fcddc9b1390f8a303bd","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-25 10:25:22.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"937b19d82e803683372a1fad7c763587a0def83d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 10:32:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5069a4b5c4f4be296fa2a40deb7d521 : SUCCESS in 6m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4128e5e4889a479786de906aa55ccfa7 : SUCCESS in 10m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fcd53c0eeca4ab8bbedd865f266d960 : SUCCESS in 9m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18b3c632af5b47039a5de4175ebd82d7 : SUCCESS in 12m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a4617547d4e4d02bcaecde40cea4a9d : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff13895d2b564a6790c3f7d8c2fea54b : SUCCESS in 3m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24bcfd91b7db49c4b16d32c64c9a1005 : SUCCESS in 8m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c8368ac383a74a64827552a9e88b153f : SUCCESS in 57m 50s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/31964cb1d13642d6a957ad696bf5e4fb : SUCCESS in 41m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a0cf3ee4be374b5da7ed3cae6a3b3c22 : SUCCESS in 44m 14s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c3726fbbbc4c4c808fa78d6ea1b816d2 : SUCCESS in 30m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c3849f02ca994e009140ac318ff59e7d : SUCCESS in 31m 03s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/7f5272f7ef21414886ae08233fc41d70 : SUCCESS in 1h 46m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c162fc079d2742c6bde798942041030f : SUCCESS in 3h 33m 42s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ae150bbb834e4a3a82165a0047361b13 : SUCCESS in 1h 06m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/713c9b98c2134a11aaa9cb09994c5226 : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":5},{"id":"f59858d1998ae98e5299de8d1022b3b5a2f348ca","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-09-25 10:36:26.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"aab11bf2ec75084058de8feb5ea81c866088a3d1","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-25 12:42:52.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"2fa1ba239d7777e9e00a447a9de51ff66347feb8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-26 01:17:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"53fbe6361884fc8c5ecd0ff5687fb3ea1b76bd5c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-26 01:22:16.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"251a8ad36e26c0a8eaebf0903429066c0270bdf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 05:08:59.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2791758a7c1428faac99764585fe2fa : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b1a71bf94baf48fc82e5370b4442a1d9 : SUCCESS in 8m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76341d0c36f7480eb8fc6b232c1eca33 : SUCCESS in 8m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f9cffff3c8b4071aa65e9a0de998110 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afad1deeac04453a83b3729f63f6b23a : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/183c224f7e134cb78c4eba784e45478b : SUCCESS in 4m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b953a9ff8f64412ab33a099d8e87a31e : SUCCESS in 8m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c265c4312503421fbfa11ec1bd2fc863 : SUCCESS in 49m 55s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/35563f6d0f344977af9d7adcc6d12a58 : SUCCESS in 40m 59s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a16d9d75460446bd99adf032d576d9cc : SUCCESS in 42m 14s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6fd11eca19ea484c80c4bcf201911f91 : SUCCESS in 29m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6869ce70765d4be68e7f9e8d135d89f6 : SUCCESS in 27m 32s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a7c1093167274bb3bd76495c9e232a87 : SUCCESS in 1h 54m 09s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d556b830b7644e0181a728d02933959b : TIMED_OUT in 3h 40m 36s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/86bc1642effc4cbb8acf7c2fb6cc9507 : SUCCESS in 1h 17m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eaf45f6cc5834337bf8aa5bf9915e899 : SUCCESS in 9m 56s","accounts_in_message":[],"_revision_number":7},{"id":"8f2695e820b3b9721fcd1fc1dfc81cc458224389","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-26 05:18:35.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"35d74e60aef20cdc7c90d86fb8d97d5a1d86d226","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-09-26 06:36:56.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)\n\nI like this. I have some consistency and doc issues mostly.","accounts_in_message":[],"_revision_number":7},{"id":"d805982655370cf65acd28d8a13bbdb92273231f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-26 07:25:08.000000000","message":"Patch Set 7:\n\n(4 comments)\n\nThanks Andras, please reply to my clarifying questions.","accounts_in_message":[],"_revision_number":7},{"id":"8ef29d3c6a52f1e30acf0c0f78b6d0b5cfb5f6eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 08:51:41.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96150c3586904ab98424744344969a0b : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/71787fcc24964b7e9edecf94c826edb0 : FAILURE in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19ed6d77179a4c059cfd2fcd3b79179e : SUCCESS in 13m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b14e3ae8ef3d4363800a027e3a8233ee : SUCCESS in 15m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2173f7bc71034d669be2fca593db4661 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d11f9977680d416c91e534dd3aca04b2 : SUCCESS in 7m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d06e371605ae4926ab6b78eefbaf652b : SUCCESS in 9m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1ee751a9aa664398b8b89b0a93d528fc : SUCCESS in 51m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bc6e595049a74cd29ab0ad1d4d3c7bea : SUCCESS in 50m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7a6e8bd70caf412fa4a665165b3727a0 : SUCCESS in 43m 21s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ffa1d4f3d2a94b1cab6d464630500932 : SUCCESS in 29m 21s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/65d1f8da13b34cf99f1a69570a8bdf33 : FAILURE in 31m 15s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/75167f2864a9474eb8f1ff7f43e9c0b3 : SUCCESS in 2h 15m 32s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4db9770f66c14e32a8ce57b5510d0eb1 : SUCCESS in 3h 17m 24s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/72afe237647744e0836f643530298662 : SUCCESS in 1h 05m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b1f0d20d8c94a5ca358265520ee1f9b : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":7},{"id":"d36697cea5b9a4b08eace84235e0b8e6177dd06f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-26 09:27:08.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"40e06772690e02aa5ad8d2838fa77fb8d2fcf4fe","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-26 09:55:34.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"27dad6a34b7d568a6f213e502917b6cd1668125a","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-09-26 10:00:07.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"a73fca928d5850f5aae74f3d22c137c696f8c55c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 13:57:24.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4fb0984218d4eb9b33bbadafe6e66b5 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9bb0af53b65c45dcad34b393dbedda61 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08c58b787c484710bbe2db45395d4335 : SUCCESS in 9m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3a23ea4bd584e2a8d80f4d2783c143b : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdc443c6525a47beb8172a90debfd670 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8c88823d2ef41ea9742e69fd55b9b0d : SUCCESS in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d4cc15876894da69a3aa728f018c10f : SUCCESS in 10m 12s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f774eba1cbdf4a7f8acca9893d1dabaf : SUCCESS in 41m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8a8c36528d124aacaa0622f6260c6fca : SUCCESS in 52m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/788d5811120e4ed2a9732640cd347755 : SUCCESS in 35m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/38394522c2694898ac030ce270d80fe8 : SUCCESS in 30m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/54f8d7b88a3949efa87c1fe44009cf62 : SUCCESS in 29m 49s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c93bd394f58142bdbd09275b0ca12812 : SUCCESS in 1h 48m 42s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4aef6ccfd11f4de1b4e423fab7fcc63f : TIMED_OUT in 3h 53m 54s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/724d505ac5ad49309a156a977cad6891 : SUCCESS in 1h 12m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0aacde2cefff4437acd2d5c205b63e27 : SUCCESS in 11m 05s","accounts_in_message":[],"_revision_number":9},{"id":"645374ec3ad7c46e228b09e2849aad2b9b389d90","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-09-26 17:35:50.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"dd1280417135de41df1ae1ead57bb613ca3d5a56","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-27 04:18:50.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fdb1751656b874f39d293a131dafc4e69c22f12f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-27 04:19:23.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"bfe1cc747202a516e59390261477ace1c5c57a6b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-27 04:33:04.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"5e4056931a85b29692a2f8bfcb2bd06323a7b8e6","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-09-27 08:00:31.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"6fdfeebe08302eaecc0299fa1c84f2ea1cc99bc3","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-09-27 08:10:12.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"840302048addcbbb51f54a74cdefe8e865557687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 08:26:01.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eabe93abd0aa47e282717c013bef11fc : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3d09dc4e443470392ce1364c2fc26be : SUCCESS in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09c22703ba53403c954332198fce649b : SUCCESS in 9m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eadae574c9d94986bfccf99eac37c373 : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc98c06231794dcfb1744dba9ddf1e42 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f8e938487cb453b8bbd283579268f73 : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a04aa8ee334b4a4a9fc34b109d98e1e8 : SUCCESS in 7m 54s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b86b743617f24aca911ddce9518d9b18 : SUCCESS in 38m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/be59767b27a14ebb822a31fe46c966fb : SUCCESS in 42m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c0d88e6eed044b1d9810ccab85831625 : SUCCESS in 43m 17s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/27bfe96abb254e5eb20771a927bf18b8 : SUCCESS in 28m 16s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/76103855836744219baf8945e09af28c : SUCCESS in 26m 24s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/24f7a4692aed40d0b71e8e572aba04bb : SUCCESS in 1h 57m 35s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/11f221d69f1f4aa1aac7535471af5eb7 : TIMED_OUT in 3h 46m 19s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f95e2517d9a84cfdbfad713d73f27692 : SUCCESS in 1h 21m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6847cbdb90b47889e468bc366eb70fd : SUCCESS in 9m 35s","accounts_in_message":[],"_revision_number":11},{"id":"422e3dc135485badfc083847776c60b60b1e3c9c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-27 11:24:32.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"0127ea1d2e5f7c3ecab90ce5416c9438cab1112e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 15:08:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42749c54db6f4c6e8ae2f2c2e9484273 : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5bf818676f824131b4ce096c7aa1fb89 : SUCCESS in 8m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a9feff48b034c94a6ef742920763219 : SUCCESS in 8m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b177bbb4286749b69d855429ee35105e : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd263addec64c638ccaecb7a433442c : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5718d163f014e55b2cd749fb83bde49 : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/117ec09aaf1546f88a5a9f6849d710f1 : SUCCESS in 9m 40s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ccb2657c780340078fe40626457df47b : SUCCESS in 43m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4485ad4e4f7b413f94a37d72d7ab7afd : SUCCESS in 44m 17s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bd92a4b1167d49d991d3ff4dddb8f8d3 : SUCCESS in 45m 03s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a3133c842b8e42c6b08812be516c1e74 : SUCCESS in 29m 50s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d6a6cf7ba6484c00be7bd043ad943033 : SUCCESS in 28m 01s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b8854b5659ff4cecb52b423754851140 : SUCCESS in 2h 23m 35s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e1773d24786e42bf89cf2544008448ad : SUCCESS in 3h 36m 14s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f2be7be469d1440fa6443b9984ddac3a : FAILURE in 1h 09m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/645b6a9ebc9047c98c2a59298eb5fa6b : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":11},{"id":"dbbebb4f45887b8bbfd99ac30ab17fbc286569b3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-27 16:09:19.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"c387e77ed5240676bfdad10490d284df2cf49ff3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 16:09:31.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"34705d6576691d66a58bbbc7e50cf232ab69d431","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 18:39:56.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ba5a6a7c9b547cd8bafc7a135af88bc : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b40f149dd264664bd04964d4bc1caf7 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbc678292272453688ca0b2212384bd5 : SUCCESS in 7m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/387820c9f96f448cb402d289f93646eb : SUCCESS in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51c029f6e40446df9f9f324df0fd021b : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/943ee70a2789474b9065041c563a4869 : SUCCESS in 4m 24s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4f701fcb93f04ec2a97a0d89504943c3 : SUCCESS in 50m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3d08b35405f84b57a5f3e275cbe12084 : SUCCESS in 43m 37s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1446a4bbc5b54377ac7a4a0d3fd5772d : SUCCESS in 29m 08s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/35ff96cd5f314527aaee64cf1b781e46 : SUCCESS in 27m 21s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/4933dc52951240479e40ca603ea41be3 : SUCCESS in 2h 16m 20s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/96602748d8ae4bfb93bdadbf6d05b9ca : SUCCESS in 49m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5787a5a3998443b7a9786c547ffe9536 : SUCCESS in 9m 12s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8814e884eca646e597423dd595a921ae : SUCCESS in 2h 21m 32s","accounts_in_message":[],"_revision_number":11},{"id":"734f5245455e52e2dc801b26574519df3347e446","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 18:39:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"d1b710e3514e987f404899d4cd7508437592fc5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 18:41:28.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfe1e13e322840df8208096bc554a76a : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/725c77caf0e94e20822f37f2b5479cdd : SUCCESS in 53s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"ac41f94d11cee8c7ed94152f159d3e0ad2a755a6","revisions":{"2d3c7dfd70cf2875ecfd2dc459f9cca689a8c343":{"kind":"REWORK","_number":1,"created":"2019-09-20 09:47:03.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/44/683344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/683344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/683344/1"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-20 05:03:46.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-20 09:31:51.000000000","tz":420},"subject":"Add an ability to disable workflow text validation","message":"Add an ability to disable workflow text validation\n\n* For the sake of the service performance, it may make sense to\n  disable validation of the workflow language syntax if it is\n  affordable for a particular use case. For example, if all\n  workflows are auto-generated by a 3rd party system and tested\n  thoroughly (either by running them with Mistral or at least\n  validating them via the special Mistral endpoint) then we can\n  safely disable validation of the language syntax when uploading\n  workflow definitions. For production systems it makes a big\n  difference if workflow texts are large (thousands of tasks).\n  This patch adds the boolean parameter \"validate\" for requests\n  like \"POST /v2/workflows\" to disable validation, if needed, and\n  the new configuration property to set a desired validation mode.\n  The option is an enumeration and has the following valid values:\n    1) \"enabled_by_default\" - enabled for all API requests unless\n       it\u0027s explicitly disabled in the request itself\n    2) \"always_enabled\" - enabled for all API requests regardless\n       of the flag in the request\n    3) \"always_disabled\" - disabled for all API requrests regardless\n       of the flag in the request\n* Minor style changes.\n\nCloses-Bug: #1844242\n\nChange-Id: Ib509653d38254954f8449be3031457e5f636ccf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2d3c7dfd70cf2875ecfd2dc459f9cca689a8c343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2d3c7dfd70cf2875ecfd2dc459f9cca689a8c343"}]},"branch":"refs/heads/master"},"e759b4b422b406cd2348fb570b67357d9a2032bf":{"kind":"REWORK","_number":2,"created":"2019-09-20 10:08:20.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/44/683344/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/683344/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/683344/2"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-20 05:03:46.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-20 10:08:04.000000000","tz":420},"subject":"Add an ability to disable workflow text validation","message":"Add an ability to disable workflow text validation\n\n* For the sake of the service performance, it may make sense to\n  disable validation of the workflow language syntax if it is\n  affordable for a particular use case. For example, if all\n  workflows are auto-generated by a 3rd party system and tested\n  thoroughly (either by running them with Mistral or at least\n  validating them via the special Mistral endpoint) then we can\n  safely disable validation of the language syntax when uploading\n  workflow definitions. For production systems it makes a big\n  difference if workflow texts are large (thousands of tasks).\n  This patch adds the boolean parameter \"validate\" for requests\n  like \"POST /v2/workflows\" to disable validation, if needed, and\n  the new configuration property to set a desired validation mode.\n  The option is an enumeration and has the following valid values:\n    1) \"enabled_by_default\" - enabled for all API requests unless\n       it\u0027s explicitly disabled in the request itself\n    2) \"always_enabled\" - enabled for all API requests regardless\n       of the flag in the request\n    3) \"always_disabled\" - disabled for all API requrests regardless\n       of the flag in the request\n* Minor style changes.\n\nCloses-Bug: #1844242\n\nChange-Id: Ib509653d38254954f8449be3031457e5f636ccf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e759b4b422b406cd2348fb570b67357d9a2032bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e759b4b422b406cd2348fb570b67357d9a2032bf"}]},"branch":"refs/heads/master"},"729d47bf989b5fd6b484a41becc31f810910f532":{"kind":"REWORK","_number":3,"created":"2019-09-24 08:48:11.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/44/683344/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/683344/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/683344/3"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-20 05:03:46.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-24 08:47:53.000000000","tz":420},"subject":"Add an ability to disable workflow text validation","message":"Add an ability to disable workflow text validation\n\n* For the sake of the service performance, it may make sense to\n  disable validation of the workflow language syntax if it is\n  affordable for a particular use case. For example, if all\n  workflows are auto-generated by a 3rd party system and tested\n  thoroughly (either by running them with Mistral or at least\n  validating them via the special Mistral endpoint) then we can\n  safely disable validation of the language syntax when uploading\n  workflow definitions. For production systems it makes a big\n  difference if workflow texts are large (thousands of tasks).\n  This patch adds the boolean parameter \"validate\" for requests\n  like \"POST /v2/workflows\" to disable validation, if needed, and\n  the new configuration property to set a desired validation mode.\n  The option is an enumeration and has the following valid values:\n    1) \"enabled_by_default\" - enabled for all API requests unless\n       it\u0027s explicitly disabled in the request itself\n    2) \"always_enabled\" - enabled for all API requests regardless\n       of the flag in the request\n    3) \"always_disabled\" - disabled for all API requrests regardless\n       of the flag in the request\n* Minor style changes.\n\nCloses-Bug: #1844242\n\nChange-Id: Ib509653d38254954f8449be3031457e5f636ccf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/729d47bf989b5fd6b484a41becc31f810910f532"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/729d47bf989b5fd6b484a41becc31f810910f532"}]},"branch":"refs/heads/master"},"543aa9d1154316518d1696fbd758ec92ca2dca00":{"kind":"REWORK","_number":4,"created":"2019-09-24 09:53:20.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/44/683344/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/683344/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/683344/4"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-20 05:03:46.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-24 09:51:33.000000000","tz":420},"subject":"Add an ability to disable workflow text validation","message":"Add an ability to disable workflow text validation\n\n* For the sake of the service performance, it may make sense to\n  disable validation of the workflow language syntax if it is\n  affordable for a particular use case. For example, if all\n  workflows are auto-generated by a 3rd party system and tested\n  thoroughly (either by running them with Mistral or at least\n  validating them via the special Mistral endpoint) then we can\n  safely disable validation of the language syntax when uploading\n  workflow definitions. For production systems it makes a big\n  difference if workflow texts are large (thousands of tasks).\n  This patch adds the boolean parameter \"skip_validation\" for API\n  requests like \"POST /v2/workflows\" to disable validation, if\n  needed, and the new configuration property to set a desired\n  validation mode.\n  The option is an enumeration and has the following valid values:\n    1) \"enabled_by_default\" - enabled for all API requests unless\n       it\u0027s explicitly disabled in the request itself\n    2) \"always_enabled\" - enabled for all API requests regardless\n       of the flag in the request\n    3) \"always_disabled\" - disabled for all API requrests regardless\n       of the flag in the request\n* Minor style changes.\n\nCloses-Bug: #1844242\n\nChange-Id: Ib509653d38254954f8449be3031457e5f636ccf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/543aa9d1154316518d1696fbd758ec92ca2dca00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/543aa9d1154316518d1696fbd758ec92ca2dca00"}]},"branch":"refs/heads/master"},"acb993549bfaf63a08818960121b433c1fa99a43":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-09-24 09:56:13.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/44/683344/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/683344/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/683344/5"}}},"commit":{"parents":[{"commit":"47172f64a117d4b2d9d7dfc21efbfc893bf3e429","subject":"Mock the ironic client so not to need a real session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/47172f64a117d4b2d9d7dfc21efbfc893bf3e429"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-20 05:03:46.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-24 09:56:13.000000000","tz":0},"subject":"Add an ability to disable workflow text validation","message":"Add an ability to disable workflow text validation\n\n* For the sake of the service performance, it may make sense to\n  disable validation of the workflow language syntax if it is\n  affordable for a particular use case. For example, if all\n  workflows are auto-generated by a 3rd party system and tested\n  thoroughly (either by running them with Mistral or at least\n  validating them via the special Mistral endpoint) then we can\n  safely disable validation of the language syntax when uploading\n  workflow definitions. For production systems it makes a big\n  difference if workflow texts are large (thousands of tasks).\n  This patch adds the boolean parameter \"skip_validation\" for API\n  requests like \"POST /v2/workflows\" to disable validation, if\n  needed, and the new configuration property to set a desired\n  validation mode.\n  The option is an enumeration and has the following valid values:\n    1) \"enabled_by_default\" - enabled for all API requests unless\n       it\u0027s explicitly disabled in the request itself\n    2) \"always_enabled\" - enabled for all API requests regardless\n       of the flag in the request\n    3) \"always_disabled\" - disabled for all API requrests regardless\n       of the flag in the request\n* Minor style changes.\n\nCloses-Bug: #1844242\n\nChange-Id: Ib509653d38254954f8449be3031457e5f636ccf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/acb993549bfaf63a08818960121b433c1fa99a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/acb993549bfaf63a08818960121b433c1fa99a43"}]},"branch":"refs/heads/master"},"2e3762e1c3c24362cb3200e1f06ed4004e027c1a":{"kind":"REWORK","_number":6,"created":"2019-09-26 01:17:35.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/44/683344/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/683344/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/683344/6"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-20 05:03:46.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-26 01:15:28.000000000","tz":420},"subject":"Add an ability to disable workflow text validation","message":"Add an ability to disable workflow text validation\n\n* For the sake of the service performance, it may make sense to\n  disable validation of the workflow language syntax if it is\n  affordable for a particular use case. For example, if all\n  workflows are auto-generated by a 3rd party system and tested\n  thoroughly (either by running them with Mistral or at least\n  validating them via the special Mistral endpoint) then we can\n  safely disable validation of the language syntax when uploading\n  workflow definitions. For production systems it makes a big\n  difference if workflow texts are large (thousands of tasks).\n  This patch adds the boolean parameter \"skip_validation\" for API\n  requests like \"POST /v2/workflows\" to disable validation, if\n  needed, and the new configuration property to set a desired\n  validation mode.\n  The option is an enumeration and has the following valid values:\n    1) \"enabled_by_default\" - enabled for all API requests unless\n       it\u0027s explicitly disabled in the request itself\n    2) \"always_enabled\" - enabled for all API requests regardless\n       of the flag in the request\n    3) \"always_disabled\" - disabled for all API requrests regardless\n       of the flag in the request\n  \"always_enabled is choosen as the default value for this new\n  property to keep compatibility with the previous versions.\n* Minor style changes.\n\nCloses-Bug: #1844242\n\nChange-Id: Ib509653d38254954f8449be3031457e5f636ccf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2e3762e1c3c24362cb3200e1f06ed4004e027c1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2e3762e1c3c24362cb3200e1f06ed4004e027c1a"}]},"branch":"refs/heads/master"},"c60959eea0eac6b0bb7c031e54d28c243bbc5dfe":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-09-26 01:22:16.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/44/683344/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/683344/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/683344/7"}}},"commit":{"parents":[{"commit":"0de247948b324bd3b0f91b606803a7ec5411423e","subject":"Pass a real session to ironicclient in _get_fake_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0de247948b324bd3b0f91b606803a7ec5411423e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-20 05:03:46.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-26 01:22:16.000000000","tz":0},"subject":"Add an ability to disable workflow text validation","message":"Add an ability to disable workflow text validation\n\n* For the sake of the service performance, it may make sense to\n  disable validation of the workflow language syntax if it is\n  affordable for a particular use case. For example, if all\n  workflows are auto-generated by a 3rd party system and tested\n  thoroughly (either by running them with Mistral or at least\n  validating them via the special Mistral endpoint) then we can\n  safely disable validation of the language syntax when uploading\n  workflow definitions. For production systems it makes a big\n  difference if workflow texts are large (thousands of tasks).\n  This patch adds the boolean parameter \"skip_validation\" for API\n  requests like \"POST /v2/workflows\" to disable validation, if\n  needed, and the new configuration property to set a desired\n  validation mode.\n  The option is an enumeration and has the following valid values:\n    1) \"enabled_by_default\" - enabled for all API requests unless\n       it\u0027s explicitly disabled in the request itself\n    2) \"always_enabled\" - enabled for all API requests regardless\n       of the flag in the request\n    3) \"always_disabled\" - disabled for all API requrests regardless\n       of the flag in the request\n  \"always_enabled is choosen as the default value for this new\n  property to keep compatibility with the previous versions.\n* Minor style changes.\n\nCloses-Bug: #1844242\n\nChange-Id: Ib509653d38254954f8449be3031457e5f636ccf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c60959eea0eac6b0bb7c031e54d28c243bbc5dfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c60959eea0eac6b0bb7c031e54d28c243bbc5dfe"}]},"branch":"refs/heads/master"},"d6dbcb809dcd1885e91935f9b363978966fe961e":{"kind":"REWORK","_number":8,"created":"2019-09-26 09:27:08.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/44/683344/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/683344/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/683344/8"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-20 05:03:46.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-26 09:16:23.000000000","tz":420},"subject":"Add an ability to disable workflow text validation","message":"Add an ability to disable workflow text validation\n\n* For the sake of the service performance, it may make sense to\n  disable validation of the workflow language syntax if it is\n  affordable for a particular use case. For example, if all\n  workflows are auto-generated by a 3rd party system and tested\n  thoroughly (either by running them with Mistral or at least\n  validating them via the special Mistral endpoint) then we can\n  safely disable validation of the language syntax when uploading\n  workflow definitions. For production systems it makes a big\n  difference if workflow texts are large (thousands of tasks).\n  This patch adds the boolean parameter \"skip_validation\" for API\n  requests like \"POST /v2/workflows\" to disable validation, if\n  needed, and the new configuration property \"validation_mode\"\n  to set a desired validation mode.\n  The option is an enumeration and has the following valid values:\n    1) \"enabled\" - enabled for all API requests unless it\u0027s\n       explicitly disabled in the request itself\n    2) \"mandatory\" - enabled for all API requests regardless\n       of the flag in the request\n    3) \"disabled\" - disabled for all API requrests regardless\n       of the flag in the request\n  \"mandatory\" is choosen as the default value for this new\n  property to keep compatibility with the previous versions.\n* Minor style changes.\n\nCloses-Bug: #1844242\n\nChange-Id: Ib509653d38254954f8449be3031457e5f636ccf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6dbcb809dcd1885e91935f9b363978966fe961e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6dbcb809dcd1885e91935f9b363978966fe961e"}]},"branch":"refs/heads/master"},"7bc51ee72e11ffe36927dff6a85cb015d860b564":{"kind":"REWORK","_number":9,"created":"2019-09-26 09:55:34.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/44/683344/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/683344/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/683344/9"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-20 05:03:46.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-26 09:55:07.000000000","tz":420},"subject":"Add an ability to disable workflow text validation","message":"Add an ability to disable workflow text validation\n\n* For the sake of the service performance, it may make sense to\n  disable validation of the workflow language syntax if it is\n  affordable for a particular use case. For example, if all\n  workflows are auto-generated by a 3rd party system and tested\n  thoroughly (either by running them with Mistral or at least\n  validating them via the special Mistral endpoint) then we can\n  safely disable validation of the language syntax when uploading\n  workflow definitions. For production systems it makes a big\n  difference if workflow texts are large (thousands of tasks).\n  This patch adds the boolean parameter \"skip_validation\" for API\n  requests like \"POST /v2/workflows\" to disable validation, if\n  needed, and the new configuration property \"validation_mode\"\n  to set a desired validation mode.\n  The option is an enumeration and has the following valid values:\n    1) \"enabled\" - enabled for all API requests unless it\u0027s\n       explicitly disabled in the request itself\n    2) \"mandatory\" - enabled for all API requests regardless\n       of the flag in the request\n    3) \"disabled\" - disabled for all API requrests regardless\n       of the flag in the request\n  \"mandatory\" is choosen as the default value for this new\n  property to keep compatibility with the previous versions.\n* Minor style changes.\n\nCloses-Bug: #1844242\n\nChange-Id: Ib509653d38254954f8449be3031457e5f636ccf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7bc51ee72e11ffe36927dff6a85cb015d860b564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7bc51ee72e11ffe36927dff6a85cb015d860b564"}]},"branch":"refs/heads/master"},"4d8698a73081fbc57e4c3fcaa8699f805429d802":{"kind":"REWORK","_number":10,"created":"2019-09-27 04:19:23.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/44/683344/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/683344/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/683344/10"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-20 05:03:46.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-27 04:19:04.000000000","tz":420},"subject":"Add an ability to disable workflow text validation","message":"Add an ability to disable workflow text validation\n\n* For the sake of the service performance, it may make sense to\n  disable validation of the workflow language syntax if it is\n  affordable for a particular use case. For example, if all\n  workflows are auto-generated by a 3rd party system and tested\n  thoroughly (either by running them with Mistral or at least\n  validating them via the special Mistral endpoint) then we can\n  safely disable validation of the language syntax when uploading\n  workflow definitions. For production systems it makes a big\n  difference if workflow texts are large (thousands of tasks).\n  This patch adds the boolean parameter \"skip_validation\" for API\n  requests like \"POST /v2/workflows\" to disable validation, if\n  needed, and the new configuration property \"validation_mode\"\n  to set a desired validation mode.\n  The option is an enumeration and has the following valid values:\n    1) \"enabled\" - enabled for all API requests unless it\u0027s\n       explicitly disabled in the request itself\n    2) \"mandatory\" - enabled for all API requests regardless\n       of the flag in the request\n    3) \"disabled\" - disabled for all API requrests regardless\n       of the flag in the request\n  \"mandatory\" is choosen as the default value for this new\n  property to keep compatibility with the previous versions.\n* Minor style changes.\n\nCloses-Bug: #1844242\n\nChange-Id: Ib509653d38254954f8449be3031457e5f636ccf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4d8698a73081fbc57e4c3fcaa8699f805429d802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4d8698a73081fbc57e4c3fcaa8699f805429d802"}]},"branch":"refs/heads/master"},"ac41f94d11cee8c7ed94152f159d3e0ad2a755a6":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-09-27 04:33:04.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/44/683344/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/683344/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/683344/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/683344/11"}}},"commit":{"parents":[{"commit":"0de247948b324bd3b0f91b606803a7ec5411423e","subject":"Pass a real session to ironicclient in _get_fake_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0de247948b324bd3b0f91b606803a7ec5411423e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-20 05:03:46.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-27 04:33:04.000000000","tz":0},"subject":"Add an ability to disable workflow text validation","message":"Add an ability to disable workflow text validation\n\n* For the sake of the service performance, it may make sense to\n  disable validation of the workflow language syntax if it is\n  affordable for a particular use case. For example, if all\n  workflows are auto-generated by a 3rd party system and tested\n  thoroughly (either by running them with Mistral or at least\n  validating them via the special Mistral endpoint) then we can\n  safely disable validation of the language syntax when uploading\n  workflow definitions. For production systems it makes a big\n  difference if workflow texts are large (thousands of tasks).\n  This patch adds the boolean parameter \"skip_validation\" for API\n  requests like \"POST /v2/workflows\" to disable validation, if\n  needed, and the new configuration property \"validation_mode\"\n  to set a desired validation mode.\n  The option is an enumeration and has the following valid values:\n    1) \"enabled\" - enabled for all API requests unless it\u0027s\n       explicitly disabled in the request itself\n    2) \"mandatory\" - enabled for all API requests regardless\n       of the flag in the request\n    3) \"disabled\" - disabled for all API requrests regardless\n       of the flag in the request\n  \"mandatory\" is choosen as the default value for this new\n  property to keep compatibility with the previous versions.\n* Minor style changes.\n\nCloses-Bug: #1844242\n\nChange-Id: Ib509653d38254954f8449be3031457e5f636ccf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ac41f94d11cee8c7ed94152f159d3e0ad2a755a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ac41f94d11cee8c7ed94152f159d3e0ad2a755a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
