)]}'
{"id":"openstack%2Fmistral~684511","triplet_id":"openstack%2Fmistral~master~I4e4bdc24f7aa57799e61f683c06533ef2ea4ed7d","project":"openstack/mistral","branch":"master","topic":"eyalb/cinder","hashtags":[],"change_id":"I4e4bdc24f7aa57799e61f683c06533ef2ea4ed7d","subject":"Remove volumes.promote and volumes.reenable action from cinder","status":"ABANDONED","created":"2019-09-25 06:10:00.000000000","updated":"2019-09-26 08:41:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ae597646ba3e8ffcde8cf6652ec61e15ce1da729","_number":684511,"virtual_id_number":684511,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-09-25 10:20:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"all":[{"value":1,"date":"2019-09-25 06:12:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":2,"date":"2019-09-25 06:43:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 06:12:00.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2019-09-25 06:43:08.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-09-25 10:20:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"849063f73f76a62e61e1a901fed2ce7510f294b2","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-25 06:10:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7eaa6d2ba4dbee7a2325d3a5223b0b121df909be","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-09-25 06:12:00.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31e8a5ea538bc6a075d60a49b44b4165ea436a95","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-25 06:43:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3686c6f82d2c98616449d045787a54f92cc0f7be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 10:20:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d238132f58344318b5a2a396330af94a : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/15a0828c4b834a2c97f7e652137a1b28 : SUCCESS in 9m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9dd8627c599f476eb0f6263c31ce6869 : SUCCESS in 9m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1b550ac3d924606bb91cfe0cb02ddee : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ee5c8b6448941f3b8b99a84d877d404 : SUCCESS in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9656b47f3d7e4da1935e38595eb59a86 : SUCCESS in 10m 28s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/55256edb04fd4b4b8c13ac6336c79c0c : SUCCESS in 42m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2d8aac0ee1f144019b2797b85dedd11d : SUCCESS in 45m 37s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/07408a9102fa4ee99926a12ed9a446d9 : SUCCESS in 49m 32s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/450603ba13244b1193b8e3986dacf89c : SUCCESS in 29m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3266fb68fa114cbcb97d71ddb4276d6e : SUCCESS in 28m 29s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b719cdff3f25455b847e92cee066f85f : SUCCESS in 2h 08m 12s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ecc2a086fb114404aa68e13b00de85aa : SUCCESS in 3h 31m 39s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/aa62f3007cba4a78a6d1d2b6b6c2b786 : SUCCESS in 1h 14m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54a603dc63fa45f293803dd9593c47ff : SUCCESS in 10m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f4a7d4f629224db695df4ff687cb4be0fd678d72","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-26 08:41:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19469ad01bc406ce7f0b2541e110ef07d78bdbd5","revisions":{"19469ad01bc406ce7f0b2541e110ef07d78bdbd5":{"kind":"REWORK","_number":1,"created":"2019-09-25 06:10:00.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/11/684511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/684511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/684511/1"}}},"commit":{"parents":[{"commit":"47172f64a117d4b2d9d7dfc21efbfc893bf3e429","subject":"Mock the ironic client so not to need a real session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/47172f64a117d4b2d9d7dfc21efbfc893bf3e429"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-09-25 05:24:34.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-09-25 05:24:34.000000000","tz":180},"subject":"Remove volumes.promote and volumes.reenable action from cinder","message":"Remove volumes.promote and volumes.reenable action from cinder\n\nThey were removed from cinder api see https://review.opendev.org/#/c/658318/\nalso move to cinder v3 (v2 is deprecated)\n\nChange-Id: I4e4bdc24f7aa57799e61f683c06533ef2ea4ed7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/19469ad01bc406ce7f0b2541e110ef07d78bdbd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/19469ad01bc406ce7f0b2541e110ef07d78bdbd5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
