)]}'
{"id":"openstack%2Fmistral~684711","triplet_id":"openstack%2Fmistral~master~Iba05b9acbfea06214a3ab93911d2eaaa4729278f","project":"openstack/mistral","branch":"master","topic":"bug/1791449","hashtags":[],"change_id":"Iba05b9acbfea06214a3ab93911d2eaaa4729278f","subject":"Task publish is ignored silently","status":"MERGED","created":"2019-09-25 11:22:53.000000000","updated":"2019-11-12 09:46:49.000000000","submitted":"2019-11-12 09:45:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684711-1573551919528-e7a1f455","meta_rev_id":"277b0081112280080798b716ed8e9ea4d6c7fc0c","_number":684711,"virtual_id_number":684711,"owner":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-12 09:45:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2019-11-12 08:00:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2019-11-12 07:18:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":1,"date":"2019-11-12 08:00:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-30 08:18:08.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-11-12 07:18:52.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-12 08:00:09.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-11-12 09:45:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1e1fd8d9178f1482ab154969f68a4c5840ad02e","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-09-25 11:22:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3776f7c7032460c03e454d29d5a2ddf6aa84d193","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 15:11:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42bfb25602e642cea331cae0b7a19580 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cd49de5f3a80410d8eae129853c1fe63 : FAILURE in 10m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/caf156ecf31a4c3cba4f6af9c7df332d : FAILURE in 10m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13466fa830824ad1bee41826ddbb7e85 : FAILURE in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d26facc45264d47a94ee61b34ffa85d : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b907a582217248f8a39a401aba2b9307 : FAILURE in 10m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/aacc4c94e32d45efbbd403d5525e025b : FAILURE in 1h 22m 46s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/98cd5836cef54e5fa9d978a35b9108f6 : FAILURE in 1h 01m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f0fe15528753491c8d06c91551d5e37d : FAILURE in 1h 08m 52s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a725a474a1d74a0982cc3e52c46d6da5 : FAILURE in 36m 28s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3cda8a3595b24c2caf3989752af02235 : FAILURE in 34m 39s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/74922ae9ccfd4383993b4c15be311471 : SUCCESS in 2h 20m 01s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/68a749cba05a48ae851bdbed6aee66e9 : SUCCESS in 3h 30m 15s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4c651d758eef4990bad90cb25bd4338a : SUCCESS in 1h 08m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16c9aef938b242369eba7052f22fef70 : FAILURE in 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9d684842b7cf5a29e5973669072fbc6e74ff0849","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-02 06:59:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d7ab35446624c3b65ad4573f24355db92f27c12a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-02 09:00:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"078ef159816c2c82ba57262ba0c09a427f7190c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 09:48:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88c444fc3a84400787efe1018e54c4ce : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/deeffebbb8fc430699bd39f5e6bc2979 : FAILURE in 11m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8757469e43947d388897cab376c9a82 : FAILURE in 9m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67dc030ed2f140378d74cb5c11f7c84d : FAILURE in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55cde658050f4401ac64cc0eb363d0a2 : FAILURE in 4m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4b2253a82554778870f2871da2f31c1 : FAILURE in 11m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1885b16a6afe4914b129e2e75e9de21f : FAILURE in 1h 00m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/066a866883fd45b38d369d23e74387cf : FAILURE in 58m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/da1c2ca6e9ab4ddfa3af2f9e0268cd6a : FAILURE in 1h 06m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/85b32270829c421ca3dd9a564d33c02e : FAILURE in 35m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/15ee862a86ae43eb8bd0ff6294751f03 : FAILURE in 36m 44s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/05ee4dd5deb84a49aa69c6a174b1cfd4 : SUCCESS in 1h 44m 15s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4f00ca8f18ea441e9f72c48467234649 : SUCCESS in 2h 46m 51s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6f3c41271ff949c689a846f307103dd2 : SUCCESS in 1h 01m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0bc2f6ad8c54da78d8edaf96829e0be : FAILURE in 13m 10s","accounts_in_message":[],"_revision_number":1},{"id":"cae774acf133b2d6a4715b825b10d734aea16fd9","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-02 11:15:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b81f7eb808f58bd9c0c6a9b8d44746a9194a3c96","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-02 11:16:48.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1c3e3419190e1a0dd6ee44530dd4de982cab7942","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-02 11:17:05.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"52730d70576995f94468091d88c516ed761852c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 14:51:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c52033ac44a4dae9f1da3313f95e04c : FAILURE in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7737895ec8ee421ba9e8d17570d11ee2 : FAILURE in 11m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1e30a03d2964f3ba3c90bc85a24ddb6 : FAILURE in 10m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2383dc4964549b285aaf59be50c9817 : FAILURE in 9m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c078823e664049a0c59b192893c643 : FAILURE in 4m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89419954ce244cf4abebfcf0ce42e037 : FAILURE in 12m 13s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/84a630701f1249e78fe765739fc77f9d : FAILURE in 1h 00m 51s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ce8007a496d2429a85798f9966312d30 : FAILURE in 1h 05m 01s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/eae8333864e14d219b5533b4d799f178 : FAILURE in 58m 59s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e209f3566bed4b72821616e53135aae5 : FAILURE in 36m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/521775ded648472983ca7d35ff82beda : FAILURE in 35m 42s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a1f31204684d4445bb5d5a776ab91de4 : SUCCESS in 1h 52m 02s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d33113faacfb4d91bbe8a0eda4381a0b : SUCCESS in 3h 12m 16s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bfbb07a0b2874d37954264d0e18524fc : FAILURE in 1h 06m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4af1e659cc3948cebbcf4abe42d565be : FAILURE in 10m 59s","accounts_in_message":[],"_revision_number":3},{"id":"3111126a1c0cb1ffd0ffdf8db36785359acd15a5","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-03 09:07:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5c9cb7f2f6702f891f7321e8b39c3dee2fdc883e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 12:15:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b87a13a160714ba1ad1918937f7e2fd9 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ac42371ddca45918718cd068bc9d864 : FAILURE in 9m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7178d1e9e3d44eeb6fcb14c7ee11e3b : FAILURE in 11m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d79ba3b1d94a4161a0abdb31cc62e81c : FAILURE in 10m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4a53eb45c1d4977a12fdb786de4c4e9 : SUCCESS in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da270530646340a5a2d4d4ffc90164ef : FAILURE in 9m 17s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8de7f68a9b2d4ecd9d18aaea0de1d401 : FAILURE in 1h 00m 51s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ff1a1f93f6ff43ce85a39c9670dc1dd8 : FAILURE in 1h 02m 06s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/52f2f91bf63c41ddbac01f171777fc4f : FAILURE in 1h 12m 42s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a5d396a9770144b3afdee307b1c9b729 : FAILURE in 34m 32s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/567403fa8aab4ef2b511aa804a3d704b : FAILURE in 35m 02s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/6c0fb39aa95044eebb1a0ce4909653d9 : SUCCESS in 1h 53m 13s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/60f32b81d28d45d9a9e163bd93443057 : SUCCESS in 3h 03m 05s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/acab18748bc0476f9b0328c06a45c20d : SUCCESS in 1h 02m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47e33d2ed56144f49831da542e600051 : FAILURE in 10m 24s","accounts_in_message":[],"_revision_number":4},{"id":"947d108293318aa5e80ebc8b636af5da63735943","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-03 13:19:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ca9c6b4a238d9be9c9a7b9653bd133059a5a994d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 16:55:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03b9116f4f984f1796c5fd3b63b3d12e : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8bbd1ff1541b4919a85d457a45b9e116 : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63c25ac2e4714a3190ee297468bf2193 : SUCCESS in 8m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20e77d6454e44defad8ac0db2ab7eba4 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e7ed3e18aa042b48315dc586a25389a : SUCCESS in 5m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3347584a41b74e20a299203e0dd4a376 : SUCCESS in 8m 32s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5e384bb996524699950ddd07b90b64b3 : SUCCESS in 37m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f5f9035583794a42be5d0f5aeea13578 : SUCCESS in 43m 54s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cfd90a5b66ce459eae34f23b4207d4a4 : SUCCESS in 42m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3c1b1744b91f40e2b6a4db1f6d07cdb9 : FAILURE in 28m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/deb7b7d08e81475d90e30adb4e550f02 : SUCCESS in 26m 24s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/0c353685f4e7433f96d2d9bbf67e16af : SUCCESS in 1h 59m 14s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fa12763870c54d7e8bdd8a31a7fb3119 : SUCCESS in 3h 23m 59s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2bddb91f03f94553b2b00dec081c3ab9 : SUCCESS in 1h 16m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a808181e50654936be78fdbb6e37f9f7 : SUCCESS in 10m 49s","accounts_in_message":[],"_revision_number":5},{"id":"ac58d848054a8dc6f608a8dfc6a1bb3d382111e9","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-06 07:08:11.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c12359765cedc929e8daf0926b0f2f7b4b9b313e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-06 10:14:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fca0b980f4704ed2b107a05da405bb04 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/764cd48c8ede4a8994f574aebe3f98d9 : SUCCESS in 10m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcc9e40147064b88af2a77d655810332 : SUCCESS in 10m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8bb98889f6cc4647a37bc2c0f04b058a : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102946846751479ca4e39428f6fa015b : SUCCESS in 5m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7840fa770308404f81ec53fe77640671 : SUCCESS in 9m 22s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7d083ca4ea594b62a61c6ab263492b92 : SUCCESS in 50m 38s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/20a59b063eea494093765341ad4ab6b1 : SUCCESS in 36m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ed430fea4f7d4964a061247f2e252256 : SUCCESS in 50m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d2e8003774384d63aa874db04409bb05 : SUCCESS in 28m 21s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/dcc1c4f071414345899567176abf80d7 : SUCCESS in 26m 56s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/3aadb427822348199318be38ea933f2f : SUCCESS in 1h 53m 03s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/28bda33d45324c4da8d0e4372424e261 : SUCCESS in 3h 03m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2f0c684038d043b7aea1cc14452fa0ca : SUCCESS in 55m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe029b3d5739452596c5f1911b99dbdb : SUCCESS in 9m 22s","accounts_in_message":[],"_revision_number":5},{"id":"521f7979bff2f695982949a6f7feeb29f8e8f784","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-06 10:23:50.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0f50c263ffc6f8b5887556a79b8b92a6e97611ec","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-16 05:32:13.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e6bc6ae1062317d953c950277bd3a9d3830409ce","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-17 09:18:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6abf0b1bb98cd77aed0cbb66585e9123b3a20a2b","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-17 09:19:28.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"bd54c7e59708950167f77b85187db728099fb62b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 12:27:07.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2753de52ff64a4e9cd9109709995de5 : FAILURE in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2acd43ae53914e3faf3024bb66c6dabe : SUCCESS in 10m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00f12a6874674d4280880d9d01517549 : SUCCESS in 8m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b27baf1a6d6441c6a4d430a68187936f : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f880cbe2408c47b2a5d6d8711a5f753d : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6539a59d61bf4e398648faef68304d18 : SUCCESS in 10m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2247295adaa94e19a10fd58bea7b8227 : SUCCESS in 40m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f153eff5d4bb4674885ec2b121955f69 : SUCCESS in 40m 59s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4fdb3a2fc6cb4e17b5cf5e0f9d225fab : SUCCESS in 42m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/78bfba310cac454f96626bd09092ee93 : FAILURE in 28m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8f548a9b78c4451f9e2e9499efaca91f : FAILURE in 26m 58s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/616c02069a0146f5be5da0cdf47197d0 : SUCCESS in 1h 32m 59s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b4af02b2f6704f26a72c2c2023f995b4 : SUCCESS in 3h 02m 15s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9f763d94e72d4913a3f4b6371387ab50 : SUCCESS in 1h 10m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04b0bdcf490d47b98ea90ff3064db0cb : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":7},{"id":"3739dd232e2552166be0ebf2c56349d77d6e4133","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-17 12:52:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2b1fb183aaff1f0e292919f2ac9cdd6b6881e122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 16:14:30.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94e31104fef14703aa2efdecdaf2c8d8 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/569728b372904b0292cd4f3df72524f0 : SUCCESS in 9m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ae4959b1fda4b978023e2a2b5e96d86 : SUCCESS in 8m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1cef982b8c954badbbfa6c8d075655e0 : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed78ed3c4acf4c9d902a5cb8855d5085 : SUCCESS in 5m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2997a9218030445482ec0a9d9b453d50 : SUCCESS in 10m 17s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5a3db720262b4b3f9053c0f232891e72 : SUCCESS in 40m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6434a43c0b3b4ec2924fa389c9b8af3b : SUCCESS in 49m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8187588540f149df8c88717e382c0161 : SUCCESS in 40m 03s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0adc94cbcc7f4cf7aa691da35d2efeeb : SUCCESS in 29m 19s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/edf8d7faff59436a9b1b5b2374e74677 : FAILURE in 27m 32s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/82a580171f324e52aab242b88028dea1 : SUCCESS in 1h 35m 49s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3cbe81fb21384f4b889c0b268bd05cb1 : SUCCESS in 3h 09m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7e70df347dfe40d2990f2edb51cbb03e : SUCCESS in 1h 04m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ffa3cbf44774ea5b4977252ea4dff93 : SUCCESS in 10m 55s","accounts_in_message":[],"_revision_number":8},{"id":"06345124d852bd230ee9b8a53080750c1e75b5a5","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-22 06:12:47.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"b645f807ac9c30e4306ba06a7c106eb02a044f52","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-22 10:02:04.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"45995e4c2a711072d87c1b7ec36401fa42b85a50","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-22 11:36:53.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"41b85436c3c9e85d241a1f695031b8d62dbbda48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 14:25:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97f77a029b5c4039b55ca231867edc93 : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6691dff8a5de42f7a32f8c7d9b0f1aad : SUCCESS in 9m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d474f4b65c941a298382c5b1ea06167 : SUCCESS in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a4ad6834f434119997c778f7c658b88 : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fac20e64852848e98062afb593c47e92 : SUCCESS in 4m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cef49719c864b1f9e11cbd9a488bd87 : SUCCESS in 10m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/bb21657d7ecb41a0a832b7c18663dc2c : SUCCESS in 41m 59s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/565cce895b8942409d654191f54e0ea2 : SUCCESS in 40m 37s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2f3b66315b744772b385836c65c94bbb : SUCCESS in 45m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e0f7de70e3614499b0c044785849b2a7 : SUCCESS in 29m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c76db72d6f8d4e708a0988c188e1eee1 : SUCCESS in 27m 38s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/16ddfcfafaa44f17b19c452ea35ff5f8 : SUCCESS in 1h 58m 13s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/be206918f267423a8eed28ef92d25a92 : SUCCESS in 2h 39m 16s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8d38e2d39e7246348fbbd89a7cc9a22b : SUCCESS in 55m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11e3bba7402d420885dbb9657aeb7ddc : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":10},{"id":"b0391263f32c43a9e8c1d0750a7464881e55caab","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-30 08:18:08.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"b4a91ae24e6179b960282ad0a5d79f6b685f8467","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-06 05:51:46.000000000","message":"Patch Set 10: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"d082f5fec83c156700d31741ef311b768511610e","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-11 09:41:51.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0ebe680fefb07a0b517dafa3333b14b9b0dd68fd","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-11 09:46:29.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"fb23a34cd3ff6ff91759ee0af43cd83636e3790a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 12:39:57.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fc71618c29442099e2ffde56bd464ea : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d9fab3c5e77b4235b3eeeea5d85b4e3d : SUCCESS in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/689ff7a4c00d4a50bead028442004fec : SUCCESS in 8m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0aea961e13314ec5876fc5a46c7dae7e : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bcacba08c8b4c9c9d0f71496d93a393 : SUCCESS in 5m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5f033e699ef41b092268290bcdd47cf : SUCCESS in 8m 40s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/24ba75ec47b94999898fcdfc7a798bd6 : SUCCESS in 46m 59s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/69521440f3c1406a88b3924bba0a2344 : SUCCESS in 44m 10s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c1a4dfc15d474aebb461ca7ec808a57a : SUCCESS in 48m 47s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f8ef310b82f44fa4859eac25b83d5296 : SUCCESS in 28m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a2dd9b63954b43a58c66c609025c72f5 : SUCCESS in 27m 15s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e1b304792fa04e76bfebdd8b0dfe971a : SUCCESS in 1h 32m 01s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/81d5c0929ee34f10bfc8586e623ea243 : SUCCESS in 2h 45m 25s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a9143e837155430a9490c501e298a77d : SUCCESS in 1h 01m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3c36ae805cc47b5b80b95609606b144 : SUCCESS in 9m 56s","accounts_in_message":[],"_revision_number":11},{"id":"7e6efd45f396a9b48c951d625f77c88cc69278b6","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-12 07:18:52.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"ff54dfdbedbb44080ee519caef651b9d79527bff","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-11-12 08:00:09.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"61d18ccc12efb2cae800de94b723b255ad1c8542","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 08:00:22.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"1ea1150aa0573b562d2039df0ba4f7246aef1fa1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 09:45:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"69c450f34f1c5729a0c2bc186d880079a9fb60dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 09:45:19.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8ec2cfa635e4741b1e7571f34e6ba06 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f20a3397ec7e4cc990eba38a4793b23a : SUCCESS in 12m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/776623b2641745d78e5c6447526eea76 : SUCCESS in 8m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd1bb868570846109af52394d745f36b : SUCCESS in 13m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4574b19efc694871bb2ce028f9835449 : SUCCESS in 8m 20s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/43b3913abb184af98e1947519f0c8db2 : SUCCESS in 59m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7e1fad44167f4ae885c06d1f8447b0d1 : SUCCESS in 41m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/34ed88fc2e274e7a9024602571763a92 : SUCCESS in 34m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a80f80c625fb4895ac678f4b858b4a38 : SUCCESS in 26m 23s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/6b3f25ea1227411096597ab6ebed8952 : SUCCESS in 1h 40m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e131c03452634d0a94c0f3c8501910fd : SUCCESS in 45m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e361bc1fb88408bbf1ff41caa1b9199 : SUCCESS in 9m 51s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/66ec8ad30607459f8e428194dd55149b : SUCCESS in 1h 30m 06s","accounts_in_message":[],"_revision_number":11},{"id":"2ed3a99c3a37c3800b7c7b87805acdc831da4351","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 09:46:49.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82dc1a272c40488cb7d9fb5c20623fd0 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"c7a54d22df990795141dc990300ab76704aa5d30","revisions":{"1c72ff952c0ced4789b79c8524a9dd2fd9f90cb6":{"kind":"REWORK","_number":1,"created":"2019-09-25 11:22:53.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/11/684711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/684711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/684711/1"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:21:54.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:21:54.000000000","tz":0},"subject":"Task publish is ignored silently:-","message":"Task publish is ignored silently:-\n\nbug-fix, on-success, on complete published variables are not merged with regular published variables\n\nChange-Id: Iba05b9acbfea06214a3ab93911d2eaaa4729278f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1c72ff952c0ced4789b79c8524a9dd2fd9f90cb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1c72ff952c0ced4789b79c8524a9dd2fd9f90cb6"}]},"branch":"refs/heads/master"},"2bc62409046d3b65fe863dd5007dfb58d7c7cd30":{"kind":"REWORK","_number":2,"created":"2019-10-02 11:15:37.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/11/684711/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/684711/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/684711/2"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:21:54.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-02 11:15:21.000000000","tz":0},"subject":"Task publish is ignored silently:-","message":"Task publish is ignored silently:-\n\nbug-fix, on-success, on complete published variables are not merged\n with regular published variables\n\nCloses-Bug: #1791449\nChange-Id: Iba05b9acbfea06214a3ab93911d2eaaa4729278f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2bc62409046d3b65fe863dd5007dfb58d7c7cd30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2bc62409046d3b65fe863dd5007dfb58d7c7cd30"}]},"branch":"refs/heads/master"},"33cea9a9f8afcaa82ae7d440db1ce90a249059e6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-02 11:16:48.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/11/684711/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/684711/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/684711/3"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:21:54.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-02 11:16:44.000000000","tz":0},"subject":"Task publish is ignored silently","message":"Task publish is ignored silently\n\nbug-fix, on-success, on complete published variables are not merged\n with regular published variables\n\nCloses-Bug: #1791449\nChange-Id: Iba05b9acbfea06214a3ab93911d2eaaa4729278f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/33cea9a9f8afcaa82ae7d440db1ce90a249059e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/33cea9a9f8afcaa82ae7d440db1ce90a249059e6"}]},"branch":"refs/heads/master"},"0263f6249b2085977de48ba3e4f0c659a194065d":{"kind":"REWORK","_number":4,"created":"2019-10-03 09:07:37.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/11/684711/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/684711/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/684711/4"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:21:54.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-03 09:07:18.000000000","tz":0},"subject":"Task publish is ignored silently:-","message":"Task publish is ignored silently:-\n\nbug-fix, on-success, on complete published variables are not merged\n with regular published variables\n\nCloses-Bug: #1791449\nChange-Id: Iba05b9acbfea06214a3ab93911d2eaaa4729278f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0263f6249b2085977de48ba3e4f0c659a194065d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0263f6249b2085977de48ba3e4f0c659a194065d"}]},"branch":"refs/heads/master"},"9d55253e7959dd83f2b3be8939f49e47c9430ecb":{"kind":"REWORK","_number":5,"created":"2019-10-03 13:19:20.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/11/684711/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/684711/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/684711/5"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:21:54.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-03 13:18:41.000000000","tz":0},"subject":"Task publish is ignored silently:-","message":"Task publish is ignored silently:-\n\nbug-fix, on-success, on complete published variables are not merged\n with regular published variables\n\nCloses-Bug: #1791449\nChange-Id: Iba05b9acbfea06214a3ab93911d2eaaa4729278f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9d55253e7959dd83f2b3be8939f49e47c9430ecb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9d55253e7959dd83f2b3be8939f49e47c9430ecb"}]},"branch":"refs/heads/master"},"f3664b73ce894211ddcf436be1900f3437f4d7b2":{"kind":"REWORK","_number":6,"created":"2019-10-17 09:18:03.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/11/684711/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/684711/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/684711/6"}}},"commit":{"parents":[{"commit":"7a6aac0f5f008d67eeb5d3f63ddac42bb62f8e69","subject":"Fix \"root_execution\" lazy loading issue and refactor execution.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7a6aac0f5f008d67eeb5d3f63ddac42bb62f8e69"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:21:54.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-17 09:17:38.000000000","tz":0},"subject":"Task publish is ignored silently:-","message":"Task publish is ignored silently:-\n\nbug-fix, on-success, on complete published variables are not merged\n with regular published variables\n\nCloses-Bug: #1791449\nChange-Id: Iba05b9acbfea06214a3ab93911d2eaaa4729278f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f3664b73ce894211ddcf436be1900f3437f4d7b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f3664b73ce894211ddcf436be1900f3437f4d7b2"}]},"branch":"refs/heads/master"},"3bcff58253656b59cbf542350723be4251d945ce":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-10-17 09:19:28.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/11/684711/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/684711/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/684711/7"}}},"commit":{"parents":[{"commit":"dfcd058144b46ce39a5f53036b31d1a77d85aad9","subject":"Merge \"Add missing :param statement in doc string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dfcd058144b46ce39a5f53036b31d1a77d85aad9"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:21:54.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-17 09:19:28.000000000","tz":0},"subject":"Task publish is ignored silently:-","message":"Task publish is ignored silently:-\n\nbug-fix, on-success, on complete published variables are not merged\n with regular published variables\n\nCloses-Bug: #1791449\nChange-Id: Iba05b9acbfea06214a3ab93911d2eaaa4729278f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3bcff58253656b59cbf542350723be4251d945ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3bcff58253656b59cbf542350723be4251d945ce"}]},"branch":"refs/heads/master"},"f80c977987be6bc68f5b86fb3653f944b3eab948":{"kind":"REWORK","_number":8,"created":"2019-10-17 12:52:30.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/11/684711/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/684711/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/684711/8"}}},"commit":{"parents":[{"commit":"772043881da6b33768a48a46addfc545dcc4afb2","subject":"Merge \"Log the original exception in is_sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/772043881da6b33768a48a46addfc545dcc4afb2"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:21:54.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-17 12:52:05.000000000","tz":0},"subject":"Task publish is ignored silently:-","message":"Task publish is ignored silently:-\n\nbug-fix, on-success, on complete published variables are not merged\n with regular published variables\n\nCloses-Bug: #1791449\nChange-Id: Iba05b9acbfea06214a3ab93911d2eaaa4729278f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f80c977987be6bc68f5b86fb3653f944b3eab948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f80c977987be6bc68f5b86fb3653f944b3eab948"}]},"branch":"refs/heads/master"},"e797b2cae067eefbd1d255d9f690699c7a10e4fa":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-10-22 10:02:04.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/11/684711/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/684711/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/684711/9"}}},"commit":{"parents":[{"commit":"dfcd058144b46ce39a5f53036b31d1a77d85aad9","subject":"Merge \"Add missing :param statement in doc string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dfcd058144b46ce39a5f53036b31d1a77d85aad9"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:21:54.000000000","tz":0},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-22 10:02:04.000000000","tz":0},"subject":"Task publish is ignored silently:-","message":"Task publish is ignored silently:-\n\nbug-fix, on-success, on complete published variables are not merged\n with regular published variables\n\nCloses-Bug: #1791449\nChange-Id: Iba05b9acbfea06214a3ab93911d2eaaa4729278f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e797b2cae067eefbd1d255d9f690699c7a10e4fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e797b2cae067eefbd1d255d9f690699c7a10e4fa"}]},"branch":"refs/heads/master"},"0e4bf2940e00b170fff8ce2bd684bf6d9710a144":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2019-10-22 11:36:53.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/11/684711/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/684711/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/684711/10"}}},"commit":{"parents":[{"commit":"dfcd058144b46ce39a5f53036b31d1a77d85aad9","subject":"Merge \"Add missing :param statement in doc string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dfcd058144b46ce39a5f53036b31d1a77d85aad9"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:21:54.000000000","tz":0},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-22 11:36:49.000000000","tz":0},"subject":"Task publish is ignored silently:-","message":"Task publish is ignored silently:-\n\nbug-fix, on-success, on complete published variables are not merged\n with regular published variables\n\nDepends-On: https://review.opendev.org/#/c/690040/\nCloses-Bug: #1791449\nChange-Id: Iba05b9acbfea06214a3ab93911d2eaaa4729278f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e4bf2940e00b170fff8ce2bd684bf6d9710a144"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e4bf2940e00b170fff8ce2bd684bf6d9710a144"}]},"branch":"refs/heads/master"},"c7a54d22df990795141dc990300ab76704aa5d30":{"kind":"REWORK","_number":11,"created":"2019-11-11 09:46:29.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/11/684711/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/684711/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/684711/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/684711/11"}}},"commit":{"parents":[{"commit":"fd24972befeb4988f09b20bd4a95dd1f2d9a8dbf","subject":"Fix task expression context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fd24972befeb4988f09b20bd4a95dd1f2d9a8dbf"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:21:54.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-11-11 09:46:17.000000000","tz":0},"subject":"Task publish is ignored silently","message":"Task publish is ignored silently\n\nbug-fix, on-success, on complete published variables are not merged\n with regular published variables\n\nCloses-Bug: #1791449\nChange-Id: Iba05b9acbfea06214a3ab93911d2eaaa4729278f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c7a54d22df990795141dc990300ab76704aa5d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c7a54d22df990795141dc990300ab76704aa5d30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
