)]}'
{"id":"openstack%2Fmistral~684718","triplet_id":"openstack%2Fmistral~master~I78fbf993519beb63ee9aef7058bdcb40f0a12ec3","project":"openstack/mistral","branch":"master","topic":"bug/1702421","hashtags":[],"change_id":"I78fbf993519beb63ee9aef7058bdcb40f0a12ec3","subject":"Make it possible to set None to REST API filters","status":"MERGED","created":"2019-09-25 11:35:57.000000000","updated":"2019-11-19 09:15:21.000000000","submitted":"2019-11-19 09:12:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684718-1574154767900-81677534","meta_rev_id":"e3687e482d4791a5c8fbd4f1cc4c84710aa1a6e3","_number":684718,"virtual_id_number":684718,"owner":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-19 09:12:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-11-19 07:23:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-11-18 05:31:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"date":"2019-11-19 04:57:46.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-11-19 07:23:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2019-11-19 04:57:46.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2019-11-19 04:57:36.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-06 10:13:16.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-11-13 20:06:51.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-11-18 05:31:50.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-19 04:57:36.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-11-19 04:57:46.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-11-19 04:57:46.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-11-19 07:23:13.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-11-19 09:12:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ca1113ca8de7afeb69ccd807beae52eaeba4df3","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-09-25 11:35:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d9f2faf361ae86b346e36eb25f2beb49792ef68","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-09-25 11:36:17.000000000","message":"Patch Set 1: Code-Review-1\n\nwip","accounts_in_message":[],"_revision_number":1},{"id":"cc8d6df0def19ee19a4a7a8d1aa38d4354d4f741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 12:45:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/176744273797435f870c8942e0b796b0 : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3445d654e38b4b169ee751be19f8048e : FAILURE in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e452fe53f7a4a9fbdef136dc8a1fe39 : FAILURE in 9m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c752d97512144ede8f480e13c66bfdf2 : FAILURE in 12m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c37f92f4c5345b899a7c69bf87b0100 : FAILURE in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0240b0842fc64e1daa1d5b28f66bfcaf : FAILURE in 7m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/890534cfe73f458cbe541a8e23c6f2c1 : SUCCESS in 42m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3b63fb946ab94ac2b4dffd17041646d6 : SUCCESS in 49m 03s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d655184ec5f64041ba57fe1927655dd1 : SUCCESS in 50m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e7aa46cf100c4abebd85098c4e25181c : FAILURE in 27m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c7c9106d9afe4485b35b8d89520c0cd7 : FAILURE in 26m 49s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c9c05b02ae854ff2a9693b452314f2b1 : FAILURE in 40m 46s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0728c8b948374836a3ae85687133a62b : FAILURE in 58m 00s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b219ab6d516e4be18c6da328ee886e9e : SUCCESS in 1h 02m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/494fab4587c64766a6341ae6eb2be0a8 : SUCCESS in 9m 33s","accounts_in_message":[],"_revision_number":1},{"id":"bdac66f4543187fb11de26026522f5c46db4df7c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-26 09:25:54.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"8d63e2beaf7c9fc460d1c4302fc7ecfd9d95bd42","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-09-26 10:45:31.000000000","message":"Patch Set 1: -Code-Review Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"87042af0ac0505efc33324c3e2b5faeb51cc5b33","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-02 08:25:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c42a22d0997013470c346c6628c988b94dc1a494","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-02 10:19:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cacaeb1038626d3bdbed67f55049edbfc6e739d0","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-03 09:36:40.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f0e27de4813eead59d85ede1a8acdf70e55ca111","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-03 09:40:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4242eda48a9632ad05270b8bf8c2c2e1e8a4d6ff","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-03 09:41:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dd16ae1e97a1f2cf217ff67e7ee68666fe0195fc","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-03 10:02:31.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c6bb2ba6b5c2d167027d13a9fa45289e19bc567c","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-03 10:09:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8ef49abe90642919024acc53287e62e550c95261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 13:38:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/326e9595f155454a9d264b258c5c2137 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1de7b76edbf64974adc0faa9d2b72067 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d266628e3552434d8aa591f7388c619e : SUCCESS in 8m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d833a5e4fb8e4e6ba5145fab4459dc9d : SUCCESS in 10m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43043f4642834c06adca47ec615ab27a : SUCCESS in 7m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d694659070ca4470a2fb2bf1fec85b8d : SUCCESS in 8m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/75b62a445ef54f06bfc6e4c10b1198f7 : SUCCESS in 42m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fdc45e4f968146fcb9be22c6e478c0d3 : SUCCESS in 41m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0ee7b5e48fdd46ee9d7905402f028182 : SUCCESS in 45m 04s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/aa8a2a956da642598646f5e6d1c98bc7 : SUCCESS in 28m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9a2ef06dd2ee4b32a741c58df2af57d0 : SUCCESS in 31m 03s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/18ac2d03f9044c368bfb61336a78f6ea : SUCCESS in 1h 48m 22s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8e08973fdf434d8aadc8ea4eec4e5acf : SUCCESS in 3h 20m 43s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/70aec4acf8fc40e6aceb31b1b9303b46 : SUCCESS in 1h 08m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8194911c8934b2c9cf331dc8fc251ea : SUCCESS in 9m 17s","accounts_in_message":[],"_revision_number":5},{"id":"9430b283f350c13f94b74648ccee254fb58b1fd6","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-03 13:54:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"af3d0c4c1add1432d285414435e15f84e690ef6c","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-03 14:02:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0d1adc71a9d4720e3f7b675612d1f996e3710c35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 17:12:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcc3aca496eb4204aa720101fbb37930 : FAILURE in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bcb6aecd8504ec7b94a2ffeb8366df9 : SUCCESS in 9m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2e2b839a96f43f2943f1736ec4950fb : SUCCESS in 9m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80594e3b0d89435ea1106cdf5d48b646 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/187cd1bbb1ab44aa96f64575a7a332b0 : SUCCESS in 5m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cb09388fe35436789ed8a1206426866 : SUCCESS in 12m 25s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a365876338584f6a975dd430ab993e11 : SUCCESS in 40m 48s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/697ef4a1be7043c2894acb76af9d5693 : SUCCESS in 41m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c7ccf028ed854af79a1edcc2654a4383 : SUCCESS in 44m 01s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a09c6217151745b2896518f535279115 : FAILURE in 28m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c90d4ff2bc504a82b320ebb9c633c8db : FAILURE in 26m 37s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/8402a4dbdd59463d8fb545b0293899cb : FAILURE in 1h 49m 26s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9aa57ff54af84f2f81b8352fc568814f : SUCCESS in 3h 00m 01s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a24f5bf6c3084fbf852a4da8043e321b : SUCCESS in 1h 04m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1ee070c90094c42a6a02d80ec5b9662 : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":7},{"id":"86ea85555c3e310959803000d01bf678d85f8f10","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-06 07:01:19.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"534ce3fca7dda04d96f615da62c0870a7531d3b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-06 10:07:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d43f6ded294e435cb4127df3419e50cc : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8cfaceef4d6946fa93f4f706292f1a8a : SUCCESS in 9m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69c37205e4594383b4fdd4857e457661 : SUCCESS in 8m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df8344dc9ef3419db042b06cb8bcf1ba : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba919ac29fe4800b520f272c7f6f647 : SUCCESS in 5m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23d0fba3e4a545e3bdbf035d6413bfb7 : SUCCESS in 8m 31s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f5c19b5155924f6b8e275691ab574dda : SUCCESS in 42m 07s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fe59fabe80bb43da8af42d4918fcb514 : SUCCESS in 41m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d4e6d1a1836749bbb9f1eb561a156be1 : SUCCESS in 42m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d990f252834043c3a75c5e92e915f389 : SUCCESS in 28m 30s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/272f2575ed234ed795c949a349a457a0 : SUCCESS in 28m 25s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c8520b7973e440228b74fe63224f7829 : SUCCESS in 1h 43m 29s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/cdd9323957f34b7f97fcac8f8f786d9c : SUCCESS in 3h 01m 57s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/86f8a0d4e9c64feba985494a6887686d : SUCCESS in 1h 02m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b460a4411d6e48e0a05319ea49269f4b : SUCCESS in 10m 45s","accounts_in_message":[],"_revision_number":8},{"id":"58618dbda8b52080ad21f985649463d926dd2a90","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-06 10:13:16.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"92bcecf4ceacf0be2daceeb871f6d3707c8279c0","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-06 12:23:32.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7bed4f647a23541fcbf95bf91931121ef1d359a0","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-06 12:25:33.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1e6e9f4da96689e2cb6595db8f771a0f73e40d7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-06 15:28:49.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fd6dc522cd14367a7d5cff02bfc575a : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3692731585c141a0a836f55955be490a : SUCCESS in 8m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a4c6a20fa5e4fa9ae1fd6a2057a84de : SUCCESS in 8m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a18565411224d7fb3e87c51b3571287 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95dfae803d7c42d3a78f720a73d2c1c8 : SUCCESS in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c781e03649e04dd589351a92e4051010 : SUCCESS in 7m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4f04e458b5f343a09a3660466a9bb6dc : SUCCESS in 40m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/82d3198ca8eb4e3b95c7b51c519f500c : SUCCESS in 40m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ba7fa067fc8b4b6a9319c0b555cb1efe : SUCCESS in 41m 14s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3eeb0d25fca441cab16ca0b2e81a899f : SUCCESS in 27m 37s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f44084f70e254a9d9178b83a251e0260 : SUCCESS in 26m 25s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/0f4a22b19cdd4c4d95bf0a01f5af5d73 : SUCCESS in 1h 34m 31s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9d1f821b366942fba0a59f0b5528da78 : SUCCESS in 2h 58m 09s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/846b195b6579444596ea0ca3c86cf903 : SUCCESS in 1h 01m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/751579dfe0084686b91cc11b916f88af : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":10},{"id":"3b8c1900316060623fea35c2e41b05b492b30746","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-07 05:14:11.000000000","message":"Patch Set 10: Code-Review-1\n\n(10 comments)\n\nPlease address my comments. Thanks.","accounts_in_message":[],"_revision_number":10},{"id":"a7a97748e233ad52f72205553341849dbd7541a8","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-13 11:45:27.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2ff086b191106bd45d6d9db3b5053d929e3b62fe","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-11-13 13:04:52.000000000","message":"Patch Set 11:\n\nCan you explain the purpose of this patch? I can\u0027t get it.","accounts_in_message":[],"_revision_number":11},{"id":"c3dacf172dc46933958a7be79464ff85422009ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 14:30:12.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b23b5738c92e4d96bf87bc4edef220f9 : SUCCESS in 7m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ef795bfe2a34fd7b41509e34b6f416b : SUCCESS in 10m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23431473133641548965f4e149729dd6 : SUCCESS in 8m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ef49515e7a4491fab4ad4e717f22756 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f737bd499d24a438e6f7372be579baa : SUCCESS in 4m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4e8f572e6f641cbb49636cddc052b2e : SUCCESS in 9m 38s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cf95c25a623446aa8a83deeae5c3a898 : SUCCESS in 46m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d53e1c91a3924b72852a7de7b3954503 : SUCCESS in 48m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a6982a8a47ce42d1921764719a29f1ed : SUCCESS in 47m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2a70dea0f90344eb9b78401070aea710 : SUCCESS in 28m 32s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e4babd3042874f4196d162e090aa20e5 : SUCCESS in 31m 43s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/34221529ed9342c3b89dd5a5dba60039 : SUCCESS in 1h 49m 41s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ebeefdc4875744f39dadcc918bf3a0cf : SUCCESS in 2h 39m 55s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c1d5dfec8bd6420e8a08e297f66c86d2 : SUCCESS in 1h 06m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c569c2e511f1409f96879457f210576f : SUCCESS in 13m 36s","accounts_in_message":[],"_revision_number":11},{"id":"3a4d47b9aefb80891fa9d169f4fb5b5e46da3e86","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-13 14:39:49.000000000","message":"Patch Set 11:\n\nsure, the API did not support queries with nullable fields being None(if it was None, then it will be ignored), which you might need.\nfor example, if you want to get all the root workflow_executions, its those executions with root_execution_id\u003dNULL, but you could not do that with the current API.\n\nyou can check https://bugs.launchpad.net/mistral/+bug/1702421 for more details.","accounts_in_message":[],"_revision_number":11},{"id":"8fc79f7f47915cd87ee22449a1425817629725a9","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-11-13 20:06:51.000000000","message":"Patch Set 11:\n\nI still can\u0027t understand why we can\u0027t just add new \u0027root_execution\u003dTrue\u0027","accounts_in_message":[],"_revision_number":11},{"id":"a0bfda6dd54863d220852366123608184465395c","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-16 10:02:39.000000000","message":"Patch Set 11:\n\nthis change is not just relevant for the executions API,this is generic enough to support other APIs if needed in the future.\nyou should be able to query None fields, and not need to add new logic for every nullable field you might want to query.","accounts_in_message":[],"_revision_number":11},{"id":"3310e8361bb162655956f2f31fe553b39842bb00","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-18 04:49:54.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"8db46a71d6ba8e0a88b8ab696bad45a3b81fe741","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-18 04:51:33.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"214e2a8c0a56eb6e08aae96edcb2dd2e79fcd94f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-18 05:23:40.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"c035a66f10d63b1a4a8a86863c15502e6e437d44","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-18 05:31:50.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"3bfd4dad892e3c0647aa55aeda91b78a0382ae49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 08:14:00.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0349c9bae224430faf938ba1be7b08c7 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4530470cfa5434fa49caabecd11cfe0 : SUCCESS in 8m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0ae67a5c20a4051b0f002a5c9ea8172 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b854dc16e234f13bdce3927c3689850 : SUCCESS in 7m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74ac160483274668b8bed364375cad4c : SUCCESS in 8m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c5fe6a87935b46b98ab182128e93369d : SUCCESS in 43m 42s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9ae8acfbaeab449caec1254e819e4993 : SUCCESS in 45m 14s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/20a992b27d074012af19ee58d434ef8b : SUCCESS in 43m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/67786f39297c4e27a1d7651863a550ce : SUCCESS in 28m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7d52a30c9a9d4283b10571e004a6599f : SUCCESS in 28m 36s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/88480993ad9044949af1d01baaa9b9b4 : SUCCESS in 1h 29m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b8ba3ae296254429af9f392092b45532 : SUCCESS in 2h 48m 50s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7c71acf3bdc9484481d01da15e1d2730 : SUCCESS in 1h 04m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab40f575dfc44773a82d123d8f24e7d1 : SUCCESS in 9m 58s","accounts_in_message":[],"_revision_number":14},{"id":"4d6142709f9a7a567246ab8eca6881267e275690","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-11-19 07:23:13.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThe million dollar question: how to handle nulls?","accounts_in_message":[],"_revision_number":14},{"id":"73c55d88bf8da1480252cfa3582c4be2771cdb55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 07:23:26.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"d8a182b025057867c31e38fc5dbcc1f963f62c44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 09:12:47.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6706d0fa0f9c487b8c34c73092e28615 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/021337d8b5004a41b45e86838563146d : SUCCESS in 8m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b78143cfd5048929e8b92fe5c1f6d86 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cc1d9e537a54dcaae4635a03ec2170e : SUCCESS in 6m 48s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b40253751d564a5fb8a4c633ea64df73 : SUCCESS in 46m 29s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b78cf763160142a3bd648e9dd340e106 : SUCCESS in 53m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7608a5cf36ce4110b255a5204c9b47de : SUCCESS in 29m 37s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/27c4d686520a4b80b0a9c5bde4457ff6 : SUCCESS in 28m 35s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/14b29499af71407884835d6e38c91f6e : SUCCESS in 1h 40m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/920c5a6148604ac0a3b11862d5edeadc : SUCCESS in 45m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a786724c07c4b3a8ad9a5e6a2722c23 : SUCCESS in 10m 01s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a6c32de18901448c85bca5a33c2fd7b3 : SUCCESS in 1h 46m 51s","accounts_in_message":[],"_revision_number":14},{"id":"09b32c380865f650e1423f536d2b36630fc85e39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 09:12:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"08a4fff553ab5d6c6b388950c4d109b4232f3743","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 09:15:21.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90313d2178414d8db56d0bcb864bc47f : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"c49212e17168f5ebb0833d220bea1dddc5bc88a5","revisions":{"7bfd7e047756ff6dbf43d1cc441a1758d884509f":{"kind":"REWORK","_number":1,"created":"2019-09-25 11:35:57.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/18/684718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/684718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/684718/1"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:35:46.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:35:46.000000000","tz":0},"subject":"It\u0027s impossible to set \"equal to None\" filter when calling REST API:-","message":"It\u0027s impossible to set \"equal to None\" filter when calling REST API:-\n\nadded a list to get_all executions API which includes columns that\n are required to be null in the query.\n\nChange-Id: I78fbf993519beb63ee9aef7058bdcb40f0a12ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7bfd7e047756ff6dbf43d1cc441a1758d884509f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7bfd7e047756ff6dbf43d1cc441a1758d884509f"}]},"branch":"refs/heads/master"},"4716b06f725258a85b3095400f51223a9f0d5eb4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-03 09:36:40.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/18/684718/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/684718/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/684718/2"}}},"commit":{"parents":[{"commit":"9103b5439dd4e28c9e3341a1434c8a1705502a28","subject":"Add the explicit dependency from kombu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9103b5439dd4e28c9e3341a1434c8a1705502a28"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:35:46.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-03 09:36:40.000000000","tz":0},"subject":"It\u0027s impossible to set \"equal to None\" filter when calling REST API:-","message":"It\u0027s impossible to set \"equal to None\" filter when calling REST API:-\n\nadded a list to get_all executions API which includes columns that\n are required to be null in the query.\n\nChange-Id: I78fbf993519beb63ee9aef7058bdcb40f0a12ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4716b06f725258a85b3095400f51223a9f0d5eb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4716b06f725258a85b3095400f51223a9f0d5eb4"}]},"branch":"refs/heads/master"},"888a9a85090f758db5edf15acb183e27b7e9c533":{"kind":"REWORK","_number":3,"created":"2019-10-03 09:40:19.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/18/684718/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/684718/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/684718/3"}}},"commit":{"parents":[{"commit":"9103b5439dd4e28c9e3341a1434c8a1705502a28","subject":"Add the explicit dependency from kombu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9103b5439dd4e28c9e3341a1434c8a1705502a28"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:35:46.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-03 09:40:04.000000000","tz":0},"subject":"It\u0027s impossible to set \"equal to None\" filter when calling REST API:-","message":"It\u0027s impossible to set \"equal to None\" filter when calling REST API:-\n\nadded a list to get_all executions API which includes columns that\n are required to be null in the query.\n\nCloses-Bug: #1702421\nChange-Id: I78fbf993519beb63ee9aef7058bdcb40f0a12ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/888a9a85090f758db5edf15acb183e27b7e9c533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/888a9a85090f758db5edf15acb183e27b7e9c533"}]},"branch":"refs/heads/master"},"3c23f5509fe15629d4443ed9880167beb8857ced":{"kind":"REWORK","_number":4,"created":"2019-10-03 09:41:14.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/18/684718/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/684718/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/684718/4"}}},"commit":{"parents":[{"commit":"9103b5439dd4e28c9e3341a1434c8a1705502a28","subject":"Add the explicit dependency from kombu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9103b5439dd4e28c9e3341a1434c8a1705502a28"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:35:46.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-03 09:40:59.000000000","tz":0},"subject":"It\u0027s impossible to set \"equal to None\" filter when calling REST API:-","message":"It\u0027s impossible to set \"equal to None\" filter when calling REST API:-\n\nadded a list to get_all executions API which includes columns that\n are required to be null in the query.\n\nCloses-Bug: #1702421\nChange-Id: I78fbf993519beb63ee9aef7058bdcb40f0a12ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3c23f5509fe15629d4443ed9880167beb8857ced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3c23f5509fe15629d4443ed9880167beb8857ced"}]},"branch":"refs/heads/master"},"682238f1a1c624a4139651e09df90a4fe3f35e98":{"kind":"REWORK","_number":5,"created":"2019-10-03 10:09:40.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/18/684718/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/684718/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/684718/5"}}},"commit":{"parents":[{"commit":"9103b5439dd4e28c9e3341a1434c8a1705502a28","subject":"Add the explicit dependency from kombu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9103b5439dd4e28c9e3341a1434c8a1705502a28"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:35:46.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-03 10:09:20.000000000","tz":0},"subject":"It\u0027s impossible to set \"equal to None\" filter when calling REST API","message":"It\u0027s impossible to set \"equal to None\" filter\nwhen calling REST API\n\nadded a list to get_all executions API which includes columns that\n are required to be null in the query.\n\nCloses-Bug: #1702421\nChange-Id: I78fbf993519beb63ee9aef7058bdcb40f0a12ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/682238f1a1c624a4139651e09df90a4fe3f35e98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/682238f1a1c624a4139651e09df90a4fe3f35e98"}]},"branch":"refs/heads/master"},"b8d9987129ffd6471faf6e7b7a854856682a6c09":{"kind":"REWORK","_number":6,"created":"2019-10-03 13:54:42.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/18/684718/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/684718/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/684718/6"}}},"commit":{"parents":[{"commit":"9103b5439dd4e28c9e3341a1434c8a1705502a28","subject":"Add the explicit dependency from kombu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9103b5439dd4e28c9e3341a1434c8a1705502a28"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:35:46.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-03 13:54:19.000000000","tz":0},"subject":"It\u0027s impossible to set \"equal to None\" filter when calling REST API","message":"It\u0027s impossible to set \"equal to None\" filter\nwhen calling REST API\n\nadded a list to get_all executions API which includes columns that\n are required to be null in the query.\n\nCloses-Bug: #1702421\nChange-Id: I78fbf993519beb63ee9aef7058bdcb40f0a12ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b8d9987129ffd6471faf6e7b7a854856682a6c09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b8d9987129ffd6471faf6e7b7a854856682a6c09"}]},"branch":"refs/heads/master"},"7d03c40d2f775afa9a576f02ccd178be21bd50d2":{"kind":"REWORK","_number":7,"created":"2019-10-03 14:02:08.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/18/684718/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/684718/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/684718/7"}}},"commit":{"parents":[{"commit":"9103b5439dd4e28c9e3341a1434c8a1705502a28","subject":"Add the explicit dependency from kombu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9103b5439dd4e28c9e3341a1434c8a1705502a28"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:35:46.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-03 14:01:49.000000000","tz":0},"subject":"It\u0027s impossible to set \"equal to None\" filter when calling REST API","message":"It\u0027s impossible to set \"equal to None\" filter\nwhen calling REST API\n\nadded a list to get_all executions API which includes columns that\n are required to be null in the query.\n\nCloses-Bug: #1702421\nChange-Id: I78fbf993519beb63ee9aef7058bdcb40f0a12ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7d03c40d2f775afa9a576f02ccd178be21bd50d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7d03c40d2f775afa9a576f02ccd178be21bd50d2"}]},"branch":"refs/heads/master"},"54027945fa9060cc4976e10998471725a925fa12":{"kind":"REWORK","_number":8,"created":"2019-10-06 07:01:19.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/18/684718/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/684718/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/684718/8"}}},"commit":{"parents":[{"commit":"9103b5439dd4e28c9e3341a1434c8a1705502a28","subject":"Add the explicit dependency from kombu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9103b5439dd4e28c9e3341a1434c8a1705502a28"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-09-25 11:35:46.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-06 07:00:51.000000000","tz":0},"subject":"It\u0027s impossible to set \"equal to None\" filter when calling REST API","message":"It\u0027s impossible to set \"equal to None\" filter\nwhen calling REST API\n\nadded a list to get_all executions API which includes columns that\n are required to be null in the query.\n\nCloses-Bug: #1702421\nChange-Id: I78fbf993519beb63ee9aef7058bdcb40f0a12ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/54027945fa9060cc4976e10998471725a925fa12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/54027945fa9060cc4976e10998471725a925fa12"}]},"branch":"refs/heads/master"},"66293322a264e34cddf1b1b923ccdd21ddca03f5":{"kind":"REWORK","_number":9,"created":"2019-10-06 12:23:32.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/18/684718/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/684718/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/684718/9"}}},"commit":{"parents":[{"commit":"7a6aac0f5f008d67eeb5d3f63ddac42bb62f8e69","subject":"Fix \"root_execution\" lazy loading issue and refactor execution.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7a6aac0f5f008d67eeb5d3f63ddac42bb62f8e69"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-06 12:18:42.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-06 12:21:56.000000000","tz":0},"subject":"It\u0027s impossible to set \"equal to None\" filter when calling REST API","message":"It\u0027s impossible to set \"equal to None\" filter\nwhen calling REST API\n\nadded a list to get_all executions API which includes columns that\n are required to be null in the query.\n\nCloses-Bug: #1702421\nChange-Id: I78fbf993519beb63ee9aef7058bdcb40f0a12ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66293322a264e34cddf1b1b923ccdd21ddca03f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66293322a264e34cddf1b1b923ccdd21ddca03f5"}]},"branch":"refs/heads/master"},"89338d275fe4d9a89e0c946cb25485209c763a6a":{"kind":"REWORK","_number":10,"created":"2019-10-06 12:25:33.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/18/684718/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/684718/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/684718/10"}}},"commit":{"parents":[{"commit":"7a6aac0f5f008d67eeb5d3f63ddac42bb62f8e69","subject":"Fix \"root_execution\" lazy loading issue and refactor execution.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7a6aac0f5f008d67eeb5d3f63ddac42bb62f8e69"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-06 12:18:42.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-10-06 12:25:22.000000000","tz":0},"subject":"It\u0027s impossible to set \"equal to None\" filter when calling REST API","message":"It\u0027s impossible to set \"equal to None\" filter\nwhen calling REST API\n\nadded a list to get_all executions API which includes columns that\n are required to be null in the query.\n\nCloses-Bug: #1702421\nChange-Id: I78fbf993519beb63ee9aef7058bdcb40f0a12ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/89338d275fe4d9a89e0c946cb25485209c763a6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/89338d275fe4d9a89e0c946cb25485209c763a6a"}]},"branch":"refs/heads/master"},"6f4713d76dc5d2d923eb2dd44f1046edc73e9911":{"kind":"REWORK","_number":11,"created":"2019-11-13 11:45:27.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/18/684718/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/684718/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/684718/11"}}},"commit":{"parents":[{"commit":"66d1776f1b992d3b5f593240f4a9bfa75e572f76","subject":"Merge \"Extend capabilities to clean up old executions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66d1776f1b992d3b5f593240f4a9bfa75e572f76"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-07 09:09:33.000000000","tz":420},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-11-13 11:44:41.000000000","tz":0},"subject":"Make it possible to set None to REST API filters","message":"Make it possible to set None to REST API filters\n\n added a list to get_all executions API which includes columns that\n are required to be null in the query.\n\nCloses-Bug: #1702421\nChange-Id: I78fbf993519beb63ee9aef7058bdcb40f0a12ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6f4713d76dc5d2d923eb2dd44f1046edc73e9911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6f4713d76dc5d2d923eb2dd44f1046edc73e9911"}]},"branch":"refs/heads/master"},"d6e1c8a18422699fc373603224ec024499b51b28":{"kind":"REWORK","_number":12,"created":"2019-11-18 04:49:54.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/18/684718/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/684718/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/684718/12"}}},"commit":{"parents":[{"commit":"66d1776f1b992d3b5f593240f4a9bfa75e572f76","subject":"Merge \"Extend capabilities to clean up old executions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66d1776f1b992d3b5f593240f4a9bfa75e572f76"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-07 09:09:33.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-18 04:49:31.000000000","tz":0},"subject":"Make it possible to set None to REST API filters","message":"Make it possible to set None to REST API filters\n\n added a list to get_all executions API which includes columns that\n are required to be null in the query.\n\nCloses-Bug: #1702421\nChange-Id: I78fbf993519beb63ee9aef7058bdcb40f0a12ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6e1c8a18422699fc373603224ec024499b51b28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6e1c8a18422699fc373603224ec024499b51b28"}]},"branch":"refs/heads/master"},"ab94b15296b09d2d078c4212807a7ef15ac837ec":{"kind":"REWORK","_number":13,"created":"2019-11-18 04:51:33.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/18/684718/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/684718/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/684718/13"}}},"commit":{"parents":[{"commit":"66d1776f1b992d3b5f593240f4a9bfa75e572f76","subject":"Merge \"Extend capabilities to clean up old executions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66d1776f1b992d3b5f593240f4a9bfa75e572f76"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-07 09:09:33.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-18 04:51:30.000000000","tz":0},"subject":"Make it possible to set None to REST API filters","message":"Make it possible to set None to REST API filters\n\n added a list to get_all executions API which includes columns that\n are required to be null in the query.\n\nCloses-Bug: #1702421\nChange-Id: I78fbf993519beb63ee9aef7058bdcb40f0a12ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ab94b15296b09d2d078c4212807a7ef15ac837ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ab94b15296b09d2d078c4212807a7ef15ac837ec"}]},"branch":"refs/heads/master"},"c49212e17168f5ebb0833d220bea1dddc5bc88a5":{"kind":"REWORK","_number":14,"created":"2019-11-18 05:23:40.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/18/684718/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/684718/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/684718/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/684718/14"}}},"commit":{"parents":[{"commit":"66d1776f1b992d3b5f593240f4a9bfa75e572f76","subject":"Merge \"Extend capabilities to clean up old executions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66d1776f1b992d3b5f593240f4a9bfa75e572f76"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-07 09:09:33.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-18 05:15:34.000000000","tz":420},"subject":"Make it possible to set None to REST API filters","message":"Make it possible to set None to REST API filters\n\n* Added a list argument to the method get_all() of the executions\n  API controller class that may contain names of columns that a\n  client wants to set to the null value (None internally in the\n  code) in the query string. Thereby we\u0027re getting the ability to\n  filter API entities (currently only workflow executions) fetched\n  with the get_all() by some fields that are assigned to the None\n  (null from the API perspective) value, i.e. typically it means\n  that a value of a field is not defined.\n\nCloses-Bug: #1702421\nChange-Id: I78fbf993519beb63ee9aef7058bdcb40f0a12ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c49212e17168f5ebb0833d220bea1dddc5bc88a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c49212e17168f5ebb0833d220bea1dddc5bc88a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
