)]}'
{"id":"openstack%2Fmistral~684731","triplet_id":"openstack%2Fmistral~master~I59c113b22c60f04e89a631ade8039c4fa62933dc","project":"openstack/mistral","branch":"master","topic":"ironicclient","hashtags":[],"change_id":"I59c113b22c60f04e89a631ade8039c4fa62933dc","subject":"Pass a real session to ironicclient in _get_fake_client","status":"MERGED","created":"2019-09-25 12:30:15.000000000","updated":"2019-09-26 03:41:47.000000000","submitted":"2019-09-26 03:40:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684731-1569469219698-e03b07bf","meta_rev_id":"e9b24498d1acc6ddfcb0c5e3c11e5e8695d1279b","_number":684731,"virtual_id_number":684731,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 03:40:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-09-26 01:19:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-09-26 01:19:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 01:19:39.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-09-26 03:40:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"43b3710fffd4d0d8624d9ce97896023ca0242c3f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-25 12:30:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a19ca4f43fd78cdb7321717c4edfc29611b374b1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-25 12:30:39.000000000","message":"Patch Set 1:\n\nNote that this won\u0027t pass with ironicclient 3.0.0, which is pretty broken.","accounts_in_message":[],"_revision_number":1},{"id":"3b828afa1342c2e44af3b88eccd0d0cef06959a2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-25 12:32:09.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d61442527f223da97c4fbc3581ef455e9dd22f1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 16:30:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f539e27a5444cd3b80c59a44ee02515 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba483a20bc284e9fab70dafd140a8c09 : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e88b048ccb4c441385de9b81c0c311fc : SUCCESS in 8m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76be72dfea174c4da3c1ffb7717629be : SUCCESS in 9m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f2b5a245094fd6b84e2eb7ff6ea27a : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39adf353d42d458684cee8854dced869 : SUCCESS in 9m 09s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2e84a30273394f309144583dc964c0ba : SUCCESS in 47m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/100f3986eec045c19c317e75921063b8 : SUCCESS in 53m 37s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4bc0f88e777940c38433d7202311cd7e : SUCCESS in 42m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7581d0ed3a9d436688c25526d589d8aa : SUCCESS in 28m 21s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/457887f31992498193c3e3a82964df9c : SUCCESS in 26m 30s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e53dbb32f75948e98521e58889c5010d : SUCCESS in 2h 03m 31s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4ea710b55373433c854c296366b8626b : SUCCESS in 3h 28m 09s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2563558deb4c47fdbca2d6043da9b201 : SUCCESS in 1h 06m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d899094f5934fac8ed40adc4e229d11 : SUCCESS in 9m 16s","accounts_in_message":[],"_revision_number":2},{"id":"92de61a7b124f6a2669b985c8f353451dc8397aa","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-26 01:19:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"448e22f45990592cb54bbfda0f803108e5e63ed9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 01:20:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e41bd483d034a0d17ba876e8a2d524cf947d5709","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 03:40:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa1b2f87cdeb4191ab32b3755e1c6e9e : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/65527a82c6a34d3a9783acd1aa991582 : SUCCESS in 9m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/525041bcc7d84854a963881d4baaeaac : SUCCESS in 9m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e8176cea73f4e77a497e6e4c4f3024d : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/609580279580480fb4dfaf0d135ac208 : SUCCESS in 4m 27s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/30d59a93ed024d32a4f90f93e93392c7 : SUCCESS in 38m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/28809540240945e7ba4ba258e02d10f1 : SUCCESS in 40m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/74d21d9c05a443969290537e46672d3a : SUCCESS in 27m 42s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8bf56917ae674a099fc96e859c4dd28d : SUCCESS in 27m 13s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/706b9aaef34e477786fb7a0ce0b2bb83 : SUCCESS in 1h 56m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fd100c8003d946228157594f0660930a : SUCCESS in 56m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49e6d156516a48b0a380cb3ce6799047 : SUCCESS in 10m 31s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/afae2af4e0744bbc87b6afdd7d0a2978 : SUCCESS in 2h 14m 38s","accounts_in_message":[],"_revision_number":2},{"id":"7830ade161da1f39a28bf328928274e18b5eab1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 03:40:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"173c2ed4d39b7beff42dd3ec8a7630eaad2e18f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 03:41:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a1bf22afa884d89a16ee787d1baf972 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0de247948b324bd3b0f91b606803a7ec5411423e","revisions":{"ff59ff97d343c6510f9e01a97065387e1f15bf63":{"kind":"REWORK","_number":1,"created":"2019-09-25 12:30:15.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/31/684731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/31/684731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/31/684731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/31/684731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/31/684731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/31/684731/1"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-25 12:29:21.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-25 12:29:21.000000000","tz":120},"subject":"Pass a real session to ironicclient in _get_fake_client","message":"Pass a real session to ironicclient in _get_fake_client\n\nUsing a dict is not guaranteed to work (and actually doesn\u0027t with\nironicclient 3.0.0, although a few other things are broken with it).\n\nChange-Id: I59c113b22c60f04e89a631ade8039c4fa62933dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ff59ff97d343c6510f9e01a97065387e1f15bf63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ff59ff97d343c6510f9e01a97065387e1f15bf63"}]},"branch":"refs/heads/master"},"0de247948b324bd3b0f91b606803a7ec5411423e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-25 12:32:09.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/31/684731/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/31/684731/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/31/684731/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/31/684731/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/31/684731/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/31/684731/2"}}},"commit":{"parents":[{"commit":"b9b21c823bf8af637c70fe11b84ed8c467681ad2","subject":"Exclude ironicclient 3.0.0 from requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b9b21c823bf8af637c70fe11b84ed8c467681ad2"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-25 12:29:21.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-25 12:31:59.000000000","tz":120},"subject":"Pass a real session to ironicclient in _get_fake_client","message":"Pass a real session to ironicclient in _get_fake_client\n\nUsing a dict is not guaranteed to work (and actually doesn\u0027t with\nironicclient 3.0.0, although a few other things are broken with it).\n\nChange-Id: I59c113b22c60f04e89a631ade8039c4fa62933dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0de247948b324bd3b0f91b606803a7ec5411423e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0de247948b324bd3b0f91b606803a7ec5411423e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
