)]}'
{"id":"openstack%2Fmistral~684732","triplet_id":"openstack%2Fmistral~master~Ib256c02d6b6810caa959de986fba51d5ea664b34","project":"openstack/mistral","branch":"master","topic":"gate","hashtags":[],"change_id":"Ib256c02d6b6810caa959de986fba51d5ea664b34","subject":"Exclude ironicclient 3.0.0 from requirements","status":"MERGED","created":"2019-09-25 12:31:51.000000000","updated":"2019-09-26 02:09:29.000000000","submitted":"2019-09-26 02:08:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684732-1569463701817-17048323","meta_rev_id":"3701e0a1d6b589d89df08fde39bd5a27749d623e","_number":684732,"virtual_id_number":684732,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 02:08:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-09-26 01:20:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-09-26 01:20:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 01:20:13.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-09-26 02:08:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"090ff7f744cabab5472a1f6808945b837503d155","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-25 12:31:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e32029b2af5322c31947f8bcd543d8b6e8975d80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 14:11:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3052babac80a47aeade9efc2d2956901 : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ceda7b556ea43bc8695c4eb783a02d3 : SUCCESS in 8m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e6dc2695d2b49d39bef8a8798b2148d : SUCCESS in 8m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d755eac28b1d4fbaafcd6dded63d9518 : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ec01195fb394341883568b1f92132bb : SUCCESS in 4m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e5c8dd5933994a56ae10152baf040eb1 : SUCCESS in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3c0c99b4c164bd9a56f03e4cf72d034 : SUCCESS in 8m 12s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3f9c0c845f544868bac4f729cef0fccc : SUCCESS in 45m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/05c2972aa52244d4bbb8e626b743815c : SUCCESS in 43m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/94b8820e449240e0a2c03fd8040e5919 : SUCCESS in 1h 07m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8c4b0603ff3246bca2c2e5b455a1b7ea : SUCCESS in 30m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f0c5534d1c4f4cd8b8a4c7147ecb8b49 : SUCCESS in 28m 07s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/25a6794fafc346f5ae1da69c7aeeafe6 : SUCCESS in 1h 22m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32ec51421827497b8c84f8a09b7011b0 : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":1},{"id":"973a7d7a18ecd6cce69eb2c3b05b1f058f336f2e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-26 01:20:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7b12751ed435fa2fab69ab6ad23d5e40914d7d0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 01:20:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"45bf1843fc25ecc15be0146af2593f5b0939071f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 02:08:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c4c8aa601b44e789efad64c1533cac1 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/559da7b0d5514ac78cd531c92370dd3d : SUCCESS in 8m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d93f8e10e1fa47df8c1d39a9770f2fe8 : SUCCESS in 9m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee097c4cdeef4bc5a5873178512cdfd0 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b76f0e90712460c97216899f7f61117 : SUCCESS in 5m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/afce86018fca48d1a1225b9187f1b1db : SUCCESS in 4m 11s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/25fdfe96508b4ae890707a1d5d5b6441 : SUCCESS in 40m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/80476a70add349fcab60164ea044fba8 : SUCCESS in 42m 18s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/031a81ec391042ac83939c290e6add00 : SUCCESS in 28m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ffcf3deb8c344603becd70103c0a1b9b : SUCCESS in 27m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8ead2db4cc684c958376e0047f021fd7 : SUCCESS in 41m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94c38a6bd31045cfa45dbe2e144dd7dc : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"6d6c929365dd8fa3a8a4e04567f1b23c37530dcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 02:08:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f8a7f5a6a607a5f52d4c88a8b9980e02707486f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 02:09:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a2766c018984c49b7a1fc00c2c9c341 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9b21c823bf8af637c70fe11b84ed8c467681ad2","revisions":{"b9b21c823bf8af637c70fe11b84ed8c467681ad2":{"kind":"REWORK","_number":1,"created":"2019-09-25 12:31:51.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/32/684732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/32/684732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/684732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/684732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/684732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/32/684732/1"}}},"commit":{"parents":[{"commit":"64c4038b2c11170a942c909c314e681f73cf895e","subject":"Optimize creation of language specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/64c4038b2c11170a942c909c314e681f73cf895e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-25 12:30:59.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-25 12:30:59.000000000","tz":120},"subject":"Exclude ironicclient 3.0.0 from requirements","message":"Exclude ironicclient 3.0.0 from requirements\n\nIt breaks detection of ironic actions. 3.1.0 should fix it.\n\nDepends-On: https://review.opendev.org/684376\nChange-Id: Ib256c02d6b6810caa959de986fba51d5ea664b34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b9b21c823bf8af637c70fe11b84ed8c467681ad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b9b21c823bf8af637c70fe11b84ed8c467681ad2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
