)]}'
{"id":"openstack%2Fmistral~685036","triplet_id":"openstack%2Fmistral~master~I764586d4f93cca42722f7d598a5c9ac0463730c2","project":"openstack/mistral","branch":"master","topic":"bug/1845482","hashtags":[],"change_id":"I764586d4f93cca42722f7d598a5c9ac0463730c2","subject":"Fix scheduled jobs migration","status":"MERGED","created":"2019-09-26 11:16:06.000000000","updated":"2019-09-27 17:38:52.000000000","submitted":"2019-09-27 17:36:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685036-1569605777052-efb49f7a","meta_rev_id":"7489f0befb020ac2613fe434f94f520838ef4001","_number":685036,"virtual_id_number":685036,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8532,"name":"Ryan Brady","email":"rbrady@redhat.com","username":"rbrady"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 17:36:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2019-09-27 08:09:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":2,"date":"2019-09-27 07:59:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8532,"name":"Ryan Brady","email":"rbrady@redhat.com","username":"rbrady"},{"value":2,"date":"2019-09-27 07:56:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8532,"name":"Ryan Brady","email":"rbrady@redhat.com","username":"rbrady"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":1,"date":"2019-09-27 08:02:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8532,"name":"Ryan Brady","email":"rbrady@redhat.com","username":"rbrady"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 11:18:09.000000000","updated_by":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"reviewer":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"state":"REVIEWER"},{"updated":"2019-09-26 12:09:40.000000000","updated_by":{"_account_id":8532,"name":"Ryan Brady","email":"rbrady@redhat.com","username":"rbrady"},"reviewer":{"_account_id":8532,"name":"Ryan Brady","email":"rbrady@redhat.com","username":"rbrady"},"state":"REVIEWER"},{"updated":"2019-09-26 12:14:45.000000000","updated_by":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"reviewer":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"state":"REVIEWER"},{"updated":"2019-09-27 07:56:40.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-09-27 07:59:12.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-09-27 08:09:30.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-09-27 17:36:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71cbc7cfe2de4ed3c489191ac4b32ae98fbbe931","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-26 11:16:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd8fa271b89eb75b6932d7b58f7a66585b49623b","author":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"date":"2019-09-26 12:14:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5647e2d90a657eaf0cba1c7d6fadfb312d1e9441","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 15:30:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64c6470c2c554f4093fb932ec3b6bdb1 : FAILURE in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/63e9c589eba04cdca1d5550419000585 : SUCCESS in 10m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35024d30f3fc4d2aab1d43cce7e8bfbf : SUCCESS in 11m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b40b3927782a4847bbe57fb8a3f5eaac : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de6aad21a684695af4f2a40fe0380df : SUCCESS in 4m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/826754972f134a3bb19078687528dcee : SUCCESS in 8m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/be107d5be4964475b9988151d0dd3a39 : SUCCESS in 44m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c6703b7b78044021bf6eade8fc1f6371 : SUCCESS in 43m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4eca4efef572463cb24815d3ba3fc54a : SUCCESS in 43m 20s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cb66392e8eff49ba886bedcf6ddf693f : FAILURE in 29m 45s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a038b0d8ff494e9d8e17858e21b246f6 : FAILURE in 30m 03s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/2dbd218d230c4898ae56b8a348d9d5c7 : SUCCESS in 1h 51m 27s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/750109f083bb4b1d92960f040cd9363f : TIMED_OUT in 3h 51m 45s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e5dd39286f21434b9a4d21e69b33fadd : SUCCESS in 1h 12m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/947604194aca4107956f707cdda74a7c : SUCCESS in 9m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9b3ae04a3e53888d23abc439be5f96502f43dc66","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-27 04:21:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bf5208742b35cd1881e44a7e383915d96d408f61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 07:49:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e831b422fea4ee1bc36e8eb8ef5c614 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e990a990d25e467eaf9029740c97a4ee : SUCCESS in 10m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e26eca1b580a473ba497238755ec148e : SUCCESS in 8m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6025b4e7c20f4a3899d23658703e424a : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce6ea1caa3f432db9262768408ca2d3 : SUCCESS in 5m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2db5610623a34d6981c3b4bbef10ab45 : SUCCESS in 9m 06s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/dd51b920224c434692638768ea8e31d0 : SUCCESS in 39m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/98d96f9bbca74a15a0e10a13ad7aa6bc : SUCCESS in 40m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cc03344f320d41e08b1bdd37b8db0d40 : SUCCESS in 40m 48s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1319ee959b164a2eaf2a46338c9f79a0 : SUCCESS in 31m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4cf0282723b240f0b574a045c89f31be : SUCCESS in 27m 31s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/9716b55e403e4e6386282a7cc7a636d1 : SUCCESS in 1h 52m 48s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0dfa2762ddf749759b5c940017806a90 : SUCCESS in 3h 20m 34s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/91fa86f5ff98462db7277b43dd7813b8 : SUCCESS in 1h 00m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08a62708ad1346d2b9ddeb08e37ce532 : SUCCESS in 9m 45s","accounts_in_message":[],"_revision_number":2},{"id":"e430776fca47d8647fe88638e853e407ba0c6d0e","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-09-27 07:56:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"651974db6e1aacde235c7e5c7754de0a8763d550","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-09-27 07:59:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3e87988cb5062470704ebe0b9dfcc688fbfb0d45","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-27 08:02:08.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bd8ebd1a35ee35272155552cf693d8bc0610d194","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 08:02:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"64491aa058827e6b20a7b353381d9c7d2e4fdf10","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-09-27 08:09:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4be0d386fcab3e92fce664c33d1ba39c8de096fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 10:42:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8df73c6998bf4c8cb5935517d1072a1d : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9a3502f0d064de09d678c9f10842aea : SUCCESS in 9m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f18271a742cb438cbb4ff142d53c30e1 : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/384008137d614c40ab4a20c0326f0216 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5de970e9400c41c799e8e0aec4142343 : SUCCESS in 5m 03s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/328acf1757814f1999ed413257cdcea3 : SUCCESS in 41m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/acb91dec943a4cae871fa66289acc298 : SUCCESS in 43m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a4afa13ff55e457b96c504797c41edd5 : FAILURE in 31m 20s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/72cecb3cbfe34f7d9cfc0fbad4ee492b : SUCCESS in 30m 40s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/31ab819c4af443f3b1a66ec051e1fc1e : SUCCESS in 1h 58m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fdadb6972bc9413ab3a1812fed1d30e4 : SUCCESS in 42m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12fbdcc2453149a7bc5564c364994869 : SUCCESS in 9m 51s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/24b388416e2c452eb1ec06d17ed0ae6d : SUCCESS in 2h 34m 28s","accounts_in_message":[],"_revision_number":2},{"id":"bd8e6233526382cb249d2cf430b3cf6cbd12d5c8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-27 11:25:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bb29a86e3a69d9d78e1243658f924e3ca0d7b333","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 15:11:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba2922295c9f4ddda26c74f87c616248 : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/40f14dde45b9427badeb516905a106f8 : SUCCESS in 10m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/156e84076d834e2db197db304321e176 : SUCCESS in 7m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f4aa6f40fffd4aeaa0d1aa0111959891 : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a1c1209ad241118bf94e15c899023d : SUCCESS in 6m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acfb914dd1bc4664acc52d77d5b21684 : SUCCESS in 8m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a219cec33e8e4a9c9052d7577e872fbb : SUCCESS in 46m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/65a8b829ef864b6b9ae44a9ad456aa18 : SUCCESS in 48m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/aeb82e8ebf844b698af8d80c42b08d9b : SUCCESS in 41m 52s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/68dfd3ed0766481ca4ec0568f5f4ca07 : SUCCESS in 28m 26s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4e05913408ae4208a7a80bcf85931706 : SUCCESS in 27m 16s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/db8d37b0058a43beb42944551b82f7d7 : SUCCESS in 2h 11m 34s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7e28d386bb134cd2bbc1e3468922b5dd : SUCCESS in 3h 41m 14s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2bd5e85bc64b4005aaffa54ee38c947e : SUCCESS in 1h 16m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca6ac11107ec463999a9a9d60d8f31ec : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":2},{"id":"9aef1725eae28cfe07db64bb836ad4d847145310","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 15:12:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36815284b5dfb9908adac6115a9529578c04fc5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 17:36:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c454e89e2d444159d09fc9cf46d127b : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/099b11aa1d51485e9e945ee47d8f276e : SUCCESS in 10m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fd49f85c5964e78a560b5cdc9fcc746 : SUCCESS in 8m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5091e2c263c94f9b92b84dd7b08ddeab : SUCCESS in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2765a1b9d0af49e0b54f37a1eeb1ee9e : SUCCESS in 4m 17s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0048094e7aaf45a48f8aff85f2a93cdb : SUCCESS in 41m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7499fbaeb5bd479b9224549b0f1ba084 : SUCCESS in 48m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8444fd857bd946e9821c5214d6b5f223 : SUCCESS in 27m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d8a60d0bbb134b0896a1ce0e78df90a7 : SUCCESS in 27m 10s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/9f63f65b694349c8b80f064d144468ba : SUCCESS in 2h 12m 50s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/72259e26b9a4420aa3f94f641e38b020 : SUCCESS in 40m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8dc290a9027049a0a002099ceeaad62c : SUCCESS in 9m 45s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/22c22d61603e4ec983c1e167727509f3 : SUCCESS in 2h 19m 30s","accounts_in_message":[],"_revision_number":2},{"id":"d85ed0c6511682554ac86fd51f3e74eec84ca6f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 17:36:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e2ac7c62c48bca8a4e8d841b32a16af9cd70d89a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 17:38:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7238a7b3e0a401eb8771055beed193f : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6fb833882110bdb309213a1e6a2680c9cb8509a8","revisions":{"7b73b06b5f73db51415aa83e6f3c384b548e5c9e":{"kind":"REWORK","_number":1,"created":"2019-09-26 11:16:06.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/36/685036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/36/685036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/36/685036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/36/685036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/36/685036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/36/685036/1"}}},"commit":{"parents":[{"commit":"0de247948b324bd3b0f91b606803a7ec5411423e","subject":"Pass a real session to ironicclient in _get_fake_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0de247948b324bd3b0f91b606803a7ec5411423e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-26 11:15:21.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-26 11:15:21.000000000","tz":420},"subject":"Fix scheduled jobs migration","message":"Fix scheduled jobs migration\n\nChange-Id: I764586d4f93cca42722f7d598a5c9ac0463730c2\nCloses-Bug: #1845482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7b73b06b5f73db51415aa83e6f3c384b548e5c9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7b73b06b5f73db51415aa83e6f3c384b548e5c9e"}]},"branch":"refs/heads/master"},"6fb833882110bdb309213a1e6a2680c9cb8509a8":{"kind":"REWORK","_number":2,"created":"2019-09-27 04:21:15.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/36/685036/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/36/685036/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/36/685036/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/36/685036/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/36/685036/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/36/685036/2"}}},"commit":{"parents":[{"commit":"0de247948b324bd3b0f91b606803a7ec5411423e","subject":"Pass a real session to ironicclient in _get_fake_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0de247948b324bd3b0f91b606803a7ec5411423e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-26 11:15:21.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-27 04:21:00.000000000","tz":420},"subject":"Fix scheduled jobs migration","message":"Fix scheduled jobs migration\n\nChange-Id: I764586d4f93cca42722f7d598a5c9ac0463730c2\nCloses-Bug: #1845482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6fb833882110bdb309213a1e6a2680c9cb8509a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6fb833882110bdb309213a1e6a2680c9cb8509a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
