)]}'
{"id":"openstack%2Fmistral~685478","triplet_id":"openstack%2Fmistral~stable%2Ftrain~Ic3ac2c8ba0ce546fc5654650ac6f5c9b533bdc04","project":"openstack/mistral","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"Ic3ac2c8ba0ce546fc5654650ac6f5c9b533bdc04","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","status":"MERGED","created":"2019-09-27 20:40:16.000000000","updated":"2019-10-09 18:11:22.000000000","submitted":"2019-10-09 18:10:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685478-1570644605795-6aa585c6","meta_rev_id":"f0282c375bdab32552e57eec8e8fd286cb33c494","_number":685478,"virtual_id_number":685478,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 18:10:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-10-09 16:11:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-10-09 16:11:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 16:11:48.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-10-09 18:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51e08a5eb6a20ed7de7370f8f848512db8a0cc1e","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-09-27 20:40:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e7748db1903cc87cbd9ed5e97b5b5bc4ce0cb15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 21:15:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49b349bc19e94181a9648485af8d451e : SUCCESS in 7m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9df650b5b0047b78a81ee52bec990d9 : SUCCESS in 10m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5909a35bbef34e19b23c6558fbdd0200 : SUCCESS in 13m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bb046161f80f4e1ea77864aec8067c3e : SUCCESS in 9m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9d0d7c6c16c445fa97cf6e40cf1ef9c : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b55577a03184841a91622707f610440 : SUCCESS in 9m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c22f5c398f040d89f96a2470e6ce1d3 : SUCCESS in 11m 38s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/84abcc12778c406b9bed883548cba290 : SUCCESS in 27m 42s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0e86b1c04cce42a697697e434d59e317 : SUCCESS in 27m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e668c210e52e4ff6929b431897be9679 : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a58c77d1d11bbb0032bb175631df89804fb21d85","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-30 04:43:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4a0bdf2d2e88f29f74d9e11efa2a13b15a3c943c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-09 16:11:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1a10b8efac66d153d757bd03f74cae19fcd89082","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-09 16:11:48.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2a4b7ea5d0e3e6be852acf4e1d4b3a1273faf6fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 17:20:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9385dd6f18c54e78afe06d501d60c07d : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32bcdb5613c24638a3bc4a7c20d1173b : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b811de89e4b24fc49356d0abcd6b5b07 : SUCCESS in 8m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d62200db7684476ae5aaf752219febf : SUCCESS in 10m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2ed0e93eaed46859303f4069568dcb2 : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31076315150640bbb4ea685cd4995a01 : SUCCESS in 4m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41e63b6373b14e429b3761878ea2f52b : SUCCESS in 8m 46s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/239eaeb39e104bd0a798411fbc43969e : SUCCESS in 40m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6fe6b4a286e542a98aad6dddf0ab68ce : SUCCESS in 42m 26s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e4ef97df8da1405f9e1afc565e93d86e : SUCCESS in 43m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/653136e4c1544fc7a5e1d9ab9e242a40 : SUCCESS in 31m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/44e41df0940f417e801bb29758d9384e : SUCCESS in 26m 52s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/444d87dc2909447e9061590ac2d41ae8 : SUCCESS in 1h 02m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/133d9ee5e3104d2a8358eacd8eef7fbc : SUCCESS in 11m 09s","accounts_in_message":[],"_revision_number":2},{"id":"7aaefb166ac99f49935cad455e883b0edc70a9d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 17:20:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fa606b5219d1268c364d435987bfcc46544bd67d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 18:10:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28850faa316d4a12949421d48b609b33 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0295d9b149934713b63908faa5f545d1 : SUCCESS in 10m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a57643eae3b4f5ea5ab00425fbd73ce : SUCCESS in 9m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ec157b81dc642ab8334c31ac003a3c8 : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7159b0a3582442cb8c5eec83c59cad7 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dd3c2ddc241410aa506d6300a084695 : SUCCESS in 4m 09s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7850eaf7490a47989094880b15c0b776 : SUCCESS in 42m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5f5bee7d97c948588ea3e99a715e423f : SUCCESS in 42m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2aaee9926d204b0dbfb072d5f77f7eb4 : SUCCESS in 27m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/33350e3db5f3460da6a51bd8786be46f : SUCCESS in 27m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/60970cf552de4364ac7899c8a5edbad4 : SUCCESS in 41m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/766ec220b78e463085e692ac43966b06 : SUCCESS in 9m 34s","accounts_in_message":[],"_revision_number":2},{"id":"b23c46a1f8e6f040061891878be3ea30feaa645a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 18:10:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"376a0c98abcf07bb2f3a7b8a65f1956980ac2570","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 18:11:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e313ff712c74462fa6f409f6a7e441b8 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5a76282f13f43ccbc8d6978ef8cdb58 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"016509eb5ea673dd66cff04bb457645be1618653","revisions":{"4385d973ce9db50eb2e746e191f67d183fff5fd6":{"kind":"REWORK","_number":1,"created":"2019-09-27 20:40:16.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/78/685478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/78/685478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/685478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/685478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/685478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/78/685478/1"}}},"commit":{"parents":[{"commit":"fdc7b4162d2c720792331e30b126767d397ef570","subject":"Update .gitreview for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fdc7b4162d2c720792331e30b126767d397ef570"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-09-27 20:40:15.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-09-27 20:40:15.000000000","tz":0},"subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","message":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/train branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ic3ac2c8ba0ce546fc5654650ac6f5c9b533bdc04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4385d973ce9db50eb2e746e191f67d183fff5fd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4385d973ce9db50eb2e746e191f67d183fff5fd6"}]},"branch":"refs/heads/stable/train"},"016509eb5ea673dd66cff04bb457645be1618653":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-09 16:11:41.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/78/685478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/78/685478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/685478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/685478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/685478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/78/685478/2"}}},"commit":{"parents":[{"commit":"41bf7406c06b7a0b9cd2cda57daadb804f8ef48e","subject":"Merge \"Update .gitreview for stable/train\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/41bf7406c06b7a0b9cd2cda57daadb804f8ef48e"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-09-27 20:40:15.000000000","tz":0},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-10-09 16:11:41.000000000","tz":0},"subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","message":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/train branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ic3ac2c8ba0ce546fc5654650ac6f5c9b533bdc04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/016509eb5ea673dd66cff04bb457645be1618653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/016509eb5ea673dd66cff04bb457645be1618653"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
