)]}'
{"id":"openstack%2Fmistral~685899","triplet_id":"openstack%2Fmistral~master~Ic8609e55930e2ed13653e79e8ca7a31c951d9030","project":"openstack/mistral","branch":"master","topic":"bug/1846152","hashtags":[],"change_id":"Ic8609e55930e2ed13653e79e8ca7a31c951d9030","subject":"Fix \"root_execution\" lazy loading issue and refactor execution.py","status":"MERGED","created":"2019-10-01 08:36:17.000000000","updated":"2019-10-03 10:28:14.000000000","submitted":"2019-10-03 10:26:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685899-1570098403144-ef2ef3aa","meta_rev_id":"1b362e2f24fd859788c49e208714e29f783ff1cc","_number":685899,"virtual_id_number":685899,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-03 10:26:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":2,"date":"2019-10-03 08:15:53.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-10-03 09:37:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":1,"date":"2019-10-02 11:02:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":1,"date":"2019-10-03 09:20:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":2,"date":"2019-10-03 08:01:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":1,"date":"2019-10-03 08:15:53.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2019-10-03 01:00:28.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2019-10-03 07:51:48.000000000","_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"date":"2019-10-03 01:00:28.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-03 01:00:15.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-02 08:37:44.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2019-10-03 01:00:15.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-10-03 01:00:28.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-10-03 01:00:28.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-10-03 07:51:48.000000000","updated_by":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"reviewer":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"state":"REVIEWER"},{"updated":"2019-10-03 08:01:51.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-10-03 08:15:53.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-10-03 09:20:20.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2019-10-03 09:37:53.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-10-03 10:26:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"038520df916f0f07d3bc74990d73202e4128db7f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-01 08:36:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b58ee586411e66ab80a1afeb3e85568907baff5","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-01 09:28:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a668dc3b3309f8ba76436e871d23470836ad383f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 13:02:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f402aa48142a4e12a846e0b7f33ee32c : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8dabbfbf5f0b47a49ab06104fa783e80 : SUCCESS in 8m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb2165f72fef43b7bb149abe766f8e17 : SUCCESS in 8m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/264a179c5a4f43af8a20ab334ef49d24 : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe7c7d3e60b40458d2a98f3a293f412 : FAILURE in 4m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0caea4cb79744cb089aea0d18744e041 : SUCCESS in 9m 16s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/38f93f255c5d40fcb52fbf641fd0adbe : SUCCESS in 41m 43s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d84b601079e1414cb84750fccc3f3398 : SUCCESS in 43m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b0ddde4996cf42f7a7774db0bafd7cfa : SUCCESS in 43m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0f9abffe336645b7b51c67e9cdd56b03 : SUCCESS in 28m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ab585e89991f4ab0b1eeb9d20a183782 : SUCCESS in 28m 59s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/bbba78bed8264f32ab1f8f2380c39bcc : SUCCESS in 2h 03m 07s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/525f487c4a0d4241a510fb60aff4bdc6 : SUCCESS in 3h 26m 22s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/78f257d339bd4c2f93563c3074bad5f7 : SUCCESS in 1h 06m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ab49a76bd274d96820e766ede8803d7 : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":2},{"id":"5fe852a36972b5871146a07470d99a732f0a15d6","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-02 05:20:14.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"218974e331497414c37ac3b09245dc95190a8fe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 08:32:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b736dad30af745349163cab829b066b3 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b621a5db3f8743388a44e899091b8c44 : SUCCESS in 8m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/675f67b8d9ed40eab454d926d9a4f39f : SUCCESS in 7m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c4e14a7af1a43f7b95ed3440c12aa36 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e82074664f1423eb8c2f070a23909b2 : FAILURE in 5m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/513467654f38412bb413dafa9cfb0166 : SUCCESS in 10m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cfc5dac80142434a8e591f3421bfa3c5 : SUCCESS in 46m 54s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5bd3b30400724f3baedc74438e0ba0a8 : SUCCESS in 55m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3600aa83b8a24f9cbbd86432600afe4c : SUCCESS in 45m 50s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/eba7b108d5534757897fa54d798fde40 : SUCCESS in 28m 45s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ae69f13e387a4b2fa19e7e5df4672a4e : FAILURE in 29m 40s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/3c03f25431db45c0988aeb18cd8158ec : SUCCESS in 1h 50m 36s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6f659355446541e1947a184ccf1e034b : SUCCESS in 3h 08m 03s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5f61c3bdbb914459a4121d68e469502a : SUCCESS in 1h 10m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/363030e4039747b8912f4870bf3092a5 : SUCCESS in 11m 35s","accounts_in_message":[],"_revision_number":3},{"id":"dcda44daf437672847231d2da02a902bec6704b7","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-10-02 08:37:44.000000000","message":"Patch Set 3: Code-Review+1\n\n\u003e Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c0e50e6887106893834216de86c43c45b7eda492","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-02 11:02:52.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"6d36b164336b3cc431da1519290683b5a60980eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 14:32:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45ed171741b94b7ca543802051612d20 : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd51ce429c954c2d88c31b856860f0d1 : SUCCESS in 9m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9618b5a6d45849629e23c70b424144d5 : SUCCESS in 10m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e268fddc291f4dbcb980ac7bebcdefd5 : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44c66ced563f4a4ca95aafd840a8aeee : SUCCESS in 4m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/800c682ce437440088be4f80b94baf8c : SUCCESS in 8m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/95ac0c0fd97140d681f66ebc340a2c3a : SUCCESS in 51m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b16b10fb7f864267881c206da624aaa6 : SUCCESS in 42m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/60f57b5104044725bb53c27e2bdad6c2 : SUCCESS in 42m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/31e0c8595e5c4506abc0e06824dc9a42 : SUCCESS in 27m 57s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b55e00e1666e43d1a0303eb0a4ef9fb3 : SUCCESS in 25m 50s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/907878f7f3af4f6e8b2e4d93cbeaf95f : SUCCESS in 1h 47m 45s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8b47acb30c1f4b4d9ee438f045af5e04 : SUCCESS in 3h 24m 05s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7e9451b170604aa397339a2438f40766 : SUCCESS in 58m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f425caaa48e24532924efe7273fd9120 : SUCCESS in 10m 18s","accounts_in_message":[],"_revision_number":4},{"id":"092d841b74c21643706dca3b1e8a3ccb08a9a499","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-10-03 08:01:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bd2aae4b9aec55f30d5d19b4a8414a54f5a23783","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-10-03 08:15:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"cae3d3d70852906485790d7e5a471634a59537fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 08:16:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b5927211cd4359e450fd55d80a27cc7edb8d2fab","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-03 08:26:41.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 5f1d28fd73236388f7abdf7474ab19a186bde306","accounts_in_message":[],"_revision_number":4},{"id":"9ec03ae93e2d2ae5b5d7f040f57a3a3d6e0f10b4","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-03 09:20:20.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"816e674980d3eadc057da43b1e4cc46f7812938e","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-10-03 09:37:53.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"915628ab06ddc7f77491506026554e5abbcf036b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 10:26:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7efc36dd789747f7a0aacbfefa0a760b : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf86117f57fd4ba4ad3619789cb0e254 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8dba5cc005642d8a168ea0e9aa3ab82 : SUCCESS in 7m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/234ae31b8bf1498a800d8b99a582b25a : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8fd7ed726644c3bb9655fe7cac9ad8f : SUCCESS in 4m 54s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3522532b23e84c7e8f7f740c44c13fdf : SUCCESS in 43m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/52c7d3ebd20b482a9ff1ca5c74bd1cf8 : SUCCESS in 40m 37s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4abd46fdf6a441b6a26aa594ae14b341 : SUCCESS in 28m 57s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4bbaccdd8eda4a01aabc2a3939c9a467 : SUCCESS in 26m 39s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e3bcef08e3f9437a86eef960385f1d27 : SUCCESS in 1h 43m 38s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f68ed5459338469d91e16c0529ff6879 : SUCCESS in 45m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea96ebe90c6a4ccfad42176ff74ede33 : SUCCESS in 8m 57s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/b3879610b44e4bbba76dd326082e74c8 : SUCCESS in 2h 04m 38s","accounts_in_message":[],"_revision_number":4},{"id":"591dbe8f4a5079eb9ac32028f9ab156312c91855","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 10:26:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"d9fe80682d116168435086790c39f52facc233d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 10:28:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ecf52503315494a82a42a4b8b189f11 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7a6aac0f5f008d67eeb5d3f63ddac42bb62f8e69","revisions":{"ab9aa20c4756d5e3fd785779798e6e06de8c2b87":{"kind":"REWORK","_number":1,"created":"2019-10-01 08:36:17.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/99/685899/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/685899/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/685899/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/685899/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/685899/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/685899/1"}}},"commit":{"parents":[{"commit":"0de247948b324bd3b0f91b606803a7ec5411423e","subject":"Pass a real session to ironicclient in _get_fake_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0de247948b324bd3b0f91b606803a7ec5411423e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-26 10:48:40.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-10-01 08:30:02.000000000","tz":420},"subject":"Fix \"root_execution\" lazy loading issue and refactor execution.py","message":"Fix \"root_execution\" lazy loading issue and refactor execution.py\n\n* There\u0027s an issue with lazyly loaded field of WorkflowExecution\n  model occuring on GET /v2/execution/\u003cid\u003e because the logic\n  that calculates \"published_global\" of the execution rest resource\n  hits \"root_execution\" field out of transaction scope indirectly\n  within the \"data_flow.get_workflow_environment_dict\" method.\n  This patch makes refactoring of this logic and calculates\n  globally published variables of the workflow execution simply\n  as its context that doesn\u0027t contain all internal data like\n  \"__execution\" and \"openstack\".\n* Other style change.\n\nCloses-Bug: #1846152\nChange-Id: Ic8609e55930e2ed13653e79e8ca7a31c951d9030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ab9aa20c4756d5e3fd785779798e6e06de8c2b87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ab9aa20c4756d5e3fd785779798e6e06de8c2b87"}]},"branch":"refs/heads/master"},"3b180d2fe8f743f50de63258df0f229a337e9de8":{"kind":"REWORK","_number":2,"created":"2019-10-01 09:28:34.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/99/685899/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/685899/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/685899/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/685899/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/685899/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/685899/2"}}},"commit":{"parents":[{"commit":"0de247948b324bd3b0f91b606803a7ec5411423e","subject":"Pass a real session to ironicclient in _get_fake_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0de247948b324bd3b0f91b606803a7ec5411423e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-26 10:48:40.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-10-01 09:28:19.000000000","tz":420},"subject":"Fix \"root_execution\" lazy loading issue and refactor execution.py","message":"Fix \"root_execution\" lazy loading issue and refactor execution.py\n\n* There\u0027s an issue with lazyly loaded field of WorkflowExecution\n  model occuring on GET /v2/execution/\u003cid\u003e because the logic\n  that calculates \"published_global\" of the execution rest resource\n  hits \"root_execution\" field out of transaction scope indirectly\n  within the \"data_flow.get_workflow_environment_dict\" method.\n  This patch makes refactoring of this logic and calculates\n  globally published variables of the workflow execution simply\n  as its context that doesn\u0027t contain all internal data like\n  \"__execution\" and \"openstack\".\n* Other style change.\n\nCloses-Bug: #1846152\nChange-Id: Ic8609e55930e2ed13653e79e8ca7a31c951d9030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3b180d2fe8f743f50de63258df0f229a337e9de8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3b180d2fe8f743f50de63258df0f229a337e9de8"}]},"branch":"refs/heads/master"},"ad2891de63790ab12662774eeca240b5f9c2bc84":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-02 05:20:14.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/99/685899/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/685899/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/685899/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/685899/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/685899/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/685899/3"}}},"commit":{"parents":[{"commit":"e4f62dab1d6160f2a41cd05b0ec893d7dcc55989","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4f62dab1d6160f2a41cd05b0ec893d7dcc55989"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-26 10:48:40.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-10-02 05:20:14.000000000","tz":0},"subject":"Fix \"root_execution\" lazy loading issue and refactor execution.py","message":"Fix \"root_execution\" lazy loading issue and refactor execution.py\n\n* There\u0027s an issue with lazyly loaded field of WorkflowExecution\n  model occuring on GET /v2/execution/\u003cid\u003e because the logic\n  that calculates \"published_global\" of the execution rest resource\n  hits \"root_execution\" field out of transaction scope indirectly\n  within the \"data_flow.get_workflow_environment_dict\" method.\n  This patch makes refactoring of this logic and calculates\n  globally published variables of the workflow execution simply\n  as its context that doesn\u0027t contain all internal data like\n  \"__execution\" and \"openstack\".\n* Other style change.\n\nCloses-Bug: #1846152\nChange-Id: Ic8609e55930e2ed13653e79e8ca7a31c951d9030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ad2891de63790ab12662774eeca240b5f9c2bc84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ad2891de63790ab12662774eeca240b5f9c2bc84"}]},"branch":"refs/heads/master"},"7a6aac0f5f008d67eeb5d3f63ddac42bb62f8e69":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-10-02 11:02:52.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/99/685899/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/685899/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/685899/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/685899/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/685899/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/685899/4"}}},"commit":{"parents":[{"commit":"9103b5439dd4e28c9e3341a1434c8a1705502a28","subject":"Add the explicit dependency from kombu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9103b5439dd4e28c9e3341a1434c8a1705502a28"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-26 10:48:40.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-10-02 11:02:52.000000000","tz":0},"subject":"Fix \"root_execution\" lazy loading issue and refactor execution.py","message":"Fix \"root_execution\" lazy loading issue and refactor execution.py\n\n* There\u0027s an issue with lazyly loaded field of WorkflowExecution\n  model occuring on GET /v2/execution/\u003cid\u003e because the logic\n  that calculates \"published_global\" of the execution rest resource\n  hits \"root_execution\" field out of transaction scope indirectly\n  within the \"data_flow.get_workflow_environment_dict\" method.\n  This patch makes refactoring of this logic and calculates\n  globally published variables of the workflow execution simply\n  as its context that doesn\u0027t contain all internal data like\n  \"__execution\" and \"openstack\".\n* Other style change.\n\nCloses-Bug: #1846152\nChange-Id: Ic8609e55930e2ed13653e79e8ca7a31c951d9030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7a6aac0f5f008d67eeb5d3f63ddac42bb62f8e69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7a6aac0f5f008d67eeb5d3f63ddac42bb62f8e69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
