)]}'
{"id":"openstack%2Fmistral~686349","triplet_id":"openstack%2Fmistral~stable%2Ftrain~Ic8609e55930e2ed13653e79e8ca7a31c951d9030","project":"openstack/mistral","branch":"stable/train","topic":"bug/1846152-stable/train","hashtags":[],"change_id":"Ic8609e55930e2ed13653e79e8ca7a31c951d9030","subject":"Fix \"root_execution\" lazy loading issue and refactor execution.py","status":"MERGED","created":"2019-10-03 08:26:41.000000000","updated":"2019-10-09 09:25:08.000000000","submitted":"2019-10-09 09:22:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686349-1570612979180-8e81ebfc","meta_rev_id":"22b784f5d195e24bb7cbdc800606dafa35bd4396","_number":686349,"virtual_id_number":686349,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 09:22:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":1,"date":"2019-10-08 09:44:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-10-08 03:40:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-10-08 03:40:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-07 17:32:45.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-10-08 09:44:10.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-10-09 09:22:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c96631c9f9b7620bd8cfcf2f5e3b59a2d9697d7","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-03 08:26:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8e88557e298d1129c38d8c3a675da05d140efb0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 09:25:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bec2466f8dfc451a93bbd447b3f03012 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dbd5ea2533344de289d432e6042d2c89 : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07fb376ebf09452187faf6b870899c3c : SUCCESS in 9m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a3ab5dda103412ea9cf38699fd9005d : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddfd30b09ac44cd6ab4c05ac40d8d4f7 : FAILURE in 5m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6a2f048f2bd4a7baaebc674884356bf : SUCCESS in 8m 03s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f128f9677aff4b948ddd2d7003f3d0c5 : SUCCESS in 27m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c16e6382ebd14bd0905c4ed464aa136e : SUCCESS in 33m 36s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/10ef56b0faa34d8db1f3baf1cdd6752d : FAILURE in 56m 56s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9c23a48d85314ac8a64d7dd2da52ca62 : FAILURE in 9m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa238a7485e545649a34c44202fab278 : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9f3bfa150f775d0350f666891288ef09349f546e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-07 03:39:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"32bdf26ba566988542b59d62b13df3e7db7ec2d1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-07 03:39:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3857a70c912d77f9c9646a3e500b745656a94c5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 04:23:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5933fb9c4e44997bede74c7b4ad5619 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb181c3af6864dc4b0035a044f4a2913 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab423ce53a764ffaacc73931c692d575 : SUCCESS in 7m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef0644428e584efebb515e13e8f68da3 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/909757d6e79a49b1bed088aedb75f40a : SUCCESS in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35257b3de76c4e3babdba39f4ecfda55 : SUCCESS in 8m 27s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8805f9aed7cb4cad88bcaed933d92069 : SUCCESS in 28m 33s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/97be48d9bef643bfb1bf3f34220a79ee : SUCCESS in 26m 50s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1a545e756e6743c9b70be442e92b688b : FAILURE in 30m 50s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/bcfbce8a97654663947fdc38e3b19e12 : FAILURE in 11m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08419ababfcb4f70b4ad11253f6cb553 : SUCCESS in 9m 58s","accounts_in_message":[],"_revision_number":1},{"id":"982bd7cdc0152e54868f374650e1d1c3e1dd167c","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-10-07 17:32:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b26beff14a31377bd6e222da79a7992e0664c76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 18:37:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17f0870a685c4660b63e928e01c7af10 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/044c5cbdeaa546c09ce6150971316d6f : SUCCESS in 8m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f75a3f4db7144cc8e2016a2fa43894b : SUCCESS in 9m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3ba2d7c2afd45108bb50331f9c99789 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd3f479cef34d3c8ca785cc4d4acd31 : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04bac217c73b47a6a3f68940058136cd : SUCCESS in 9m 08s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ebb1afb6effd43cea3ed932c84efc4e2 : SUCCESS in 48m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a7fba263230d459aba2256ac09dd7457 : SUCCESS in 46m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8d18c4732bbf4f818acf69b06da2fea0 : SUCCESS in 43m 32s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0c797e6422e041719d060679842b4df7 : SUCCESS in 29m 05s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/41b6197803844b76a27045cdda0d98e2 : SUCCESS in 26m 57s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/9619ed0672504e3c9259d553f5ca7754 : FAILURE in 35m 33s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f6f96598571e46799f7b99ae3394444b : FAILURE in 46m 25s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/db605dadf5584ba183f9a312fc6ef092 : SUCCESS in 59m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea05330584c54687a2284e7a6986e27a : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":1},{"id":"6ac2c1b80b9cefb5606524c93464b9cc774689ef","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-08 03:40:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"316959a8ceebf0f80205c7eddf8d2929fd981553","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-08 03:40:35.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7f9e2e3f0bbfa170127a193a427ffc923af2b523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 04:50:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6640cba6c3054e8bb01ffb59cfe6b53d : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ebe5b7de304649378fbb3d9837816235 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/264e2e79c82049b488ec3098e7921490 : SUCCESS in 7m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf39bd3006434c27bad0f8e2c24320c7 : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb37efbd45874bd7a9baf6e408310ab0 : SUCCESS in 4m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c41efe3274cf41048bf70b7eef5a66f4 : SUCCESS in 8m 32s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3d483f9ad01741be939d14fe0ffd3c1e : SUCCESS in 47m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c12d603fc99345f3bdf0ba75f47fa13a : SUCCESS in 52m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dc7454734c1841f6a590d08ba0113ba5 : SUCCESS in 46m 13s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4efe828b6cd747c0addc62210385ef56 : FAILURE in 27m 41s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/462dcd4480c446d493d4241646d74689 : SUCCESS in 26m 54s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/53333dc2937a474f9c8c15460b2b8d32 : FAILURE in 34m 56s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/285a87697b7641a2be73e3d57a5f63e9 : FAILURE in 31m 17s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/dc65798071ea40be9363da314d5d4aa6 : SUCCESS in 1h 03m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6a05b7eda5c498fab96bd253643a0ae : SUCCESS in 11m 33s","accounts_in_message":[],"_revision_number":2},{"id":"04788114c4b856da07c7dbe414ef465d724f19df","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-10-08 09:43:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7579e2675e324c2ecc5af0ebd226029791777dc2","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-10-08 09:43:49.000000000","message":"Removed Code-Review+1 by Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"c15b956eff4bfc9a15ffe5efbc9e18d6abc7c2be","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-10-08 09:44:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"54755fb386a98075a3737e0793ba16117af4f86f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-09 04:24:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ac84b553172fa9ce2613cd51c0ee98e1c6db7017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 07:17:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6ee0209fc364f2c8b238f040565cfe9 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/71b00ca4ab6e4c39a4152855dcb08169 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d45089a1a23a4c04ac9331d6a8771224 : SUCCESS in 8m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abe3f082f02e4493a849ed77054c6a7f : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f019cdc251467a9fa57f4226ad9271 : SUCCESS in 4m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b8fd91ba4454f11a95013d26d5a4011 : SUCCESS in 8m 32s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/50521d607b534a1d91157f4a10d9fee4 : SUCCESS in 40m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d69d41edbff042369df7c44101f7cc9e : SUCCESS in 40m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7c34d4649c114a5a81e7dcd10b609232 : SUCCESS in 40m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1db235d4aadd41bb9c57c156d1312e40 : SUCCESS in 27m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e10eaebd8173429792909b617123ec40 : SUCCESS in 26m 27s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c6c8f99f9de74affb7e68ea4f9a9cf1e : SUCCESS in 1h 46m 06s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/845a4808672a47579e75a81236540567 : SUCCESS in 2h 43m 14s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6035a1d75b534e808d7d7de3e342214f : SUCCESS in 1h 01m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2e4a6f8bb9549749b89fe6d0e68dc84 : SUCCESS in 9m 12s","accounts_in_message":[],"_revision_number":2},{"id":"c814e8110a30a40a112333733820d351498e2656","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 07:18:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1eaf7d2fbba5e2d08c187a7ebac9a50f75cb84f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 09:22:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/661922628c6d41f1832d91ad91b018d1 : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eef5ce3ece96479485ec8b17cd8f5784 : SUCCESS in 12m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c80d52cf8fa848abb629a945d577c5ba : SUCCESS in 7m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03c42949da544ee7b1911989256b865c : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca614c2cfc394b2b8294583777bddc05 : SUCCESS in 8m 08s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/44e044e148a34ad28f108e4a828c00f8 : SUCCESS in 43m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f3ce9b3bdfe44d68a8c3aecd67582d5d : SUCCESS in 39m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/04eb74ceab304bfe824d35aab1a62aea : SUCCESS in 31m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2080e3abc70d4081b2f761888d26ef8f : SUCCESS in 26m 35s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/3d518b6205534773bc6b31be7f582430 : SUCCESS in 2h 04m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8ba65246823a41329fa01ea4b4f77249 : SUCCESS in 51m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86e60723c00d4c14a508bc001fd44b2b : SUCCESS in 10m 26s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/49f655be8e3e441993f2c78a900ee191 : SUCCESS in 2h 02m 04s","accounts_in_message":[],"_revision_number":2},{"id":"59c30bd479e713a4f403f30d7a07638acf87e5c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 09:22:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"52ec614a5d84d1bf261c31c4bd1f986ffd650ecb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 09:25:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59d312e171f84021816a93507d8208cb : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"33d4a8bba35d79e8216d5babce48bbd04acfdfbd","revisions":{"5f1d28fd73236388f7abdf7474ab19a186bde306":{"kind":"REWORK","_number":1,"created":"2019-10-03 08:26:41.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/49/686349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/49/686349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/686349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/686349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/686349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/49/686349/1"}}},"commit":{"parents":[{"commit":"4f0c4c8c2a0a2dd7d5f3ca273017af5f39a9df81","subject":"Merge \"Remove volumes.promote and volumes.reenable action from cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4f0c4c8c2a0a2dd7d5f3ca273017af5f39a9df81"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-26 10:48:40.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-10-03 08:26:41.000000000","tz":0},"subject":"Fix \"root_execution\" lazy loading issue and refactor execution.py","message":"Fix \"root_execution\" lazy loading issue and refactor execution.py\n\n* There\u0027s an issue with lazyly loaded field of WorkflowExecution\n  model occuring on GET /v2/execution/\u003cid\u003e because the logic\n  that calculates \"published_global\" of the execution rest resource\n  hits \"root_execution\" field out of transaction scope indirectly\n  within the \"data_flow.get_workflow_environment_dict\" method.\n  This patch makes refactoring of this logic and calculates\n  globally published variables of the workflow execution simply\n  as its context that doesn\u0027t contain all internal data like\n  \"__execution\" and \"openstack\".\n* Other style change.\n\nCloses-Bug: #1846152\nChange-Id: Ic8609e55930e2ed13653e79e8ca7a31c951d9030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5f1d28fd73236388f7abdf7474ab19a186bde306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5f1d28fd73236388f7abdf7474ab19a186bde306"}]},"branch":"refs/heads/stable/train"},"33d4a8bba35d79e8216d5babce48bbd04acfdfbd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-08 03:40:25.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/49/686349/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/49/686349/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/686349/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/686349/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/686349/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/49/686349/2"}}},"commit":{"parents":[{"commit":"a655ff98a6e41766d85f1f154b98dcdc7e0b74ff","subject":"Add the explicit dependency from kombu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a655ff98a6e41766d85f1f154b98dcdc7e0b74ff"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-26 10:48:40.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-10-08 03:40:25.000000000","tz":0},"subject":"Fix \"root_execution\" lazy loading issue and refactor execution.py","message":"Fix \"root_execution\" lazy loading issue and refactor execution.py\n\n* There\u0027s an issue with lazyly loaded field of WorkflowExecution\n  model occuring on GET /v2/execution/\u003cid\u003e because the logic\n  that calculates \"published_global\" of the execution rest resource\n  hits \"root_execution\" field out of transaction scope indirectly\n  within the \"data_flow.get_workflow_environment_dict\" method.\n  This patch makes refactoring of this logic and calculates\n  globally published variables of the workflow execution simply\n  as its context that doesn\u0027t contain all internal data like\n  \"__execution\" and \"openstack\".\n* Other style change.\n\nCloses-Bug: #1846152\nChange-Id: Ic8609e55930e2ed13653e79e8ca7a31c951d9030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/33d4a8bba35d79e8216d5babce48bbd04acfdfbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/33d4a8bba35d79e8216d5babce48bbd04acfdfbd"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
