)]}'
{"id":"openstack%2Fmistral~686353","triplet_id":"openstack%2Fmistral~master~Id6521ba37dc5ccff727b06aecd573ac600ea8711","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"Id6521ba37dc5ccff727b06aecd573ac600ea8711","subject":"Log the original exception in is_sync","status":"MERGED","created":"2019-10-03 08:31:54.000000000","updated":"2019-10-09 16:52:36.000000000","submitted":"2019-10-09 16:51:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686353-1570639875786-9bbf562d","meta_rev_id":"431e56f0402b15ef61cc461465208f5b3710a30b","_number":686353,"virtual_id_number":686353,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 16:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":2,"date":"2019-10-09 14:50:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2019-10-03 12:12:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":1,"date":"2019-10-03 12:25:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":1,"date":"2019-10-09 14:50:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 12:12:40.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-10-03 12:25:29.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-10-09 14:50:12.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-10-09 16:51:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bfc1001853ac5e33d53742f5638a6704c6791716","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-10-03 08:31:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99bdb258932334a3d7b2a5f60b34f3aba36ee75a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 11:56:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7daa257afc6c41b1960a4cbddbfd6606 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7c47006ff954a20965e99be1e67692d : SUCCESS in 8m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f7beeb137de442780615d5454a0dc5a : SUCCESS in 9m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bb628d93c11c4cf3ac74ea84df5c7405 : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a59a0d17ce394137b91f5d6fed10b5ab : SUCCESS in 5m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9174692be1847a5a58211af9d56cffd : SUCCESS in 10m 09s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e1ba260413f24a42bc366b3efffafcf8 : SUCCESS in 50m 51s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7fe61c258bf74b17be602ee42fa1f3ad : SUCCESS in 41m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f4ce8e2df8cd473a8cc505c5c5c18f9e : SUCCESS in 46m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c75fcee602544548a800673d46b2a956 : SUCCESS in 29m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/850559f5c4484cae83e4045174c86160 : SUCCESS in 26m 38s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/381a0975a00246eaa6515bbeef8c204c : SUCCESS in 1h 49m 56s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d61f5ee328144fea967d92aa161e3f93 : SUCCESS in 3h 18m 01s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1e7e774ae8fd4d7ca71ad61b1d8caf55 : SUCCESS in 1h 07m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1f451a080af4c95b3ebc5eecc834ea8 : SUCCESS in 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"01809324afbf6ec40a81713122e7c871bd35c4fe","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-03 12:12:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"55ec20f2e89fa120fca597b3171a86886b391921","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-10-03 12:25:29.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e94046edd0033f9f0db66da85eee830b46ae8bd","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-10-09 14:50:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d0b2c1860213da6420e2b89712b322248c357ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 14:50:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee50c08d243e35504a6b073e3803ab406dba8ea0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 16:51:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"796b30ad49c67e6d2b87ca28d9d9feb9451afe24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 16:51:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd6b28f5d1e14e28bf8ff372e167a75b : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/337543654d8b4e8d94a48309e77df9e0 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c10185357df24baab0dbf21d2eeda930 : SUCCESS in 9m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e3a33a29f6949c297e6dbc78118826d : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4591c842ee940408165546b5d14adc6 : SUCCESS in 5m 19s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4c34a925cc574a13995a48f68d7c7d3d : SUCCESS in 52m 59s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/46514cfa1a5544d39cec3f1bf2821f10 : SUCCESS in 51m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/da8266503b6d4818aacecc4497ddf792 : SUCCESS in 28m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2dd8b81778184465825881d9f83127d9 : SUCCESS in 28m 20s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/bed897dd34d54ebaaca27cf93fa573ae : SUCCESS in 1h 34m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cc06032597cf4dceb644b7b84e4eaec8 : SUCCESS in 40m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d5d5e06d399453696ed19f2bcebe93b : SUCCESS in 9m 51s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/1c89666258ad44e49d290c6ead39ec35 : SUCCESS in 1h 59m 26s","accounts_in_message":[],"_revision_number":1},{"id":"431e56f0402b15ef61cc461465208f5b3710a30b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 16:52:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db1f8a7ca1b24b8cbd340a87e276128b : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c0857a7a952c4da04a85dbdde962d37ded00dfff","revisions":{"c0857a7a952c4da04a85dbdde962d37ded00dfff":{"kind":"REWORK","_number":1,"created":"2019-10-03 08:31:54.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/53/686353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/686353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/686353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/686353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/686353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/686353/1"}}},"commit":{"parents":[{"commit":"9103b5439dd4e28c9e3341a1434c8a1705502a28","subject":"Add the explicit dependency from kombu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9103b5439dd4e28c9e3341a1434c8a1705502a28"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-10-03 08:31:29.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-10-03 08:31:29.000000000","tz":60},"subject":"Log the original exception in is_sync","message":"Log the original exception in is_sync\n\nChange-Id: Id6521ba37dc5ccff727b06aecd573ac600ea8711\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c0857a7a952c4da04a85dbdde962d37ded00dfff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c0857a7a952c4da04a85dbdde962d37ded00dfff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
