)]}'
{"id":"openstack%2Fmistral~686354","triplet_id":"openstack%2Fmistral~stable%2Ftrain~I990d125e86943c019dc1b60070cb67c4f75787c0","project":"openstack/mistral","branch":"stable/train","topic":"fix_requirements-stable/train","hashtags":[],"change_id":"I990d125e86943c019dc1b60070cb67c4f75787c0","subject":"Add the explicit dependency from kombu","status":"MERGED","created":"2019-10-03 08:33:00.000000000","updated":"2019-10-03 11:38:52.000000000","submitted":"2019-10-03 11:37:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686354-1570102643747-01385176","meta_rev_id":"cea53af20d43f9f7a35795a695b9510667d297c8","_number":686354,"virtual_id_number":686354,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-03 11:37:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":1,"date":"2019-10-03 10:12:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2019-10-03 09:19:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2019-10-03 09:19:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 10:12:00.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2019-10-03 11:37:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"690929110f6821c66e236b3eb141d96905174512","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-03 08:33:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"56fe30b034e3e8ef0127c2494eedd2aa2960d63b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 09:06:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c16d263996564906bf3962a4014cff00 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab67b8f5b1a248ebb03e86c7cad1df9b : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dbb1c6c2b454be0982df6dd99395763 : SUCCESS in 9m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d1b1318be6f40e6ac6835b4a2ce7a68 : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c4d67406a643cb8584a829c196d976 : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a1a7e7cbe724e8bac7e772e67513d90 : SUCCESS in 9m 04s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9f235eee4da34df6b114cd72895dbbb1 : SUCCESS in 30m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5796816337fa41518032a7d3450011b3 : SUCCESS in 29m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7010c2260ff24dfbbfd052f0368a78c2 : SUCCESS in 10m 29s","accounts_in_message":[],"_revision_number":1},{"id":"88d5e4841ad59481abcb857b75d3a87a872a54a3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-03 09:19:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f66201fbe4955b20902a5b429bcc428d1d2cadc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 09:20:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af3629eb74041bff651bd29dba54fc249b944d8a","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-10-03 10:12:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e4b472d9a1ab8be65ca04197e07735a7cdf1e0d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 10:26:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a05e8b9d1de429190d9b3f481d46ae1 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d4ec21a24f84ba0859de3efc8da26da : SUCCESS in 12m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/256587a4ae54426caf911d23fb9a36e9 : SUCCESS in 8m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b187916563747e190bd21d7d485d7ca : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0c05c7e62cc47f5b1c61c0d913d85f7 : SUCCESS in 5m 14s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2482f6f3a67e4790ae2d137f779fcf40 : SUCCESS in 35m 00s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ec334218a39343ca9c30738c3ddda395 : FAILURE in 29m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bb87f9a3ef843268adb07805fac1afa : SUCCESS in 10m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f8f1f78e5c44f544b3425c9d59c0dd50ebf3bcd8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-03 10:29:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"34087df6d26dd7c19d205ffc05a522bd24983f3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 11:04:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f58134ebf7264332871b84512487cbce : SUCCESS in 6m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66d1091a8431455cbec58dcd90fb30bb : SUCCESS in 11m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/490612416a514340a0d4aa14120501b5 : SUCCESS in 8m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/705f4993f403421baf42ecb3a49c71c5 : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa0887a0c1a4b99b8c4607fe0cc5922 : SUCCESS in 5m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20cd3ed937c843fb8db4272b29b13174 : SUCCESS in 10m 43s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/44cd9246e3dc4606b72e16161de61717 : SUCCESS in 28m 10s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e77627e331e84830ae44eda4f962362e : SUCCESS in 28m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdceea9306e745e091a215c29d96df5f : SUCCESS in 9m 49s","accounts_in_message":[],"_revision_number":1},{"id":"42ec249f340fdeb48b52caea76f690994906ac94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 11:04:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d45a6084bdde4190bd5d7582cc0740bc8352195","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 11:37:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"826573b02242995e685898166b397b6860d09d0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 11:37:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e08080dff7241a5b6c82d7bfb5170c8 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/796586164df74848864a1d0251e8bf45 : SUCCESS in 8m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14ed4061769a412ba5715eef47ac4466 : SUCCESS in 11m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa6114981c6b4f77a3775f6adb044e15 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dbe7fb0c95945b18e39870f93ae994d : SUCCESS in 5m 50s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2eed0c6b7aac40c7a2f449a9c57568a7 : SUCCESS in 29m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c303a985f37f43f19e64484061350c7e : SUCCESS in 27m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c08b0220f594dfab7f9fe019d431af5 : SUCCESS in 11m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b749a7d4c63e28e4225376e00a1c544d85beaec2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 11:38:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df97f4a2bc384121a6c99f8b1fb2b21f : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a655ff98a6e41766d85f1f154b98dcdc7e0b74ff","revisions":{"a655ff98a6e41766d85f1f154b98dcdc7e0b74ff":{"kind":"REWORK","_number":1,"created":"2019-10-03 08:33:00.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/54/686354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/54/686354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/54/686354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/54/686354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/54/686354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/54/686354/1"}}},"commit":{"parents":[{"commit":"4f0c4c8c2a0a2dd7d5f3ca273017af5f39a9df81","subject":"Merge \"Remove volumes.promote and volumes.reenable action from cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4f0c4c8c2a0a2dd7d5f3ca273017af5f39a9df81"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-10-01 10:46:56.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-10-03 08:33:00.000000000","tz":0},"subject":"Add the explicit dependency from kombu","message":"Add the explicit dependency from kombu\n\n* The explicit dependency from kombu forces the right version of\n  amqp to be downloaded. Otherwise the \"docs\" job fails with a\n  version conflict because it downloads amqp 2.5.2 but the\n  upper constraints allow only 2.5.1. This is rather a workaround\n  that unblocks CI until https://review.opendev.org/#/c/681382/\n  is merged\n\nChange-Id: I990d125e86943c019dc1b60070cb67c4f75787c0\n(cherry picked from commit 9103b5439dd4e28c9e3341a1434c8a1705502a28)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a655ff98a6e41766d85f1f154b98dcdc7e0b74ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a655ff98a6e41766d85f1f154b98dcdc7e0b74ff"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
