)]}'
{"id":"openstack%2Fmistral~687880","triplet_id":"openstack%2Fmistral~master~I6b188e315007391c738b8f821fad91c4fd8572ae","project":"openstack/mistral","branch":"master","topic":"global_publish","hashtags":[],"change_id":"I6b188e315007391c738b8f821fad91c4fd8572ae","subject":"Fix the global publish for task","status":"MERGED","created":"2019-10-10 12:41:57.000000000","updated":"2019-10-11 10:11:38.000000000","submitted":"2019-10-11 10:10:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687880-1570788630270-0f5a3271","meta_rev_id":"97e869ccc26e235bccae79ff1a42a179f9c54aed","_number":687880,"virtual_id_number":687880,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-11 10:10:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":2,"date":"2019-10-11 08:13:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":2,"date":"2019-10-10 16:21:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":1,"date":"2019-10-11 08:13:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-10 13:31:00.000000000","updated_by":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"reviewer":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"state":"REVIEWER"},{"updated":"2019-10-10 16:21:14.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-10-11 08:13:51.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-10-11 10:10:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe5d386c57be13e186a800b768c7670d0380d0a4","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-10 12:41:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad99eeb0893e2d37104555b2018d071cdeb11b64","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-10 12:42:50.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit b3ea7a5b81fb8d1ce7d2a38af67d2ec960a92966","accounts_in_message":[],"_revision_number":1},{"id":"117763fe39b1e61ac684a8e4b3b78fc0c2b4adb4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-10 13:36:49.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d53017cf92f4fa93384eae58ad736cc1f4d6a35","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-10 14:15:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c2bc4b025b6079dc793f4858cbbe71a5fb013354","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-10 16:21:14.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks Eyal! Perfect","accounts_in_message":[],"_revision_number":2},{"id":"d05d30c613db77aa1ab1cc6e20c8736b5a26e655","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 17:25:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f620f95f54164f41bd5f8e1484ef8bac : SUCCESS in 10m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fa05f53ea9e43749cbb6a95988017a7 : SUCCESS in 10m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7fe3cbd659f4742ad1c73d3c9c6240b : SUCCESS in 9m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53dc2f0c09ee44c4a1826f0cc5fdecb0 : SUCCESS in 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/602bf4c64e8f47d3aaaa2af6017b14d9 : SUCCESS in 9m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12fd916f8abc4967848f19131a72e830 : SUCCESS in 10m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3844b383a46842618f3307f44d5ca4ca : SUCCESS in 44m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d64502d8ec784796ba787aa9a4a2fd42 : SUCCESS in 43m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/39ffcb6387474b138938bda1db2ae836 : SUCCESS in 42m 38s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/21c1dd203e2d42f99a85429560c8d42b : SUCCESS in 31m 23s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/049ae336b6f74cbea9ac05db73613fae : SUCCESS in 32m 04s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/59e8bdb12b74407c80c481a33febd622 : SUCCESS in 1h 43m 15s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/46f2518e6ca84ad59b320326a1547654 : SUCCESS in 3h 00m 21s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7ded16d3faa747cca6d24f9cb89d6d42 : SUCCESS in 1h 04m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fe6fff239ec4cee9ca13fa6c95c7bb3 : SUCCESS in 11m 24s","accounts_in_message":[],"_revision_number":2},{"id":"de9e01c638f76d42e8b32b7d5c718e85b60a8089","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-10-11 08:13:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0decbf0a896e8aac26bd8ee44e976a51bd46f5bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 08:14:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"109573098bb3f9001a0e0fb0d805568e888eb7e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 10:10:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca84b63e57844113adbeb091b5645836 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19d973ae98784968a0d7987ce8cc296e : SUCCESS in 8m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56bef6db72c44e54b8f58d210a1a7d07 : SUCCESS in 8m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b525788e20714a96b29a09143a419a40 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86e70722e2a84470aaa1499088dc95b2 : SUCCESS in 5m 22s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/725b3d8d9d6b40669d6aac6d3b7681e1 : SUCCESS in 48m 07s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9d813d5381684b8fb552c717c82c564e : SUCCESS in 39m 38s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/aad31418d1a04e679c3e4c4b20c01ed3 : SUCCESS in 28m 05s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f1962f38265344a0960b5f77e275784b : SUCCESS in 27m 28s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/dceff1779a6d4dd881fb98ab3c7e4050 : SUCCESS in 1h 40m 15s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ff63cb79f1924429b57ad0bf3d694b3f : SUCCESS in 43m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d91da02e8c584e9b84fb879e5689b633 : SUCCESS in 10m 14s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d2cdfbb0065143b6bdad77bee7889eae : SUCCESS in 1h 47m 47s","accounts_in_message":[],"_revision_number":2},{"id":"1e586aa705b3f935fcf33be57b1dde418057bffc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 10:10:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"254da6ebbbc6f6c21357a388abdaf4b64c589d39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 10:11:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be96c96ccbb941c78f67cc2d5d0f7ae3 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ab4d4a5807a9449b6a701e90c198d2cf4328bda6","revisions":{"cb4846e2815300fa3027e003b7fb69fd7f7d910a":{"kind":"REWORK","_number":1,"created":"2019-10-10 12:41:57.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/80/687880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/80/687880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/687880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/687880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/687880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/80/687880/1"}}},"commit":{"parents":[{"commit":"772043881da6b33768a48a46addfc545dcc4afb2","subject":"Merge \"Log the original exception in is_sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/772043881da6b33768a48a46addfc545dcc4afb2"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-10 11:44:33.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-10 12:41:39.000000000","tz":180},"subject":"Fix the global publish for task","message":"Fix the global publish for task\n\nIf there is a root execution then lazy load the\nparams field of the root execution\n\nChange-Id: I6b188e315007391c738b8f821fad91c4fd8572ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cb4846e2815300fa3027e003b7fb69fd7f7d910a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cb4846e2815300fa3027e003b7fb69fd7f7d910a"}]},"branch":"refs/heads/master"},"ab4d4a5807a9449b6a701e90c198d2cf4328bda6":{"kind":"REWORK","_number":2,"created":"2019-10-10 14:15:51.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/80/687880/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/80/687880/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/687880/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/687880/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/687880/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/80/687880/2"}}},"commit":{"parents":[{"commit":"772043881da6b33768a48a46addfc545dcc4afb2","subject":"Merge \"Log the original exception in is_sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/772043881da6b33768a48a46addfc545dcc4afb2"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-10 11:44:33.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-10 14:15:39.000000000","tz":180},"subject":"Fix the global publish for task","message":"Fix the global publish for task\n\nIf there is a root execution then lazy load the\nparams field of the root execution\n\nChange-Id: I6b188e315007391c738b8f821fad91c4fd8572ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ab4d4a5807a9449b6a701e90c198d2cf4328bda6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ab4d4a5807a9449b6a701e90c198d2cf4328bda6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
