)]}'
{"id":"openstack%2Fmistral~687933","triplet_id":"openstack%2Fmistral~stable%2Frocky~I793fbfec03032d9ff7137c11becb6d1c18ec54bc","project":"openstack/mistral","branch":"stable/rocky","topic":"bug/1843175-stable/rocky","hashtags":[],"change_id":"I793fbfec03032d9ff7137c11becb6d1c18ec54bc","subject":"Fix error validate token when run cron trigger","status":"MERGED","created":"2019-10-10 15:12:59.000000000","updated":"2019-10-16 10:09:54.000000000","submitted":"2019-10-16 10:09:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687933-1571220594009-f9e06b29","meta_rev_id":"46f0be649d521a050295baf405670c834eefad4c","_number":687933,"virtual_id_number":687933,"owner":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"date":"2019-10-10 15:13:13.000000000","_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"date":"2019-10-15 16:44:30.000000000","_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},{"value":0,"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 10:09:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"all":[{"value":1,"date":"2019-10-16 02:00:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},{"value":1,"date":"2019-10-10 15:20:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":2,"date":"2019-10-16 04:07:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},{"value":0,"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":1,"date":"2019-10-16 04:07:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-10 15:13:13.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-10-10 15:20:03.000000000","updated_by":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"reviewer":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"state":"REVIEWER"},{"updated":"2019-10-16 02:00:16.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2019-10-16 04:07:53.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-10-16 10:09:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5ae1ffce9ec7ca49e7ee42b91efc525a541a08d7","author":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"date":"2019-10-10 15:12:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"38cae1e77e192730b59495e18a5efb5dc4056bc8","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-10-10 15:13:13.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"5f5a343e8828518f3312c7b941610a4e3cfb12d8","author":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"date":"2019-10-10 15:20:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b7a9074712a96582f38ce7c99dff906df4382e3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 18:11:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9425da4b1fc6472e83fec2504549ad23 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d46b70e36dd401d90be2d346acdc239 : SUCCESS in 8m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a36a52f8e67c42c798fbf450a2d53634 : SUCCESS in 7m 59s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ec6958fde73942d69c4cbe58d3751963 : SUCCESS in 4m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9d7bed6ba694616a367cdd1a6618fc7 : SUCCESS in 7m 44s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8a9a72f3b34c49c9806183058785602a : SUCCESS in 53m 34s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6874da8b14964ce4ae0f87c98638f4a1 : SUCCESS in 50m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f9e66c12706546f2b7fbec0444273876 : FAILURE in 53m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bf443d2b240642f6aa5d8508208a2b85 : SUCCESS in 27m 28s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/163e93f602f0463293c54f7a2dd5cf89 : SUCCESS in 27m 03s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1144720c3e624e62a1a571c6c4e86d8e : SUCCESS in 2h 51m 10s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7c0370ebec324e96ae767e5c7e10af04 : SUCCESS in 59m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbe9bbb4391c44488e1374d32fd00135 : SUCCESS in 12m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6acb15a83d6485762533855b2844844956a13909","author":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"date":"2019-10-10 19:32:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8460422d0c2edd99bb87e648521d6779d80619b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 22:23:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/529efff9a17c43aab767401fb5543f05 : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fad5a95d476e435d87f1f5055b4c321b : SUCCESS in 8m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e18569bbcc774aa69d25f2b3972ac05f : SUCCESS in 9m 36s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9766d2d5727741b9a0d710b412db8ff2 : SUCCESS in 4m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f57e51a156fe48508325cb5b4e891f71 : SUCCESS in 11m 20s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/85adce01774348be9625fdf43ca83029 : SUCCESS in 47m 54s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/848a65b1951e42318c37833f1a565390 : SUCCESS in 43m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f5cff00eee18485d8553b5646d7683de : FAILURE in 49m 31s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cd81033faa814547afa67e799fa7ab86 : SUCCESS in 25m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b0724f4fea3e43bc867e51be73009bea : SUCCESS in 27m 30s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a448cd077d9a4f608d48cccfe4a80e7e : SUCCESS in 2h 41m 59s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/cf251b3a944b412699eb7fb1d3f96a06 : SUCCESS in 1h 00m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96186223da5a4589a26e44fb3d4623c1 : SUCCESS in 11m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a369b55babfb4867656d099dff6b315dcb3419a4","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-10-11 02:02:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"418c2fa71c01ce857ca0f9b842b252aa25c0dcae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 05:00:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2457df761c04031b03f0e8aecebecd0 : SUCCESS in 7m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa05741be49c455f9139cc3a7b6a7ac2 : SUCCESS in 8m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/399ac656e064427486c56778217f73d0 : SUCCESS in 8m 38s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8e64e5ee21534e06b13a692fe8846190 : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a39d38d51dac4328bb6efad93cd86c8e : SUCCESS in 9m 00s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d631515a2d2943c5b7e390d930a69483 : SUCCESS in 40m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/36a1a373008a488dba0f009f604d8fa8 : FAILURE in 54m 27s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/56e4472ef5424096a2487f59a7a2d707 : SUCCESS in 49m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/34e961b54a9547e3b3880206fce33f49 : FAILURE in 28m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0a9379d1d2e44c159bece801db0c024a : SUCCESS in 27m 25s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4aa472baf9984fb69343ff218689cb43 : SUCCESS in 2h 41m 58s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6895feb5ad5e4c32a23ddbfc3199fddb : SUCCESS in 49m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b4f0ab33f2149c08ffe8a1bc22d1e2a : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0c436eb44c0100072251eb01b376af90f4b09d11","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-11 08:21:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c5ebb5f72e94ec26b2a9e55052ce0be5a277a190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 11:10:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4fa9ae3fe854283836b27142d7b089f : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4262bba96d974a01a4aea7696d70593f : SUCCESS in 9m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1217da493eab461aa6371f53dba5bc7e : SUCCESS in 8m 16s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0a1f7f0b4f8c46d4a9f7a366d4ca20c6 : SUCCESS in 5m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97d427e393474b798587c163087bbade : SUCCESS in 9m 08s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4fd2fdbc434547799fa2326f36be1db5 : SUCCESS in 44m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/90790363fcfc4fcc83952028f8d7f4e6 : SUCCESS in 47m 37s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bbfcbe4ffd884bc7bcdd04b68d2ba4c5 : FAILURE in 45m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/925e876a0c364da1958e74b168414b99 : SUCCESS in 26m 51s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/265a2dfa2d8a47c583f8a0294a3e5099 : SUCCESS in 26m 08s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c5325b85ab6046c0a67acfd20d6e6e64 : SUCCESS in 2h 46m 54s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/db163441580140ad86adfe045c5513b6 : SUCCESS in 56m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d0c299db14c499b9b8f44ddc32b6078 : SUCCESS in 10m 15s","accounts_in_message":[],"_revision_number":1},{"id":"647edb9d9d7ea482cdec1709fff4036146d687a7","author":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"date":"2019-10-11 18:54:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0b9c300b6001ab59615e0db8a5b18b642747a336","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 23:23:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c26486314123480c88fe650380f1e26c : SUCCESS in 8m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a36b31ae84c94f809dbf8043df167be0 : SUCCESS in 8m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1b4320680b9e413eabe72db750ca9f46 : SUCCESS in 9m 13s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/83579aab17e1454486270428ffee676b : SUCCESS in 4m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7392deb90e7b4a539affd6d500de7e04 : SUCCESS in 10m 26s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/773da2e0ddce4977a7c30d0db011e881 : FAILURE in 52m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/900da5fdcc9e4303b451e3e6d31e28c1 : SUCCESS in 45m 03s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7055d058dd3f4794a7b8a39d65607456 : FAILURE in 48m 21s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1dffc5292d244733b4da07e1bcba4f84 : SUCCESS in 28m 41s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f8389824af664918965fe63794f4350a : SUCCESS in 26m 05s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6a4b1bf97b6543559c3a18fc6f689576 : SUCCESS in 2h 21m 05s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/dd3c9ac4e2a84cd7acd5b19fff2e2837 : SUCCESS in 56m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a023df3d8c634351945798a4fe2732d4 : SUCCESS in 11m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f2dedead34945230fe77b954477da5db9187b0c6","author":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"date":"2019-10-15 16:44:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d6e571d252e0b34ce2d8fabf163054aea3a9c2ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 17:46:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af895dc7ad234c4ca260e5848fc6c49e : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/925225ff60ac4cf39b6971309c57bb65 : SUCCESS in 9m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a7620ad6e6104ba780fa26c4131b2f20 : SUCCESS in 8m 40s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4d32ba46b9d0494c986182f86bc68bc4 : SUCCESS in 5m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2876d71ff9a3406a919550f888501b4d : SUCCESS in 9m 14s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b769b10eaa0e469aa8365f7fe5bba474 : SUCCESS in 49m 59s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/228d9ae7a63e4269b3b639569cc56f42 : SUCCESS in 46m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6f4390359c464deb89718b6cfdec0747 : SUCCESS in 52m 05s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3c09bd9ecef44b34b5ca1fe7e5fa4156 : SUCCESS in 26m 07s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0ea5a6388c5a4a08a3510b473f647db0 : SUCCESS in 26m 58s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a66e1708ad574bdab0b9010c8fbfae51 : FAILURE in 35m 57s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/22bb625170bf4e6b8c46ba59ab120ab7 : SUCCESS in 55m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf2fb6dd908247d4a9b9c92ed3653017 : SUCCESS in 9m 12s","accounts_in_message":[],"_revision_number":1},{"id":"713fe7ce0011bf9b8422187cc5c8023fbd7a8212","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-10-16 02:00:16.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e recheck","accounts_in_message":[],"_revision_number":1},{"id":"c8d42ff4acf94259d0e173c2d9e8ec2bfd1e44fd","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-16 04:07:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nPlease keep in mind that if we backport something to a stable branch we also need to write a release note and also backport it. So please do it as a follow-up on this patch.","accounts_in_message":[],"_revision_number":1},{"id":"737bc8cd0d2ee5233b188b1861f938ec666784a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 04:08:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43ffff1bf7a7f66523e965b4059e74429717d1a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 10:09:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80cd240152c04b3e804d8b3c8d409fae : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7780feb2674c4b78a5273239a1482e2a : SUCCESS in 7m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c1867b48924b40f2863a704e929394d9 : SUCCESS in 7m 41s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1880c360a0424343963ad4b5061ae077 : SUCCESS in 5m 31s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e21b6bcebec04ace956c8f1064aa3533 : SUCCESS in 44m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a247b46302aa414f85e15d1d635c40d0 : SUCCESS in 40m 26s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4aaea5fc6dae4e269be4571a6567aa83 : SUCCESS in 26m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e2df7e6692604aad92708a71ac3a79a8 : SUCCESS in 25m 40s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2c8319e1c2b0436ba7a56a2cd770c506 : SUCCESS in 2h 31m 03s (non-voting)\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4558f3674ab04ea2b6270ed79e9e4ff8 : SUCCESS in 46m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3b76ca5b85f4c5da3095624e778ad5f : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4d5ff39b2164ca90f2b28e687d1357e161a03da8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 10:09:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1b8dd1635aa45d3d644b356f883484997f31ca7","revisions":{"d1b8dd1635aa45d3d644b356f883484997f31ca7":{"kind":"REWORK","_number":1,"created":"2019-10-10 15:12:59.000000000","uploader":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"ref":"refs/changes/33/687933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/33/687933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/687933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/687933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/687933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/33/687933/1"}}},"commit":{"parents":[{"commit":"f2a0b569f767e0f7cea844ed01accccf810ca618","subject":"disable triggering version discovery in test_generator unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f2a0b569f767e0f7cea844ed01accccf810ca618"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-09-08 13:32:25.000000000","tz":420},"committer":{"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","date":"2019-10-10 15:12:59.000000000","tz":0},"subject":"Fix error validate token when run cron trigger","message":"Fix error validate token when run cron trigger\n\nA trust client can\u0027t do validate token when run cron trigger\nThis patch will fix that.\n\nChange-Id: I793fbfec03032d9ff7137c11becb6d1c18ec54bc\nCloses-Bug: #1843175\n(cherry picked from commit 51b7dd0c6d5c4d0de354719765968ca883b9f8fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d1b8dd1635aa45d3d644b356f883484997f31ca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d1b8dd1635aa45d3d644b356f883484997f31ca7"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
