)]}'
{"id":"openstack%2Fmistral~688726","triplet_id":"openstack%2Fmistral~master~I714962c91fdf4638dd2e04a8d1618df2245eaeb6","project":"openstack/mistral","branch":"master","topic":"fix-docstring","hashtags":[],"change_id":"I714962c91fdf4638dd2e04a8d1618df2245eaeb6","subject":"Adjust doc string to correct param","status":"MERGED","created":"2019-10-15 14:50:59.000000000","updated":"2019-10-16 11:55:42.000000000","submitted":"2019-10-16 11:53:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688726-1571226830044-d25b5e01","meta_rev_id":"e5de8e5bf43ad2723ceb3f2f01bed11ac9efe874","_number":688726,"virtual_id_number":688726,"owner":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-16 06:24:57.000000000","_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 11:53:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},{"value":2,"date":"2019-10-16 04:05:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},{"value":1,"date":"2019-10-16 04:05:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 04:05:21.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-10-16 11:53:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f996e863db997db33d544f4bf3764b49588cf823","author":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"date":"2019-10-15 14:50:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"187596a399a9bd07a2e527bf77ef4ca270d0210a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 17:48:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef54b01aad4f47fea186042d83e4da1d : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/65e095e9e0ca4550a52ad66cd0ceeea8 : SUCCESS in 10m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9e13d26ff2e415280e98cf8ecf9e380 : SUCCESS in 9m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f26d8d6d254a4119aec940b10eb04c52 : SUCCESS in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb5033e70dc457895fa4baf2aea9391 : SUCCESS in 4m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/274b6e75dd4643da924c81949d0005dd : SUCCESS in 9m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a8c9a08d0a3644b9b47fa13cd87a54b9 : SUCCESS in 36m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ce26a9116a2c42708d3176317bbb4d43 : SUCCESS in 50m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/17b610b4c58e42179a9d9ad21d16b617 : SUCCESS in 47m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/52d5d73987e74c26a516d556abc6c685 : SUCCESS in 29m 00s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2b8a5b25be30496c9d922ba1ac4150cc : SUCCESS in 29m 06s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/3ed0c2ad800c4f8896bc641ea7421062 : SUCCESS in 1h 43m 09s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/67095c1dcaa44ec4aed67e4333b2292f : SUCCESS in 2h 48m 01s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ff95e5a495ac4279afd9705176d8ec45 : SUCCESS in 1h 06m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8070c73eb1ec4d5ea9f625fe0183b5fe : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b5dc9918bafbe6b5bed337094e44d5d1ceedb032","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-16 04:05:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"43344bb8ad202e53165ed1120c322641ec6c4be6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 04:05:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a5a9c38f3ef8eaab7647aa45a770ff5e72459be9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 06:14:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4705619f702f4b0ebc2fbe97c0b75f90 : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/215f7e35169146468a8157784fa430fe : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ed78910d6ab4fb3bd61abd919e4740d : SUCCESS in 11m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8c2dbce3e98407ca12d1ab7aa33d72e : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b067193d51f64f2c8b55dc7240d869a2 : SUCCESS in 5m 48s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8408c3f8ab994983beb5c6ad234dbdfb : SUCCESS in 42m 55s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fcd3cd4bfefe4469bee2926c8868b424 : SUCCESS in 43m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ca6eba8bfa8e40c3b2a11c239d9b8420 : SUCCESS in 27m 15s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c08b6341850a4f1f9ee22ac795a68184 : SUCCESS in 28m 37s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/68b7a0e7ac2745d48ea94d636f8ebfac : SUCCESS in 1h 40m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0f1be14fa0814a018edf61839d983f87 : SUCCESS in 43m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5d83acb0c2a4adfae8dfcd3fbaad500 : SUCCESS in 8m 37s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/5cb97699af904e9e8ff9d9bd9561bafd : FAILURE in 2h 06m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d4a10337d91db13577a9bf09426445c1c5bb7a3a","author":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"date":"2019-10-16 06:24:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0df55d44b035d4e7c6101cba89038fa644e5c148","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 09:46:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7061d87a82ef438c9c3aa32bddda2b9c : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18c55689b6104b8fac56ed48363d98ec : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64350060f23a41809e0870c794944eb9 : SUCCESS in 8m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0af3c25ea46a4fa982e020d90dbccf20 : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/753f10e9681942c3927fd40032d8be97 : SUCCESS in 4m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71553d31cfd04cfd9b72a0e0650d8beb : SUCCESS in 9m 36s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/235d12e4328844afb3f9d398c40e123f : SUCCESS in 44m 29s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/86723dbdd28b42cd9ada45394f36e509 : SUCCESS in 48m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/59013e51e72f465eb5921533950588bb : SUCCESS in 42m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/67007d632e224cb7838910634ae37300 : SUCCESS in 33m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/92ccf1983b5d40bf8ce046d7ed52ec89 : SUCCESS in 28m 42s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/68dc4d16947e4e038ce0aa40499b51ad : SUCCESS in 1h 42m 13s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/090ba5ee0ad64912bfc2413f8281c7aa : SUCCESS in 3h 19m 12s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/497b8db625d9462898c438dafd0752ee : SUCCESS in 1h 00m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f86db7c1d127427fba6cb9bbee47f743 : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":1},{"id":"25ae5c2ce300b90004e2580ba6d3bc1d440407d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 09:46:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c78df136d7d69d13445dc251d9801598c2db00e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 11:53:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0771f8d70011410aaba50735837eca89 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/882a39555596440697ddd7e04f469871 : SUCCESS in 8m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14882677943b4bd3b98e07ab7af5fc11 : SUCCESS in 9m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c537bf8c2bc44a2c976ad7b4e024df79 : SUCCESS in 9m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85a114c20afb4b619ac2606f7975b1a8 : SUCCESS in 5m 15s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3cb664c7ac3f4a39a21feb3e280bf8a3 : SUCCESS in 42m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7c69c3413487417aaad600cef2f0721a : SUCCESS in 42m 37s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/77efad7285fb421587e58fc94847d038 : SUCCESS in 30m 10s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/19903a153f7f4437b472d9f94a492a0d : SUCCESS in 28m 16s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5642e1988acc4cdb801b33d72de160c8 : SUCCESS in 1h 51m 37s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4aaad5363de443369c370bccc8ce45e6 : SUCCESS in 45m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/705466cb440946ed917f25ffe5fea193 : SUCCESS in 11m 30s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/15a5af07335740b692f77b48128678d4 : SUCCESS in 2h 01m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b62d4afd5e3641d59e0a47a95a407176755e6652","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 11:53:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ffb296a11b0c4c8768bb9a084844683a6663b33d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 11:55:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64cfd188ec3f4b89ad54043c4e469b9d : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10fe69e147e4cc5e18a702658287f75d93c53243","revisions":{"10fe69e147e4cc5e18a702658287f75d93c53243":{"kind":"REWORK","_number":1,"created":"2019-10-15 14:50:59.000000000","uploader":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"ref":"refs/changes/26/688726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/26/688726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/26/688726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/26/688726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/26/688726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/26/688726/1"}}},"commit":{"parents":[{"commit":"8f02e8a7d7f3025aae7a99543568fe4eb451c955","subject":"Change the action error message format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8f02e8a7d7f3025aae7a99543568fe4eb451c955"}]}],"author":{"name":"Dominic Schlegel","email":"dominic.schlegel@hostpoint.ch","date":"2019-10-15 14:47:49.000000000","tz":120},"committer":{"name":"Dominic Schlegel","email":"dominic.schlegel@hostpoint.ch","date":"2019-10-15 14:47:49.000000000","tz":120},"subject":"Adjust doc string to correct param","message":"Adjust doc string to correct param\n\nThe doc string for action_ex_id has been used twice and the doc string\nfor state param was missing. This commit corrects this.\n\nChange-Id: I714962c91fdf4638dd2e04a8d1618df2245eaeb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/10fe69e147e4cc5e18a702658287f75d93c53243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/10fe69e147e4cc5e18a702658287f75d93c53243"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
