)]}'
{"id":"openstack%2Fmistral~688729","triplet_id":"openstack%2Fmistral~master~I02401b3e0471a4e29b2292da842115946867cf8e","project":"openstack/mistral","branch":"master","topic":"missing-param","hashtags":[],"change_id":"I02401b3e0471a4e29b2292da842115946867cf8e","subject":"Add missing :param statement in doc string","status":"MERGED","created":"2019-10-15 15:07:08.000000000","updated":"2019-10-16 13:31:19.000000000","submitted":"2019-10-16 13:29:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688729-1571232584570-47151255","meta_rev_id":"63c517154c324a6a1ecfd1c339c0768cf44b44dc","_number":688729,"virtual_id_number":688729,"owner":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-15 21:00:45.000000000","_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 13:29:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},{"value":2,"date":"2019-10-16 04:06:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},{"value":1,"date":"2019-10-16 04:06:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 04:06:17.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-10-16 13:29:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d398784acd4c50fd1d42814794cfce4553582f68","author":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"date":"2019-10-15 15:07:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61d9c84f4653b22fd2e1543adb182f72274e20fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 17:14:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/854f902693c54d2e87cfac6b238440d9 : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/425dfc7394734496969fa7762a4df03b : SUCCESS in 9m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f290e80eebba4ad1ab90e3f844cce53a : SUCCESS in 7m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19fdca1390dc4270b2a438f40669977f : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0328ddbf757b4347a65a78e497dd1be8 : SUCCESS in 4m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ff58ecde70e4c1eaf34387b63323c2c : SUCCESS in 8m 30s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/faaec477a3ea4a10b129e2d527484aba : SUCCESS in 43m 38s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d418178470c14c0e93a8566e4c5fbc1f : SUCCESS in 43m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e5918bc8989a439cbd0c28bdfde9201e : SUCCESS in 43m 18s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9216a7067ac2460a994b3681f0153d54 : SUCCESS in 28m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/39bfb78d67d24eca8dda1dedf4aab708 : SUCCESS in 26m 20s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/d2ef713a48544a9c8109a7b323537e24 : SUCCESS in 1h 55m 53s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f223d2cb61f84bc5928155a281bc1a90 : FAILURE in 1h 52m 23s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/70620f846fb24fc1b37d1461c0313734 : SUCCESS in 1h 04m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0938e4810eb643b1ba7f33cb6ab74623 : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f96e2f5952ce3e0d44caaffac764af1f8cc51c1d","author":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"date":"2019-10-15 21:00:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c5cd739c3c70e2435ddc51bc186f0a16b8299737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 23:51:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9a9eeb2b9c24354ababb517c6dd36ba : SUCCESS in 7m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ccd83bd42d954887afd2449087b43b33 : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2e74df258654b6082e6207afd8320bc : SUCCESS in 7m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/427d334655074e96bae8102c0df4d31a : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f60fdd8d9f7d4f758ae94a6a463d2f52 : SUCCESS in 5m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff9766b71a9b484a966546185ebbb13c : SUCCESS in 8m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/004c5932abc5408ea40be60b0a9b9017 : SUCCESS in 46m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2d2faaff410e4b9991093f4a7ba52299 : SUCCESS in 37m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3996d127a43e426f87ad506c12b6b0d2 : SUCCESS in 47m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/be68fd169c5d4bc0a9eca827fc256c35 : SUCCESS in 29m 19s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ad554e1066944a14bae9837c71005fec : SUCCESS in 28m 59s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/ba45094a9c7544cdafc8062453c522f1 : SUCCESS in 1h 47m 35s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0c531fab9e1649d5948f62172c91d7b2 : SUCCESS in 2h 45m 21s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2b5884d6906f4e168bff837f1ab2b626 : SUCCESS in 1h 05m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b317f0d77c247c685c823cfb8538e83 : SUCCESS in 9m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d5ae90588dfdebcbef8f9663d2fd18fcf9c6b406","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-16 04:06:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b1ab3ca4fee4d6e933b1db65d36006e8df107658","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 04:06:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"04d1c648d3c624cea53dffab39dab9df147dc033","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 08:16:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84b02b80e6f04170bf8b3acebf2db815 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ec21899d4d6549fc9834cfed618e1e83 : SUCCESS in 11m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea73c432b6444976b1bbb23eba8828ff : SUCCESS in 8m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16053b7640054309b9687a1172118207 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7382945b5764d2cbafb79b2ded9ec69 : SUCCESS in 7m 06s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d985f24239934a0489151eac1a4f0355 : SUCCESS in 45m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e88c4e94ba4f486fbda334e1dae590fe : SUCCESS in 43m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2ef3e6f6d9be4d86b831a9c2cecd64d6 : SUCCESS in 29m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/719440d560034899bd99ad4234357a82 : SUCCESS in 29m 06s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/4f781d23d34e4ac091ff0b7203652a8a : POST_FAILURE in 1h 16m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2060061ca8a94e87a1934ed08dd04bbe : SUCCESS in 43m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b80abbf24ce14c8ca642c75c794a56d1 : SUCCESS in 9m 41s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a17a138c54084297aedaf7853470156b : SUCCESS in 1h 55m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b2c72810bc01088d20706c90500f8f5b3905adf5","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-16 08:16:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7702f615a5e1bcc897b12d9faf555a98f5181193","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 11:26:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c065d075eac4179a144b9736c8b4480 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/266d530e0bd643aa93664bf49a5f816b : SUCCESS in 9m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82309f7e964f42ce90067a28216addaa : SUCCESS in 9m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37393a69055343f3879b854263f4cc78 : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0e3859b96794d358d5f40486d9ab811 : SUCCESS in 4m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7263217f62fa4bd59afbff521e31751c : SUCCESS in 8m 57s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/902998dbfd5845028781f41c8d188031 : SUCCESS in 41m 48s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/becc3e2545424df082f903082bbd3ecf : SUCCESS in 44m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/736bb1fe66d745de8a607365240ffa6c : SUCCESS in 42m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/27623801021447ffb0dd1b6f885fd32f : SUCCESS in 28m 30s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/83422c30ddc6461bb6e0c348155c5876 : SUCCESS in 25m 54s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/26fdc9ef431e478d9e53759868fb3b05 : SUCCESS in 1h 56m 03s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/28dfa322801244a7afdf70fe5bb5fc2c : SUCCESS in 3h 02m 17s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a3a121b3daaa4e8392142d6742be4e93 : SUCCESS in 1h 06m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6870032f85d14865957549c9ee25a3c8 : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a08decca2a1d4705b769c5be8152737e4c1c968d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 11:26:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ccf90c3e37abd327aee5ee562c6eea03edd0cee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 13:29:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8fb66326147350fb5ab92522b89f7c4706e3fcd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 13:29:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f2d3c586cfc4b5c8d31cc678b3b5245 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5277e116cd4a4da6a8e9d486e0c2ead3 : SUCCESS in 10m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32d950e4773d427bab03439335abdca5 : SUCCESS in 8m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e81261c67ff44e4bf419c5ca51d2285 : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2db0d2323bd84961a1d15223da136428 : SUCCESS in 4m 38s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1cc78de001104ac8a98f460424cc1288 : SUCCESS in 42m 38s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c6b3acb1faf742d8a9c0c90975d5738a : SUCCESS in 44m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ca2ade28e098432c9e1e75f2f31a2cfb : SUCCESS in 28m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/317f71c811f44a06aa7c26cd0ff61838 : SUCCESS in 26m 47s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/dab48f41c1fb46e0b2f5c179463a83a1 : SUCCESS in 1h 50m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6ef7bf97a4a24b4491cfa7f42ea00ba1 : SUCCESS in 44m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c1a8cb141924b339ec22462fb030eb9 : SUCCESS in 10m 12s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8335503b1e644216925a27a923f6be1e : SUCCESS in 1h 57m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a15038ea118170f2b19220b47cfeff719e63929b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 13:31:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2554d8d94724660b212444de5b26343 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6284c9dcd339cee5ae8a23282527ad83f2df8886","revisions":{"6284c9dcd339cee5ae8a23282527ad83f2df8886":{"kind":"REWORK","_number":1,"created":"2019-10-15 15:07:08.000000000","uploader":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"ref":"refs/changes/29/688729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/29/688729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/29/688729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/29/688729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/29/688729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/29/688729/1"}}},"commit":{"parents":[{"commit":"8f02e8a7d7f3025aae7a99543568fe4eb451c955","subject":"Change the action error message format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8f02e8a7d7f3025aae7a99543568fe4eb451c955"}]}],"author":{"name":"Dominic Schlegel","email":"dominic.schlegel@hostpoint.ch","date":"2019-10-15 15:06:27.000000000","tz":120},"committer":{"name":"Dominic Schlegel","email":"dominic.schlegel@hostpoint.ch","date":"2019-10-15 15:06:27.000000000","tz":120},"subject":"Add missing :param statement in doc string","message":"Add missing :param statement in doc string\n\nChange-Id: I02401b3e0471a4e29b2292da842115946867cf8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6284c9dcd339cee5ae8a23282527ad83f2df8886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6284c9dcd339cee5ae8a23282527ad83f2df8886"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
