)]}'
{"id":"openstack%2Fmistral~688876","triplet_id":"openstack%2Fmistral~master~If52bd09f9178a1d6f2b0d476a43b25daa26a03be","project":"openstack/mistral","branch":"master","topic":"bug/1848037","hashtags":[],"change_id":"If52bd09f9178a1d6f2b0d476a43b25daa26a03be","subject":"Prohibit creation workflows with spaces in name","status":"MERGED","created":"2019-10-16 08:43:38.000000000","updated":"2019-10-27 14:15:06.000000000","submitted":"2019-10-27 14:13:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688876-1572185636912-2bb4bfd8","meta_rev_id":"01ccbc3d9dc54209b8011e833b3615bff8c91f16","_number":688876,"virtual_id_number":688876,"owner":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},{"value":0,"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-27 14:13:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-27 07:51:53.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":2,"date":"2019-10-27 12:08:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},{"value":0,"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":1,"date":"2019-10-27 12:08:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},{"value":0,"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 08:58:30.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-10-22 06:00:46.000000000","updated_by":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"reviewer":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"state":"REVIEWER"},{"updated":"2019-10-27 07:51:53.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-10-27 12:08:09.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-10-27 14:13:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e1ce3ca802ff860895f7f713d5b0daed1935f2b","author":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"date":"2019-10-16 08:43:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"884111e21d5e524f29c85d2db29d960c9526d0da","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-16 08:58:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(9 comments)\n\nThanks for contributing!\n\nPlease address my comments.","accounts_in_message":[],"_revision_number":1},{"id":"5ba6f964adce9de00cd5fe28923462c31fe9d820","author":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"date":"2019-10-16 09:33:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fd8bee482e11131321ccea1750349e666cb010ac","author":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"date":"2019-10-16 09:36:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ec434a2705a022530554a47906079bb22bdb000c","author":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"date":"2019-10-16 09:37:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fdfa79163197b267f88566547294fc48693f2524","author":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"date":"2019-10-16 09:38:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dea783468a1f98ae7aee2d969fb8df6e4f935e5d","author":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"date":"2019-10-16 09:39:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0ed2a55bb35eccaa3c0dee383aeb0c5309cf570c","author":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"date":"2019-10-16 09:43:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9e59df067f595526bcb17aef5de50d93254da8ad","author":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"date":"2019-10-16 09:48:46.000000000","message":"Patch Set 4:\n\n\u003e (1 comment)\n\nBut WorkflowDefinitionTest and api tests already exist for it","accounts_in_message":[],"_revision_number":4},{"id":"82263dcf484dbdd10a8d323d77537f7971d9cc1c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-16 12:13:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bdd0da410a3b279ebbc8e49d2defd07781624556","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 12:53:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4064b74154cb42e4aa6eb4cf16d5d83a : FAILURE in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b01f6f9dd344ba9980478df118c1c00 : FAILURE in 9m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be0183752f3f430aa5d25dcf783bdd89 : FAILURE in 8m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c45bf62806d44a4ae9421d866ceed5f : FAILURE in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6f844a5ded44450aad2b0e275c0b1e3 : SUCCESS in 4m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc3f8193f4dc4f7e86a663aff0f3b1d3 : FAILURE in 8m 53s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1258dea2517c4f6c9797a9ef6ef4e9ca : SUCCESS in 40m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3efb2cb3374e436984126145c8383b02 : SUCCESS in 45m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4c5ac90ee7984c348fa41dd5db2fcb54 : SUCCESS in 41m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c3db2c61fd474c108bff4c24d41a9e2e : FAILURE in 27m 29s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/152c96f9dffc4544a3c4d13cdfcab35a : FAILURE in 28m 32s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5e5ee78a7a8c40a7ab8d4cf17916ec2d : SUCCESS in 2h 05m 54s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/20e1b275f05b45b6bb97508568afce28 : SUCCESS in 3h 01m 56s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2cf8ebee2734447faa509811eff15826 : SUCCESS in 1h 02m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/058ed8d2e40c4860b20b44b3a949bffe : FAILURE in 11m 19s","accounts_in_message":[],"_revision_number":4},{"id":"bb99713d77083a471f2c2e45e2dfb139f98ccd31","author":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"date":"2019-10-17 12:18:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"89e4c831a2ef5be72003e51042ed34749afb39a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 15:17:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eaf3c38ea9d4032b267e5cb3fd5ab2b : FAILURE in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dcae006713444529a0bfa090600cf754 : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c0a23173be540bd8a7c5cb6262889b5 : SUCCESS in 10m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b6a72f67c444a778c694af86780f375 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/243af5a0bd6e43b8927dbed4bcbda860 : SUCCESS in 5m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db3adfe1cf654b5fbaedb8426eb4596f : SUCCESS in 10m 13s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/760216b97d47401796b78aaa3e5f0ddb : SUCCESS in 44m 46s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1b30010b02344e69b02d47693739fce0 : FAILURE in 46m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7320459da869429b828841ee3faf861f : SUCCESS in 45m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/99c508a53ada48f0994e829d5b711ba5 : FAILURE in 29m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/de912e08b2e24e07a567703de906d83d : FAILURE in 28m 45s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5bf798a4ffa4453b8cb901e208e673b5 : SUCCESS in 1h 30m 24s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6235d148e1a94743a0b561752514f41b : SUCCESS in 2h 57m 15s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/785c7fe2f1f941e697ab7f5f37b891ed : SUCCESS in 1h 02m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f2f10ebd26b4147a98763f9b764d72f : SUCCESS in 9m 28s","accounts_in_message":[],"_revision_number":5},{"id":"1d26e57953eb568af8fda90633deb088c31ad68b","author":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"date":"2019-10-18 09:04:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1251a25183b1135c61092179f5f9e34dfd8579e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 12:06:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f403ff815751442fb7da601ce165a2df : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/06265d4189bf4510b01bcc54cc572ce3 : FAILURE in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f0c472a05e24b26b4219a743db2ec39 : SUCCESS in 8m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d955a2ac4b9e4a9cbe3d86d4380a247f : SUCCESS in 9m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0207f52f42d2466cb9d55ad582c5062a : SUCCESS in 5m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5513277fe05d43369f95597acd06f579 : FAILURE in 8m 26s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/23f3686574664a17bce5ae06a53563ed : SUCCESS in 50m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bc67219d105e4d2f839b43204033b28b : SUCCESS in 44m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1ef50cb099bf4effa6b58940bd6a68df : SUCCESS in 39m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5c75d707e4f84048a2c590176e1d960a : FAILURE in 31m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/370fa4e7fa8c4bb5861a625c00763cdc : SUCCESS in 27m 25s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/92593a753c6f42f78f08ae1f01dcf211 : SUCCESS in 2h 16m 32s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/de28481c8b8d4fccb4c4b059e29a5c92 : SUCCESS in 2h 56m 32s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/01b7e4bd53b3482ea17b95bf01b2f599 : SUCCESS in 58m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6eb3bb9437674ff69affcd6febcdeb9f : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":6},{"id":"4648ae6302488b5f4fd26781c5ea94fd21896a31","author":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"date":"2019-10-18 12:12:46.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"811c36859fed90c54df8bf7d0d6ee7a50d323a1f","author":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"date":"2019-10-22 06:00:46.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"5e42b76f7101cdba132c3ebb2b7a49e7164939b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 09:01:04.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3786f1bbdff5430cab70037c7e945fdb : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bff06a805edb41df857e268cf1345a5f : SUCCESS in 12m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2f5f7aeb2ce4619b88e5c8bb27a6934 : SUCCESS in 12m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ed06e81e65f4bf782fbb4915362eec6 : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/948c8c3ea8b54835b49f51c2104787ce : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/347c523cfe7d4520900b9f9258a9845b : SUCCESS in 10m 02s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8b64e72b2ec0448aaa80957d7060fc97 : SUCCESS in 49m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c321f9dc3f144a2385ab346a2bc148f2 : SUCCESS in 51m 17s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dbbf3e2c139e4ce5a3ac2cd02ceae42a : SUCCESS in 51m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8d6ca4306fcb44dbb0fe537cc82c0c83 : SUCCESS in 28m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/22feba9fdd3e40e7a5ff7e3b1f79816a : FAILURE in 4m 17s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/84a0e1b4bb264e7489edd25a4d9fb405 : SUCCESS in 1h 59m 40s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a452db789bb24e8caea3bb5a0bba3833 : SUCCESS in 2h 35m 25s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/98a7996e54054162aef869beaede871d : SUCCESS in 1h 20m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1817bdea7b1422aa969d068883d9903 : SUCCESS in 10m 28s","accounts_in_message":[],"_revision_number":6},{"id":"bc3a721b026816e9bbfcf7af44c7595ebb9c518c","author":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"date":"2019-10-22 11:25:52.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"4a2a4365e83bd1afef2756cd931765fe3124a471","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 14:43:05.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e7723431e784e47a77f66f2c5e9defd : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52264c3d98fb46f3b54b01c06a182a88 : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4d4b2fb3f3a456ba3cbf839434194f9 : SUCCESS in 9m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19aff0ed3a3740e187f77ab736c1886a : SUCCESS in 13m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c2bac56744a4061b54c58133e57ff3d : SUCCESS in 4m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17b77cbf08714031b895bb79ec5a1fab : SUCCESS in 9m 31s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ac2dbeb9f4684cc8993b03e2aa1ad293 : SUCCESS in 47m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e87d3162f52142bcb84fc008e79da5e0 : SUCCESS in 44m 27s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/40e57a6e66824c97827998eb5559609a : SUCCESS in 48m 06s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/53a1bd75621045ca9487d5b20ad39e37 : SUCCESS in 31m 00s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9b5aaba3e2fe46b68636c7671515fdd4 : FAILURE in 4m 58s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/ce3aecf844674d2ab9af3fbb09bad10f : SUCCESS in 1h 45m 50s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/589747bcea37419bbab9222e9ca131a0 : SUCCESS in 3h 07m 51s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/3cd0d31f56a94450ba0f3479562cd0e9 : SUCCESS in 1h 02m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21aece47cbff4714b68609a1bff2f2fe : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":6},{"id":"63f493e844f9e49eee09ee6c4203505a1b3dcb52","author":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"date":"2019-10-24 05:32:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"bcde67ab5f11a08e457ee29b89b75e087a481774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 08:25:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9e295e0ec974a2da85f8537a6d482f6 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5b740c2e00c148c4963a101f3b843a9e : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edbaaafb4fee4301b3093be383182d89 : SUCCESS in 7m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0e4fc283a56417d9942fa4ebf6c2df6 : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b64d1511bbe546daafeca985d32e574f : SUCCESS in 4m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efcfe526a34f40d18c79353edf6782eb : SUCCESS in 8m 54s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0526bb3bb4e24adf88674d7c7830c2e0 : SUCCESS in 47m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/62c0adba37004d0e9844a6cd00d80ada : SUCCESS in 42m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6c39c233ccf245cc93428b08d9cc3f39 : SUCCESS in 40m 48s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/084ba68d7fee4e2a982bc2ce0547661b : SUCCESS in 28m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7e6c10a14c59413c88805e5743f32b37 : FAILURE in 3m 55s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a72b1ff170d34e3d96d80d1d1d4d3422 : SUCCESS in 1h 32m 18s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a8a5ea997c5a4a0fb2f3f1cc4219a7f0 : SUCCESS in 2h 48m 47s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b5a353ece17f45c5a0cb74898763824d : SUCCESS in 1h 05m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15ef4d49244242bda8a58eaa12e5e6c9 : SUCCESS in 9m 58s","accounts_in_message":[],"_revision_number":7},{"id":"9dccfc5dc1f960872f817c7ed7bfd47a195a1e28","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-24 11:27:57.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"6ab5e7fc7957272463cf4d813b36e10322281488","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 14:55:44.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b7c93f5ad4b42f8890f71a3524bf2f2 : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c89106820004f888b0ad00c82702562 : SUCCESS in 10m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5be1cbe2646e4bde88dc973fd6d0662e : SUCCESS in 8m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d1b3890b56a4bd4a5781ee1be8fff0d : SUCCESS in 9m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa07ec174f204cc09435f7d4b0a72837 : SUCCESS in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f3f31abb8874637b28c84ed8658388a : SUCCESS in 10m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/720536dcd93143bb89cabdb7a8c04fc4 : SUCCESS in 57m 48s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f2bffef3bb4f4b3e9e82ed2832f94c47 : SUCCESS in 37m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/11cc9d49c99247d58b3cd4d1e0cb8517 : SUCCESS in 42m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/65d127a2cdd94b7d8c2f6502fe763766 : SUCCESS in 31m 23s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3843de9a874241a6807d82c264b0a24e : SUCCESS in 29m 59s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/806598b830274f3c830b9cc5a58fa363 : SUCCESS in 1h 31m 10s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8d1f47c705b34c7e8cfeaa0402c2ba35 : TIMED_OUT in 3h 18m 42s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8305744f1e164033920f987291ca7a01 : SUCCESS in 1h 20m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c06f4f85f1924fe78dfd303fc458424e : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":7},{"id":"771409b7028149b3ff4df9b5347a4cbeb78643bd","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-27 07:51:53.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"aef6fae538ef78ebf622c4087f4399fa7b73d85d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 10:52:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39568c3e063749788a9daf1fca2ffc2a : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/361ec357df754d66a14e94b107a4166e : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b7f964dbec3433fa1223194c570e7ed : SUCCESS in 9m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e408546a4d444529abb44b82f963475 : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eb6156f68c644ceb635966030afacc5 : SUCCESS in 4m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73aad1a1bdf149ab812902ca0854b949 : SUCCESS in 8m 02s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6ac8b8b4a86d4630a186198371d90769 : SUCCESS in 39m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bd60561f654b4aedaf87d2dc47225979 : SUCCESS in 50m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/74d5c5648f35448a9a15dcedb82d9907 : SUCCESS in 42m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a25b8fb0a9ed42f39120dec99d1dd052 : SUCCESS in 27m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a6edd7f3c2034de4917c1b9abd2eaef3 : SUCCESS in 26m 04s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/99f37f2c106f45119f6a9e5b28e57701 : SUCCESS in 1h 26m 05s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4dee4b3a11d142a98d8ff72509d46d16 : SUCCESS in 2h 55m 24s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/168202ca977e4d39a285c8550de0ea3d : SUCCESS in 1h 06m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/662dd764178748df814343015f763320 : SUCCESS in 10m 31s","accounts_in_message":[],"_revision_number":7},{"id":"8ddcffeee2f4d3a017b6e395b9d474e6e8dfbd46","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-10-27 12:08:09.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"2c665ebda125447146f4bbe266bb49397ecd35c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 12:08:21.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"38ea45847df5937918508a0088884497c51f9ebb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 14:13:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"9d5187373b2cc81d35268cb1b1030c7d82b5b4e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 14:13:56.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c6a1d5b7c2a43cb9369ddc33f0200cd : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc3159f884f4471685e6c3f37c7793c0 : SUCCESS in 8m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74153bdc6a8146f58ebf843eaff00a41 : SUCCESS in 8m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72fb0541f3154d07aa550d4a5dac5fab : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc470252d2545e88de954df0fb285b1 : SUCCESS in 5m 44s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a06b0c9e1ca449b0a409cbf71173b578 : SUCCESS in 40m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8fd6d383f84c4b4ca920d45fdc9b3f46 : SUCCESS in 42m 20s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e4a0a0f688cc4ca291bdab405ced93d9 : SUCCESS in 30m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b41d55cf638a4c8a925e83923aaf411e : SUCCESS in 28m 38s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b80b030b04a548399952fe19eaa85b58 : SUCCESS in 1h 37m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9146781b59d84a9ea791a42deaea1c04 : SUCCESS in 40m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea5f2325604543219fdf0175884f3ab9 : SUCCESS in 9m 45s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/fdd0fdabc7d04029bb8d24b618a7cec4 : SUCCESS in 1h 52m 07s","accounts_in_message":[],"_revision_number":7},{"id":"258cdc1cfc2db8554ecb18bf09fcd3fc3b3a29f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 14:15:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b69e26f8e054d578307850f4ad67ef8 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f01d17726e145144d504173484cab4e5aa19abf3","revisions":{"fdb4394837f10b2ff16244001b38c0fae9d1f297":{"kind":"REWORK","_number":1,"created":"2019-10-16 08:43:38.000000000","uploader":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"ref":"refs/changes/76/688876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/688876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/688876/1"}}},"commit":{"parents":[{"commit":"8f02e8a7d7f3025aae7a99543568fe4eb451c955","subject":"Change the action error message format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8f02e8a7d7f3025aae7a99543568fe4eb451c955"}]}],"author":{"name":"Artem Lapin","email":"arteom.lap@yandex.ru","date":"2019-10-16 08:40:38.000000000","tz":180},"committer":{"name":"Artem Lapin","email":"arteom.lap@yandex.ru","date":"2019-10-16 08:40:38.000000000","tz":180},"subject":"Added additional check for workflow name to prevent creation workflows with spaces in name","message":"Added additional check for workflow name to\nprevent creation workflows with spaces in\nname\n\nChange-Id: If52bd09f9178a1d6f2b0d476a43b25daa26a03be\nCloses-Bug: 1848037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fdb4394837f10b2ff16244001b38c0fae9d1f297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fdb4394837f10b2ff16244001b38c0fae9d1f297"}]},"branch":"refs/heads/master"},"85b81157d908ffda10da6ef9dfe20b4142fd1534":{"kind":"REWORK","_number":2,"created":"2019-10-16 09:33:57.000000000","uploader":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"ref":"refs/changes/76/688876/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/688876/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/688876/2"}}},"commit":{"parents":[{"commit":"8f02e8a7d7f3025aae7a99543568fe4eb451c955","subject":"Change the action error message format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8f02e8a7d7f3025aae7a99543568fe4eb451c955"}]}],"author":{"name":"Artem Lapin","email":"arteom.lap@yandex.ru","date":"2019-10-16 08:40:38.000000000","tz":180},"committer":{"name":"Artem Lapin","email":"arteom.lap@yandex.ru","date":"2019-10-16 09:29:56.000000000","tz":180},"subject":"Prohibit creation workflows with spaces in name","message":"Prohibit creation workflows with spaces in name\n\nAdded additional check for workflow name to\nprevent creation workflows with spaces in\nname\n\nChange-Id: If52bd09f9178a1d6f2b0d476a43b25daa26a03be\nCloses-Bug: 1848037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/85b81157d908ffda10da6ef9dfe20b4142fd1534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/85b81157d908ffda10da6ef9dfe20b4142fd1534"}]},"branch":"refs/heads/master"},"c66163873ad83b91b686e510a0729e9b1a8a2393":{"kind":"REWORK","_number":3,"created":"2019-10-16 09:36:08.000000000","uploader":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"ref":"refs/changes/76/688876/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/688876/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/688876/3"}}},"commit":{"parents":[{"commit":"8f02e8a7d7f3025aae7a99543568fe4eb451c955","subject":"Change the action error message format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8f02e8a7d7f3025aae7a99543568fe4eb451c955"}]}],"author":{"name":"Artem Lapin","email":"arteom.lap@yandex.ru","date":"2019-10-16 08:40:38.000000000","tz":180},"committer":{"name":"Artem Lapin","email":"arteom.lap@yandex.ru","date":"2019-10-16 09:35:25.000000000","tz":180},"subject":"Prohibit creation workflows with spaces in name","message":"Prohibit creation workflows with spaces in name\n\nAdded additional check for workflow name to\nprevent creation workflows with spaces in\nname\n\nChange-Id: If52bd09f9178a1d6f2b0d476a43b25daa26a03be\nCloses-Bug: 1848037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c66163873ad83b91b686e510a0729e9b1a8a2393"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c66163873ad83b91b686e510a0729e9b1a8a2393"}]},"branch":"refs/heads/master"},"80fdeff03ed9843a87b0415d7663e046bed41077":{"kind":"REWORK","_number":4,"created":"2019-10-16 09:39:02.000000000","uploader":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"ref":"refs/changes/76/688876/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/688876/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/688876/4"}}},"commit":{"parents":[{"commit":"8f02e8a7d7f3025aae7a99543568fe4eb451c955","subject":"Change the action error message format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8f02e8a7d7f3025aae7a99543568fe4eb451c955"}]}],"author":{"name":"Artem Lapin","email":"arteom.lap@yandex.ru","date":"2019-10-16 08:40:38.000000000","tz":180},"committer":{"name":"Artem Lapin","email":"arteom.lap@yandex.ru","date":"2019-10-16 09:38:25.000000000","tz":180},"subject":"Prohibit creation workflows with spaces in name","message":"Prohibit creation workflows with spaces in name\n\nAdded additional check for workflow name to\nprevent creation workflows with spaces in\nname\n\nChange-Id: If52bd09f9178a1d6f2b0d476a43b25daa26a03be\nCloses-Bug: 1848037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/80fdeff03ed9843a87b0415d7663e046bed41077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/80fdeff03ed9843a87b0415d7663e046bed41077"}]},"branch":"refs/heads/master"},"c4149c725eb2399bcc5ff3127068ffdd75506b61":{"kind":"REWORK","_number":5,"created":"2019-10-17 12:18:36.000000000","uploader":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"ref":"refs/changes/76/688876/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/688876/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/688876/5"}}},"commit":{"parents":[{"commit":"8f02e8a7d7f3025aae7a99543568fe4eb451c955","subject":"Change the action error message format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8f02e8a7d7f3025aae7a99543568fe4eb451c955"}]}],"author":{"name":"Artem Lapin","email":"arteom.lap@yandex.ru","date":"2019-10-16 08:40:38.000000000","tz":180},"committer":{"name":"Artem Lapin","email":"arteom.lap@yandex.ru","date":"2019-10-17 12:18:04.000000000","tz":180},"subject":"Prohibit creation workflows with spaces in name","message":"Prohibit creation workflows with spaces in name\n\nAdded additional check for workflow name to\nprevent creation workflows with spaces in\nname\n\nChange-Id: If52bd09f9178a1d6f2b0d476a43b25daa26a03be\nCloses-Bug: 1848037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4149c725eb2399bcc5ff3127068ffdd75506b61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4149c725eb2399bcc5ff3127068ffdd75506b61"}]},"branch":"refs/heads/master"},"56dbf8eb9234a14f8323e6638f67cc2a2b9244a3":{"kind":"REWORK","_number":6,"created":"2019-10-18 09:04:19.000000000","uploader":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"ref":"refs/changes/76/688876/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/688876/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/688876/6"}}},"commit":{"parents":[{"commit":"8f02e8a7d7f3025aae7a99543568fe4eb451c955","subject":"Change the action error message format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8f02e8a7d7f3025aae7a99543568fe4eb451c955"}]}],"author":{"name":"Artem Lapin","email":"arteom.lap@yandex.ru","date":"2019-10-16 08:40:38.000000000","tz":180},"committer":{"name":"Artem Lapin","email":"arteom.lap@yandex.ru","date":"2019-10-18 09:03:38.000000000","tz":180},"subject":"Prohibit creation workflows with spaces in name","message":"Prohibit creation workflows with spaces in name\n\nAdded additional check for workflow name to\nprevent creation workflows with spaces in\nname\n\nChange-Id: If52bd09f9178a1d6f2b0d476a43b25daa26a03be\nCloses-Bug: 1848037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/56dbf8eb9234a14f8323e6638f67cc2a2b9244a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/56dbf8eb9234a14f8323e6638f67cc2a2b9244a3"}]},"branch":"refs/heads/master"},"f01d17726e145144d504173484cab4e5aa19abf3":{"kind":"REWORK","_number":7,"created":"2019-10-24 05:32:20.000000000","uploader":{"_account_id":30775,"name":"Artem Lapin","email":"arteom.lap@yandex.ru","username":"Rimger"},"ref":"refs/changes/76/688876/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/688876/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/688876/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/688876/7"}}},"commit":{"parents":[{"commit":"8f02e8a7d7f3025aae7a99543568fe4eb451c955","subject":"Change the action error message format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8f02e8a7d7f3025aae7a99543568fe4eb451c955"}]}],"author":{"name":"Artem Lapin","email":"arteom.lap@yandex.ru","date":"2019-10-16 08:40:38.000000000","tz":180},"committer":{"name":"Artem Lapin","email":"arteom.lap@yandex.ru","date":"2019-10-24 05:31:44.000000000","tz":180},"subject":"Prohibit creation workflows with spaces in name","message":"Prohibit creation workflows with spaces in name\n\nAdded additional check for workflow name to\nprevent creation workflows with spaces in\nname\n\nChange-Id: If52bd09f9178a1d6f2b0d476a43b25daa26a03be\nCloses-Bug: 1848037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f01d17726e145144d504173484cab4e5aa19abf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f01d17726e145144d504173484cab4e5aa19abf3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
