)]}'
{"id":"openstack%2Fmistral~688965","triplet_id":"openstack%2Fmistral~master~I2571246947ab5ae06731fc7d35de7a5954325a82","project":"openstack/mistral","branch":"master","topic":"releasenote-bug-1843175","hashtags":[],"change_id":"I2571246947ab5ae06731fc7d35de7a5954325a82","subject":"Add release note for fix error of cron trigger run","status":"MERGED","created":"2019-10-16 15:51:58.000000000","updated":"2019-11-21 01:07:39.000000000","submitted":"2019-11-05 07:25:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688965-1572938720492-aeac7fa8","meta_rev_id":"31807fe949000bee4e52d7235dcbd994254f96a3","_number":688965,"virtual_id_number":688965,"owner":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-05 07:25:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"all":[{"value":1,"date":"2019-10-30 09:38:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":1,"date":"2019-11-04 09:52:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2019-10-25 17:24:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":2,"date":"2019-10-17 04:57:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-30 09:35:20.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":1,"date":"2019-11-05 07:16:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 17:24:35.000000000","updated_by":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"reviewer":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"state":"REVIEWER"},{"updated":"2019-10-30 09:35:20.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-10-30 09:38:04.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2019-11-04 09:52:33.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2019-11-05 07:16:08.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-05 07:25:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb7ca38fce5db39273653c35294c96ccdccc5b29","author":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"date":"2019-10-16 15:51:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1834d5113c2b6b0fb7860bcb85fb8c1f9b26b5c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:01:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a95ac465eecd4dcbbafc502cbe217377 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a3f01e5647f4c50a0631669fc54f428 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5ccdfa9286e42108c304e3339b8865c : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f72f5280e5f8731e2fde80d9ffef09bce91baede","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-17 04:57:42.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"6866143edc148317c91d774249202ce1d4190754","author":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"date":"2019-10-25 17:24:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"036b82c8256fd61fbe002f41b55f39fe69d1e58e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-30 09:35:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"edcc285f971e8125c2f15b01512faf10262db29f","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-10-30 09:38:04.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e Uploaded patch set 1.\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"40639ecc9d84176aebc7a2f148157a1f7e0d3ab5","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-04 09:52:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0680bb8948f07923a7e5c777699de62b9ea15033","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-05 07:16:08.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"287d1baa8473daddbcfc6a42d718c4799dc53caf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 07:16:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4dd7ebe3647d6d4b318d1fff9d9b0d49a8fccb57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 07:25:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"194fbc0f0be982db04dd703b1bc919a3fa0492ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 07:25:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a28891f1b0484ab4b121707d2575aa8e : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61290a6607744ef4b4ce43b6c56d18ec : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f61718805a1b4676bdc735351f963276 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"4ba7c18af54d0322a5c58a9afd6f3b0c0fff0e06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 07:26:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f818ca5ff7d47e384000511d0e34fa5 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ecff99a319f4c9481b86d22a0f7de88 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1},{"id":"048182dbd8902b27d9bfa3b93427c9c2835ef94e","author":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"date":"2019-11-05 14:37:51.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit eab767f3ed1a6dfd4f50b214556142346c830c35","accounts_in_message":[],"_revision_number":1},{"id":"ff8490b651931c2a4270751564636e87ba58d218","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-11-21 01:07:39.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 66da715c84fadba2120a59f02b68ba92ae6807b8","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"948ceb150b66b86ba56b11821e1bb9cb8a7209cc","revisions":{"948ceb150b66b86ba56b11821e1bb9cb8a7209cc":{"kind":"REWORK","_number":1,"created":"2019-10-16 15:51:58.000000000","uploader":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"ref":"refs/changes/65/688965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/65/688965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/65/688965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/65/688965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/65/688965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/65/688965/1"}}},"commit":{"parents":[{"commit":"dfcd058144b46ce39a5f53036b31d1a77d85aad9","subject":"Merge \"Add missing :param statement in doc string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dfcd058144b46ce39a5f53036b31d1a77d85aad9"}]}],"author":{"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","date":"2019-10-16 15:32:28.000000000","tz":-240},"committer":{"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","date":"2019-10-16 15:32:28.000000000","tz":-240},"subject":"Add release note for fix error of cron trigger run","message":"Add release note for fix error of cron trigger run\n\nIt relates to the change I793fbfec03032d9ff7137c11becb6d1c18ec54bc,\ncommit 51b7dd0c6d.\n\nChange-Id: I2571246947ab5ae06731fc7d35de7a5954325a82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/948ceb150b66b86ba56b11821e1bb9cb8a7209cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/948ceb150b66b86ba56b11821e1bb9cb8a7209cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
