)]}'
{"id":"openstack%2Fmistral~689680","triplet_id":"openstack%2Fmistral~master~I194d17b64a953ed0497da02b9ec079a536965668","project":"openstack/mistral","branch":"master","topic":"fix-docstring2","hashtags":[],"change_id":"I194d17b64a953ed0497da02b9ec079a536965668","subject":"adjust doc string to correct key","status":"MERGED","created":"2019-10-21 07:14:54.000000000","updated":"2019-11-06 06:31:20.000000000","submitted":"2019-11-06 06:29:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689680-1573021741340-752aeac9","meta_rev_id":"b1c21c9d6b6e31d68eb2660a3061eb5f6fbbe5b5","_number":689680,"virtual_id_number":689680,"owner":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2019-10-24 10:45:16.000000000","_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-06 06:29:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":1,"date":"2019-11-04 09:55:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},{"value":2,"date":"2019-11-06 04:42:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-30 09:31:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},{"value":1,"date":"2019-11-06 04:42:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-30 09:31:28.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-11-04 09:55:18.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2019-11-06 04:42:13.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-06 06:29:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a665bd22a2fec94123b903c1f4470f0c67c3ba84","author":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"date":"2019-10-21 07:14:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69ae1ee4911d19cabedfd52fc1de1cea5c23c37f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 10:04:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4247296bb5d549af8d11201d1a31700f : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afd5e39e215847aab4eccb247416338b : SUCCESS in 8m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3b96daae15f49e8b4ccebfbf32f7b79 : SUCCESS in 8m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56d77afd2e1649f98a01b01229add2fc : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68b03c6946304267904c2fff751a50ee : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1b2a61ad08943a88d845b53d74485e4 : SUCCESS in 8m 22s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b193572d46184450ab5839b3aa899e63 : SUCCESS in 42m 09s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7c544c306fd54016900bbee0ed2bcdb6 : SUCCESS in 42m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/696cbfd8fc034ccabf13214cc612976e : SUCCESS in 39m 37s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b3520ea7b0d342c1a9bf4b76ec333211 : SUCCESS in 28m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/42e4133e76a24574a277e25eebe8c8de : FAILURE in 4m 31s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/28dfedcdbb3240bcb7c1baeb24228588 : SUCCESS in 1h 59m 45s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c35326055e8c4294961c965634602a45 : SUCCESS in 2h 45m 07s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d111bd898a0f4d6d89bcce631ad04acf : SUCCESS in 1h 06m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be4dc6fb8d4447e1bb2406533b07daea : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":1},{"id":"dd48a63bfc4f8774b25f60fb8cca50423de21a04","author":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"date":"2019-10-21 15:10:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6be0459019f095a77fbecc53ba8e4ac0c61ff2ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 18:09:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9976cffc9b3b471580e45f6e1113e00f : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/399e6d9df11e4c01a6b61548e9dbd45d : SUCCESS in 9m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b1f26603a204829bce0417a90c0cb67 : SUCCESS in 9m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f53207fb4cc4ca4b8b0f952935b5448 : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b80e9cdc2db0461d8b59e92104f82d64 : SUCCESS in 7m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cb8d2228e7245188e55a38529f389c0 : SUCCESS in 10m 24s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/dff9e2e09fae427eaf88553360f21c58 : SUCCESS in 47m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8ea728421bad4e3baf88a835601deec1 : SUCCESS in 39m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/96c1211ba24d4e82b0c7987fda9de2e1 : SUCCESS in 49m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ae9babb6de5141e4806c3d2c6a51d41b : SUCCESS in 30m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2f57e21a128f49be874afbaa12e4b382 : FAILURE in 4m 56s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e65f20b5db7d4f02b168ba51f828c323 : SUCCESS in 2h 08m 05s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2b7790165a5f4934a9c84e813587d2d9 : SUCCESS in 2h 51m 37s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f777aec7b1b24354affbfa22fe6cd154 : SUCCESS in 58m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3e7640e1fe74924a71624229f3df934 : SUCCESS in 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1d7cb5ef9a479d77049b9b9729a61cc8441134a1","author":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"date":"2019-10-23 01:01:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c2913795c5048d57dac1ce8165773ba31ed4079f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 03:23:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e76443029dd4b9197bd6569f4ae71c3 : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38871f2c13074c56a1878a539a5c8d74 : SUCCESS in 9m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/508ea079500f44c2baebc9c160d0a65c : SUCCESS in 10m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab8535c7c258468ea5b38c27381e484e : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89eae6765ac54f2f90678aa796417048 : SUCCESS in 5m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3adaa4ca951491eaf13034d93f2c1d0 : SUCCESS in 10m 14s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6eb8ff91dbfe4ca7a0909360f1e72a03 : SUCCESS in 49m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/043c06baeeb04cec8efd44da9f86c833 : SUCCESS in 43m 15s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/735772c4faac41adb0398e32ac9939e9 : SUCCESS in 47m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1647c566cfd84e2ca0377361cb17ade6 : SUCCESS in 30m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9cd8489ac87a4955859f45f57d7df91b : FAILURE in 5m 32s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/06066006449a42ba99765ad3faedfd60 : SUCCESS in 1h 24m 07s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/bd66874328ff4df6867e510a2ee4f902 : SUCCESS in 2h 20m 57s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/93dd54164a65467090b0ab3cb315ba5d : SUCCESS in 56m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7215c192e5fb4c05ade1f6471895d1d7 : SUCCESS in 9m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c1af582bcdd8e8f41458a41ca02df32534ce699a","author":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"date":"2019-10-24 10:45:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f554d73dfe66c77223ef67a70e11abae6001c7b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 13:44:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36b94c5e58d404d8349fe6f1850d840 : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5caf267a8c040d49e7c850981f81615 : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/655e2e0ed1154e3384ff5494ffab9774 : SUCCESS in 8m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e42ca53a1304938bfbd33f918040f8c : SUCCESS in 9m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1666a2cb44c644faae96b35d841df54f : SUCCESS in 4m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0318350f593485c8564840e942d7e3a : SUCCESS in 9m 29s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1b2b58d998054794ac55572981793733 : SUCCESS in 55m 54s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/09ea9bc7c3ae49c99c246aa7a395fead : SUCCESS in 59m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/51e7f3fab4804148873c67aeddee510b : SUCCESS in 41m 05s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fc72682338a34fafa1940feabf781f05 : SUCCESS in 28m 30s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a7bc78a51bb94b14a1b48609804a5d92 : SUCCESS in 27m 22s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/9b1a0879e7384e40a0c16ec554a637b0 : SUCCESS in 1h 33m 29s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/39fa28ba3df5488caf0043f1fdbc305b : SUCCESS in 2h 48m 55s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4ee3ed0e955545809e5073095a315819 : SUCCESS in 1h 10m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b316f0f7256f4785bf8a61917151c61a : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":1},{"id":"88fe1bd6a5281af39acb1d7584b593553f65de6e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-30 09:31:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6353fadcc919559d10bd2dbe7235437b5c13ccd4","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-04 09:55:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9f90117bea74ebd05a182ef8138feda296c89831","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-06 04:42:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e8a2522cd715050ea20c654126af2a920cde01ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 04:42:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5ba1e0f66b6d8d79041a20789eb9ba8c18c0d59e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 06:29:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3bbbd5fa4dbc77fbe40f0e8dc6bc9085a84929e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 06:29:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/082fef1223a641f9b01e12009d752a6d : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/48c550fdabb44fdda344f86a5c1e1a04 : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/466376b323a84b288bc687a8732829dd : SUCCESS in 7m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/228c110a7e364071a1223fb870aba9b7 : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/129df42e3d3c4b26b575ae007f3070ba : SUCCESS in 4m 22s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7cf2b9103e9947729d9726a33145f85b : SUCCESS in 45m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cc9f4b14920d43b78f0bed36a18049a6 : SUCCESS in 41m 13s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4be1db15fcda4b10823c5b288363eb0b : SUCCESS in 29m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0fc4e9a1e6354611829947815c67c527 : SUCCESS in 26m 02s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/0aa2432935c549059ca4a28cf6102d20 : SUCCESS in 1h 29m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7968becc0119479fb3ae6a12f0c9ba31 : SUCCESS in 41m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdabb302990d4556a95d93d0ed733777 : SUCCESS in 9m 41s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/128ba1b9f867410ea67a02008d2066de : SUCCESS in 1h 42m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f16ba5dcac56260a914e86c19158695e0b63b071","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 06:31:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4efafd2c141741d1919966b2fd27af99 : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6ec559cac46ebe1e4edd2166328794d201b802b","revisions":{"f6ec559cac46ebe1e4edd2166328794d201b802b":{"kind":"REWORK","_number":1,"created":"2019-10-21 07:14:54.000000000","uploader":{"_account_id":26693,"name":"do3meli","email":"info@dominicschlegel.ch","username":"do3meli"},"ref":"refs/changes/80/689680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/80/689680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/689680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/689680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/689680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/80/689680/1"}}},"commit":{"parents":[{"commit":"dfcd058144b46ce39a5f53036b31d1a77d85aad9","subject":"Merge \"Add missing :param statement in doc string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dfcd058144b46ce39a5f53036b31d1a77d85aad9"}]}],"author":{"name":"Dominic Schlegel","email":"dominic.schlegel@hostpoint.ch","date":"2019-10-21 07:14:09.000000000","tz":120},"committer":{"name":"Dominic Schlegel","email":"dominic.schlegel@hostpoint.ch","date":"2019-10-21 07:14:09.000000000","tz":120},"subject":"adjust doc string to correct key","message":"adjust doc string to correct key\n\nChange-Id: I194d17b64a953ed0497da02b9ec079a536965668\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f6ec559cac46ebe1e4edd2166328794d201b802b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f6ec559cac46ebe1e4edd2166328794d201b802b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
