)]}'
{"id":"openstack%2Fmistral~690015","triplet_id":"openstack%2Fmistral~stable%2Ftrain~If133cf472d1e0ecadd96bc07e7fd0d1ae4068740","project":"openstack/mistral","branch":"stable/train","topic":"bug/1849280","hashtags":[],"change_id":"If133cf472d1e0ecadd96bc07e7fd0d1ae4068740","subject":"Using std.ssh without private_key_filename causes TypeError","status":"MERGED","created":"2019-10-22 09:06:00.000000000","updated":"2019-10-23 21:41:19.000000000","submitted":"2019-10-23 21:40:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690015-1571866810648-5cb63f8a","meta_rev_id":"dcdea674a173501c043a4a62b80e1f7b24e4edf4","_number":690015,"virtual_id_number":690015,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 21:40:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6734,"name":"Jose Castro Leon","email":"josecastroleon@gmail.com","username":"josecastroleon"},{"value":0,"date":"2019-10-23 16:43:40.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":1,"date":"2019-10-23 08:24:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":1,"date":"2019-10-23 09:13:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-10-23 11:16:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6734,"name":"Jose Castro Leon","email":"josecastroleon@gmail.com","username":"josecastroleon"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-10-23 08:24:13.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"date":"2019-10-23 07:38:23.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2019-10-23 07:50:03.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2019-10-23 08:24:13.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-10-23 11:16:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-23 08:24:43.000000000","_account_id":6734,"name":"Jose Castro Leon","email":"josecastroleon@gmail.com","username":"josecastroleon"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6734,"name":"Jose Castro Leon","email":"josecastroleon@gmail.com","username":"josecastroleon"},{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 07:38:23.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-10-23 07:50:03.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-10-23 08:24:13.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-10-23 08:24:13.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-10-23 08:24:43.000000000","updated_by":{"_account_id":6734,"name":"Jose Castro Leon","email":"josecastroleon@gmail.com","username":"josecastroleon"},"reviewer":{"_account_id":6734,"name":"Jose Castro Leon","email":"josecastroleon@gmail.com","username":"josecastroleon"},"state":"REVIEWER"},{"updated":"2019-10-23 08:24:53.000000000","updated_by":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"reviewer":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"state":"REVIEWER"},{"updated":"2019-10-23 09:13:55.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-10-23 11:16:09.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-10-23 21:40:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd33d252e563f7c8ebceb543dc4064c7ce074263","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-22 09:06:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b96ae13110f6d6865610515934b063152d5f5c6","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-22 09:22:28.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch master as commit 660b79a9d20b53e86cfacb3de196e89c982fca86","accounts_in_message":[],"_revision_number":1},{"id":"899212dc265bdd49acf851218796e310124f874e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 11:27:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ca22fc1c8e74fbf81759be74082f29c : SUCCESS in 7m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8276cb414c34a9f9b8562a2008538c3 : SUCCESS in 12m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bce3b6522a634fa099baea7050c93949 : SUCCESS in 12m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d77fd1d6d9242d69bae15e9895b90cc : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5d2faa0db15460c97f6b8e32dd00af8 : SUCCESS in 6m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a3e5a165b384672b02e617b47c68e09 : SUCCESS in 8m 35s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5c207860d5b44c60b7fb501f98ece5ac : SUCCESS in 51m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7cfedeecf5814e59b64ec855a16bff89 : SUCCESS in 52m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e6704d8b0c9a476cb05a66d1b9ffc77d : SUCCESS in 52m 38s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6eb17fb0c938499bad3ab38e489128e9 : SUCCESS in 31m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/eb3e647d8bc6410fb158d56279cf1a6a : FAILURE in 4m 11s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/abe6337d2e7a4ee5ba3068e49b147840 : SUCCESS in 2h 00m 17s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c0f678f1db5c4d778441839070ec2464 : SUCCESS in 2h 12m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0df7dc5f9e1d4d6b9e5f21b1b4203c82 : SUCCESS in 1h 04m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76b3583b767d41d5993ae78c0bc6037e : SUCCESS in 10m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1c94ded667f2df8c16addf66ae3fe13452a08229","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-22 11:29:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f264fa7bc801113acb2359183fabb5504580434c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 14:44:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29afbeec68c14918ad2fc4e61ec383ae : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e8db5ba2e9641f29c51e401de90e583 : SUCCESS in 9m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9d3df567ab942e9890b854c06d6c2cb : SUCCESS in 8m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/acffdf38d88140bfabc60b7a23cef097 : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2436034001dd497f97bf0c02601c8b91 : SUCCESS in 5m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4f9d03c2dfe47839e1cc3063f337bd6 : SUCCESS in 9m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/37d392a7450b4735aa91a96189b618a4 : SUCCESS in 41m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2aa8f0538e8b45598661f75e3701ab9e : SUCCESS in 43m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/26e76e3870194abfa3a7b1d5406e399b : SUCCESS in 51m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/150d3f0a09c4471780107d883817443a : SUCCESS in 28m 52s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1b9cb73912a2467798ef26888c5d5a99 : SUCCESS in 26m 39s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/502eb080720f47cb89397fcd502e66ef : SUCCESS in 2h 14m 44s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d769d3d221a34bb39368eccc52c96c98 : SUCCESS in 3h 01m 50s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6d3e33d4375e4a019d08e08453f95940 : SUCCESS in 56m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbf1863381764895b6a97eb071988c18 : SUCCESS in 9m 33s","accounts_in_message":[],"_revision_number":2},{"id":"4702cce2a0be01a9c553d376d677d28845675ac8","author":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"date":"2019-10-22 18:27:30.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ea55718b3768a4b923a82be014f33768fcb15bb2","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-23 05:12:17.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6dce5da0a7e35ba52d158a723bacb3a2425161d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 08:20:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cb9cae364a34720a332dbc57651cd36 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7da429d612c743a7b26f3ce2c3513c9f : SUCCESS in 10m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c7eee45c95b4afa9b4169d07797f78c : SUCCESS in 10m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6abe3892afaa4701ba4da3413dbed3ed : SUCCESS in 10m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2c4bdb1ef18461e8d1e03aa40d262f0 : SUCCESS in 4m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc8f2cfb1a8a4459aa5962cb332e3926 : SUCCESS in 10m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fdfb8e54c02e455cb1ef709ac8327570 : SUCCESS in 42m 53s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/586d1c9835034bdc95e445906003b418 : SUCCESS in 43m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8b41917660244ca39a11b02ac8140391 : SUCCESS in 45m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bbcdc5486f3a493fa35ecd73b9502b08 : SUCCESS in 29m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0cf6998b8bb943f69e05fe37b8d2c5db : SUCCESS in 28m 02s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/2a82f967a2de4d9abd2b17b282befc91 : SUCCESS in 1h 34m 09s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3f8797646edd4f6bbc3f69fd5899442c : SUCCESS in 2h 37m 49s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4392a148aa084c108abbc7195e3b81f0 : SUCCESS in 1h 07m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eaa078f8443941b6b84f336f936a7855 : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":3},{"id":"297bd3ab62de3a91c6f78e6807bcf82b2436f0c6","author":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"date":"2019-10-23 08:24:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"04ebee32fc44c722e8cdd164ffd7369bba0dc786","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-10-23 09:13:55.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"64ba597d312af573ad2249341f6a6a37f2cdb4e3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-23 11:16:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7377aab7feade189d327badce519a952830bb057","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 11:16:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"623f4db7398259c6cf703a6a4ff585540a1a389c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 13:23:28.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1acba97cb0e84b2199c6104f8adee6cf : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0a1d3d41cc694398a14606bfec5ed7f5 : SUCCESS in 10m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c308f2b8431d498dbb4a255477e50001 : SUCCESS in 8m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/260048cf3cd3436396b2f90c3e42c0fe : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7ca5bb6015c4cd9bdcb6aa8676136fe : SUCCESS in 6m 11s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b7cd4165abaa433daf11d0e58d69a36e : FAILURE in 39m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9da8abd46ee9484592c9c4a251aea95e : SUCCESS in 34m 44s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/444f08ba97424dd8911192f3c61676db : SUCCESS in 27m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/01e75e958a5a408bbafe0367a74461fe : SUCCESS in 26m 34s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b11c0106155343388b708473d5b56aa5 : SUCCESS in 2h 00m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a1315a35959f46c4933349816c4c5088 : FAILURE in 48m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f68f91e78d84ded9cd20bfd05cd3278 : SUCCESS in 10m 09s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/74e2d3a4b6c248608070675b825d1b35 : SUCCESS in 1h 44m 39s","accounts_in_message":[],"_revision_number":3},{"id":"26c547741c467b2cc7f713a3719b51db263e186f","author":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"date":"2019-10-23 13:27:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a9243f2800df3cc0c8fc0e3384f5edad22c7f54a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 16:42:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8e58ab80416434294a5ffbb374f06af : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c346c1c0c4f6414ba7ca773241d7f04a : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c5cce12361043e5a8b90eead5a11fca : SUCCESS in 9m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4afcf44e4b034ad58781da4cd007c18c : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46ca4b9681bc4c02870a28ea11ead27c : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0df69d3d5bc49749be7afe6d3881ffc : SUCCESS in 9m 49s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8a3c8713f539467ba047a42a67a310e6 : SUCCESS in 53m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b41db34e123242e1b2b25ed937eb6e14 : SUCCESS in 46m 27s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8e4e127db2774c129f14d5749621a670 : SUCCESS in 44m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8ea4459b708f4290951b7486b2d27cc1 : FAILURE in 27m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a926c2b6df8546cca07d06343ad99ed3 : SUCCESS in 27m 28s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e9970483a3904425ada21024ccd3f377 : SUCCESS in 1h 31m 55s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ceb62ad941d546d5bd551b3074586711 : SUCCESS in 2h 50m 55s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ff0f1b826af34d16a65134925389b2aa : SUCCESS in 1h 00m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/024fc13dd453426fafe7a10eee35c039 : SUCCESS in 10m 31s","accounts_in_message":[],"_revision_number":3},{"id":"bb4cdc3bbdb09f35d3b4d8137f11e77228097b00","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-23 16:43:40.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e55d0858b698d5273d3ed9e760bb93a985e229e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 19:44:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84aeea5701d4437fb4a937bd58d69f08 : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3fb51e37562841d8acb02218278cd20c : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f281fb4e6ac40f4bf77374f37096175 : SUCCESS in 8m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63753447f31f4044b2db7bc47f812b0c : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/643a65b5d4e44de6a1d33ca15f10108b : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe03d5f4ca954ae8b1128c06d55c402b : SUCCESS in 10m 03s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/68be229711dd4af8a79277d8681a4edf : SUCCESS in 38m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f12db9a224c34d80b71ddaf999e9684c : SUCCESS in 50m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7fef6d9b2a4d496c8dc640401c827599 : SUCCESS in 46m 50s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/48f191c6640e4ef6aeea8e98278db9a6 : SUCCESS in 29m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/06b2a73b1a5142ce851f3c1721d31cff : SUCCESS in 27m 44s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/99141ba95d0a4c7ca4f1ab69a7bcc535 : SUCCESS in 1h 40m 27s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/77f221c437ba4b55a39a0090fe9d8f5b : SUCCESS in 2h 53m 19s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/417d49d5d6ec431196cabdd041a6fca0 : SUCCESS in 1h 10m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/664fd8d2424848689c52a24989d9976d : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":3},{"id":"1e816c90d06f1f22ebeb2bb3d500b32093f240bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 19:44:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b00a946000d09c436eae3eba00ff0997d80028cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 21:40:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e28a080698a46909bfbfc9a14b6f278 : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac99cc0e2762423db7431db2c38c2178 : SUCCESS in 8m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac0b861fa88c429497a72839cf02112c : SUCCESS in 7m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b33a700990045289f7a6a0ce47d4ec6 : SUCCESS in 9m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dab51eafc654511be7e11ddf591a5d6 : SUCCESS in 4m 20s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c23163dd609c43ae86363fde806153eb : SUCCESS in 39m 55s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/dcb8aa46201a4556b70011c7c25f40e8 : SUCCESS in 41m 45s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a5eba538222f441cb223b53b126b267c : SUCCESS in 29m 39s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c6923dc95f084effa3bb200a32889791 : SUCCESS in 27m 48s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/620770fcdf8546318e605967cdc3b593 : SUCCESS in 1h 33m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/05011df28f7848ae9f6ff32045791a14 : SUCCESS in 56m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08b3e74d5b294e66b16f1ac7aacff134 : SUCCESS in 11m 26s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/cdaf158b71094730a75b88194d3a407d : SUCCESS in 1h 53m 21s","accounts_in_message":[],"_revision_number":3},{"id":"1d9a09d92c9acc1979ae32f280cedec26f208c65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 21:40:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"19f204c606a0bca18112f4bbee4184c81b82f625","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 21:41:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96604eb870974aa1903157a557e89e6b : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0a3cbfb5d6dd4f8b926b4d0f2ab5f9a2ea281419","revisions":{"855f21ebcc3b132bf71da2a28b5013a7cf75e5e5":{"kind":"REWORK","_number":1,"created":"2019-10-22 09:06:00.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/15/690015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/15/690015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/690015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/690015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/690015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/15/690015/1"}}},"commit":{"parents":[{"commit":"5d387db6c2a53e3c28d46ffd7da45cf1f52a3d8b","subject":"Change the action error message format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5d387db6c2a53e3c28d46ffd7da45cf1f52a3d8b"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-22 09:05:43.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-22 09:05:43.000000000","tz":180},"subject":"Using std.ssh with username and password causes TypeError","message":"Using std.ssh with username and password causes TypeError\n\nThis fixes this bug\n\nChange-Id: If133cf472d1e0ecadd96bc07e7fd0d1ae4068740\nCloses-Bug: #1849280\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/855f21ebcc3b132bf71da2a28b5013a7cf75e5e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/855f21ebcc3b132bf71da2a28b5013a7cf75e5e5"}]},"branch":"refs/heads/stable/train"},"45ff5b26ff7e000f2b095a8920702d1eff2b3b3e":{"kind":"REWORK","_number":2,"created":"2019-10-22 11:29:02.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/15/690015/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/15/690015/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/690015/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/690015/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/690015/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/15/690015/2"}}},"commit":{"parents":[{"commit":"5d387db6c2a53e3c28d46ffd7da45cf1f52a3d8b","subject":"Change the action error message format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5d387db6c2a53e3c28d46ffd7da45cf1f52a3d8b"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-22 09:05:43.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-22 11:28:58.000000000","tz":0},"subject":"Using std.ssh with username and password causes TypeError","message":"Using std.ssh with username and password causes TypeError\n\nThis fixes this bug\n\nChange-Id: If133cf472d1e0ecadd96bc07e7fd0d1ae4068740\nCloses-Bug: #1849280\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/45ff5b26ff7e000f2b095a8920702d1eff2b3b3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/45ff5b26ff7e000f2b095a8920702d1eff2b3b3e"}]},"branch":"refs/heads/stable/train"},"0a3cbfb5d6dd4f8b926b4d0f2ab5f9a2ea281419":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-23 05:12:17.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/15/690015/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/15/690015/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/690015/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/690015/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/15/690015/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/15/690015/3"}}},"commit":{"parents":[{"commit":"5d387db6c2a53e3c28d46ffd7da45cf1f52a3d8b","subject":"Change the action error message format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5d387db6c2a53e3c28d46ffd7da45cf1f52a3d8b"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-22 09:05:43.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-23 05:12:14.000000000","tz":0},"subject":"Using std.ssh without private_key_filename causes TypeError","message":"Using std.ssh without private_key_filename causes TypeError\n\nCheck the private_key_filename parameter only if it is not None\n\nChange-Id: If133cf472d1e0ecadd96bc07e7fd0d1ae4068740\nCloses-Bug: #1849280\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0a3cbfb5d6dd4f8b926b4d0f2ab5f9a2ea281419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0a3cbfb5d6dd4f8b926b4d0f2ab5f9a2ea281419"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
