)]}'
{"id":"openstack%2Fmistral~691194","triplet_id":"openstack%2Fmistral~master~I6619c686847ca0ec9271a755cc480dd10a961816","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I6619c686847ca0ec9271a755cc480dd10a961816","subject":"Make test_action_definition_cache_ttl more robust","status":"MERGED","created":"2019-10-25 05:54:03.000000000","updated":"2019-11-03 06:58:17.000000000","submitted":"2019-11-03 06:56:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691194-1572764199356-752aeac9","meta_rev_id":"8a745a0fb42d32d57acf08df2ae6245a308a713b","_number":691194,"virtual_id_number":691194,"owner":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-03 06:56:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2019-10-25 06:23:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-11-03 05:22:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-25 06:24:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-10-25 05:54:57.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"date":"2019-10-25 05:54:30.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-11-03 05:22:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 05:54:30.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-10-25 05:54:57.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-10-25 06:23:41.000000000","updated_by":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"reviewer":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"state":"REVIEWER"},{"updated":"2019-10-25 06:24:34.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-11-03 05:22:43.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-03 06:56:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"616eacdc485870ca12d44762987457e171669ccc","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-10-25 05:54:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d8aa1dc64a05ec6df1938ce4ccfeaff9e841076","author":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"date":"2019-10-25 06:23:41.000000000","message":"Patch Set 1: Code-Review+1\n\nImproving robustness of tests is very important. Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"0749e3ca8a033c965cb200729466fedd363bd7e7","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-25 06:24:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6a55a7181d6ea97983643b61ad0e2f33617dd1a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 08:57:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec4d95ac3a5343bf89456f765280a393 : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb36e424df42472992126ce92794e3aa : SUCCESS in 9m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/336db1bf885e4dee8d29a41a2bb6cb13 : SUCCESS in 9m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbd5cd5e3d8846c38c28d56c620174df : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c5d6d4ea848451597157fe88f2e6d91 : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ecdc374e2454fb6918d6d91fd4c8fc0 : SUCCESS in 10m 21s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fc6da023d9db4d2fa8164cb06aa6df64 : SUCCESS in 44m 55s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4a581497e1e644adb90f44a9de774903 : SUCCESS in 43m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ca99e517543d46479396044a2b069200 : SUCCESS in 44m 47s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2464ea5e807c4e8c94fcc0bfaa8fac72 : SUCCESS in 30m 37s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/360f08f8a39f48328f8fa54a1fe1fe41 : SUCCESS in 30m 01s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/d2a9d8a21b1e425088b23b15028a36a7 : SUCCESS in 2h 08m 56s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/19796bfc0f904de1b0848329db2a677b : SUCCESS in 2h 58m 15s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b1d3b0c84aa74a1b98f00824ccebea05 : SUCCESS in 1h 10m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbf64f53b143438cb8ddfc34ffcedbab : SUCCESS in 10m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c6f2817807023c8d03ad7a84063f0d321b7b2cc4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-03 05:22:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4e1f6f843a18fe620f6af37a8345c112a53cd82d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-03 05:22:43.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7e2d30b5b1cae5180193cb767a6e3ddf4e7c6d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 05:22:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2dada85255af44dec3f848bd216d8fa6c78afdaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 06:56:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b221a39e33d545b48ed1ba5543524a93 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f071b6925d64dd39e40a5625836e780 : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d497cfe321664ca095063e49f9e4dbe0 : SUCCESS in 8m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef3f90f0c19e477a913220931e3763b6 : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24aa9d3b590e4ba990e96f7a001ac4fb : SUCCESS in 4m 02s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/213821da20764e21afb2d4d0499cd21a : SUCCESS in 42m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6e54a40c26524f818767db20f058e165 : SUCCESS in 38m 14s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/912c71ab144547cc97c6e1695543c003 : SUCCESS in 29m 01s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/dd9d7445fa1c4f818f03cdf8d5e191a5 : SUCCESS in 28m 01s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/eef13f08d28d499cb6632fea9b3f05ad : SUCCESS in 1h 19m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/490f045781ca49ea9a190def780010bf : SUCCESS in 39m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d8e048b69fc44bab5547631b8b2a0d5 : SUCCESS in 9m 54s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/09835325e67343e58a1df1bc5a2e6b4a : SUCCESS in 1h 32m 11s","accounts_in_message":[],"_revision_number":1},{"id":"53f32bd9a87e6af8c3a3996037c42be5d64e926c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 06:56:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8a745a0fb42d32d57acf08df2ae6245a308a713b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 06:58:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf7cbd02c54c465ea4853343aca2d303 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50e9d31a5c4d4067486812b91ba8d1abb9daf1dc","revisions":{"50e9d31a5c4d4067486812b91ba8d1abb9daf1dc":{"kind":"REWORK","_number":1,"created":"2019-10-25 05:54:03.000000000","uploader":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"ref":"refs/changes/94/691194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/94/691194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/94/691194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/94/691194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/94/691194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/94/691194/1"}}},"commit":{"parents":[{"commit":"aaf687941360978401090b4c66244bdd5885fed8","subject":"Cap psycopg2 to 2.8.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aaf687941360978401090b4c66244bdd5885fed8"}]}],"author":{"name":"Andras Kovi","email":"akovi@nokia.com","date":"2019-10-25 05:53:46.000000000","tz":120},"committer":{"name":"Andras Kovi","email":"akovi@nokia.com","date":"2019-10-25 05:53:46.000000000","tz":120},"subject":"Make test_action_definition_cache_ttl more robust","message":"Make test_action_definition_cache_ttl more robust\n\nIn the Zuul pipeline the test may run very slow with real DB. It is better\nto control the expiration of the cache deterministically.\n\nChange-Id: I6619c686847ca0ec9271a755cc480dd10a961816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/50e9d31a5c4d4067486812b91ba8d1abb9daf1dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/50e9d31a5c4d4067486812b91ba8d1abb9daf1dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
