)]}'
{"id":"openstack%2Fmistral~691585","triplet_id":"openstack%2Fmistral~master~I4d027fb28f95c3aa0803f7b5a011f835ec7d615a","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I4d027fb28f95c3aa0803f7b5a011f835ec7d615a","subject":"Stop testing python2.7","status":"ABANDONED","created":"2019-10-28 05:47:23.000000000","updated":"2019-11-18 04:37:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d73b6d99ba836ad09f7d6f28f99ec1deab3be392","_number":691585,"virtual_id_number":691585,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-28 07:03:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":-1,"date":"2019-10-31 10:51:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":-1,"date":"2019-11-15 14:49:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 07:03:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-31 10:51:04.000000000","updated_by":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"reviewer":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"state":"REVIEWER"},{"updated":"2019-11-15 14:49:22.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"815c15cd18c915ac13bc4b95400a7f5a8a1170f7","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-10-28 05:47:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ffe23adc7f1c149b4f1b2777c7c5424b93a6564","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 07:03:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/50a4e68d3ff14627a778861be0122a30 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9149c7c0e08b4748b87fa4db1e4f2456 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/393c855899ea4b5295c8bf3274933e13 : SUCCESS in 8m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/757397d6aaf64fe9be8c4a9b8b5be768 : SUCCESS in 10m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d09a2755d56a4aa3ab3b4d065c865d13 : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7c7c71161ef4681ae0f9cb5d9e1a914 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcf85bc21a4d450abdabacbaa752ea16 : SUCCESS in 4m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4725ce13b3bf432e8cc559a38146cab6 : SUCCESS in 9m 29s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/36d834598056415a87401bffc5374ebb : SUCCESS in 40m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c0c2fb1f71cf4f75b83e55145e16719e : SUCCESS in 43m 19s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/954fa4381ec34bf1831ec524e45fff93 : SUCCESS in 39m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/dfd124c6888e46aba7a7fc342ab6fe24 : SUCCESS in 29m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/11a213823c8849dd82a98d37b2d93146 : SUCCESS in 29m 11s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1621831b77344b00a4eaed7c5d1e72fc : SUCCESS in 1h 07m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0203339d33f04c48bc1ccf58d8e72392 : SUCCESS in 9m 39s","accounts_in_message":[],"_revision_number":1},{"id":"33d767b32c85b81eb3129d1ac490221a4b6c8e71","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-10-31 10:51:04.000000000","message":"Patch Set 1: Code-Review-1\n\nPatch Set 1: Code-Review-1\n\nStill, zuul job for python 2 is running, as seen in the zuul check section of this patch.\n\nSo, please do these changes also:\n1) Remove the python2 job from zuul conf file.\n2) Please add the release note also.","accounts_in_message":[],"_revision_number":1},{"id":"4c00e8c134fb2f21dff6e8d777977798016a0db5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-15 14:49:22.000000000","message":"Patch Set 1: Code-Review-1\n\ndoing all the updates as per goal - https://review.opendev.org/#/c/694556/","accounts_in_message":[],"_revision_number":1},{"id":"d73b6d99ba836ad09f7d6f28f99ec1deab3be392","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-18 04:37:44.000000000","message":"Abandoned\n\nDuplicates the mentioned patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8fd91ec865e5d24d7f447115c99ebed3faf5c28","revisions":{"e8fd91ec865e5d24d7f447115c99ebed3faf5c28":{"kind":"REWORK","_number":1,"created":"2019-10-28 05:47:23.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/85/691585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/691585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/691585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/691585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/691585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/691585/1"}}},"commit":{"parents":[{"commit":"33959241933cad0c65d47877dbc8722f4192505e","subject":"Merge \"Prohibit creation workflows with spaces in name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/33959241933cad0c65d47877dbc8722f4192505e"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-10-28 05:46:44.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-10-28 05:47:16.000000000","tz":480},"subject":"Stop testing python2.7","message":"Stop testing python2.7\n\nChange-Id: I4d027fb28f95c3aa0803f7b5a011f835ec7d615a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e8fd91ec865e5d24d7f447115c99ebed3faf5c28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e8fd91ec865e5d24d7f447115c99ebed3faf5c28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
