)]}'
{"id":"openstack%2Fmistral~691856","triplet_id":"openstack%2Fmistral~master~I54a4cd38e962d363fd2626476bcae9ec0aa8dad6","project":"openstack/mistral","branch":"master","topic":"bug/1850315","hashtags":[],"change_id":"I54a4cd38e962d363fd2626476bcae9ec0aa8dad6","subject":"Evaluate input expression should check the in_context","status":"MERGED","created":"2019-10-29 12:20:52.000000000","updated":"2019-10-30 11:15:30.000000000","submitted":"2019-10-30 11:13:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691856-1572434033380-771dd57b","meta_rev_id":"17d80cf02c554443d6a6a8889d58485334c658a3","_number":691856,"virtual_id_number":691856,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-30 11:13:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":2,"date":"2019-10-30 09:40:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-10-30 09:37:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-10-30 07:00:24.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"date":"2019-10-30 07:00:24.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2019-10-30 07:00:24.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"date":"2019-10-30 07:00:24.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2019-10-30 09:37:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 15:53:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-10-29 16:10:40.000000000","updated_by":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"reviewer":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"state":"REVIEWER"},{"updated":"2019-10-30 07:00:24.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-10-30 07:00:24.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-10-30 07:00:24.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-10-30 07:00:24.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-10-30 09:37:35.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-10-30 09:40:13.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-10-30 11:13:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18d78d287d1989649309c55244f0929eaaa79769","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-29 12:20:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f4469247ab0b0a2c799202e42abdbe2c2b57b25","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-29 13:03:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3e3879b98b98cfe7c10e6217f1f5961691d04433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 15:47:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59ce914a329e41ec8c2c1da422f55f8b : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04454a8300d1466f814d915b3438ed3d : SUCCESS in 9m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6901988923c0484988161ed3532964a7 : SUCCESS in 8m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1fed420b9b434275ad7ed6e0f69e407d : SUCCESS in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e2b0f3bf3d445369ef67483d5cf70b8 : SUCCESS in 5m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a421013e1f5412e94d73de59e4acc05 : SUCCESS in 8m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a4f7ab434d9d480c9153aa9afe0259ce : SUCCESS in 52m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/83f59dba51334bb09545b13a38533a88 : SUCCESS in 36m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0580ed676dbc44699c426ecca64109b2 : SUCCESS in 41m 44s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/81cf7f29893c472c9ae024ab625d557f : SUCCESS in 29m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ab3f92c7e8e44468b5f102fae0c99852 : SUCCESS in 29m 29s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/12fd61593c2040789d5ef3eb86cb2263 : SUCCESS in 1h 49m 54s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/070098bfe654499d99e6f2acb953439f : SUCCESS in 2h 42m 07s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bb6b6864672f4c11a30f8322c7a1d6a2 : SUCCESS in 1h 00m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6bcd3e02c7eb4c3ea249ca331aed095e : SUCCESS in 10m 17s","accounts_in_message":[],"_revision_number":2},{"id":"066b1592ff457c9b349e35356e764852ae78cd21","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-29 15:53:15.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nChanges LGTM but see my comment inline:","accounts_in_message":[],"_revision_number":2},{"id":"8ff38afc7d772588067d07151510255e43fb5829","author":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"date":"2019-10-29 16:10:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"862cebf7121a420d0478d6d083213755abea6572","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-30 04:08:07.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nI wonder how we overlooked it.. looks good to me, the commit message though is not so clear.","accounts_in_message":[],"_revision_number":2},{"id":"da9760e90d99dac42598baf81c5fde411aa6e364","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-30 06:15:56.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"114ae7eb4988767f8e60b53c264f728d637243fc","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-30 06:35:01.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 090a255a58aa1efa80ea98203509440b3da18446","accounts_in_message":[],"_revision_number":3},{"id":"27cb0967616914d08966584dfc94ba68ee60885f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 09:20:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/414982bc8f6048dc8a904f811c8777e2 : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad7eeda12d7b4a8582e90c3fee673330 : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fd5cc6103664d449f31f24338c2b107 : SUCCESS in 9m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8b0703bfefe4802be170c2e0881005e : SUCCESS in 10m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa8fffd5bb664c49b8c318e4c8ed5c98 : SUCCESS in 6m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b4eb7c43bf24305b7a02522edd972f9 : SUCCESS in 10m 57s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ae42b4d31fbb47ddb6727484c7ed2782 : SUCCESS in 39m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/98e6eb2d59d048809006f816d3af135d : SUCCESS in 37m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d9c279ae415c4044813e50cd71941aa4 : SUCCESS in 37m 59s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/84c901d3afe4476993375811e1447515 : SUCCESS in 30m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/22e3b58b04f94fb1ae418b4b3334c4a5 : SUCCESS in 31m 20s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/66d2bb09325c48398ee23c687c30fc58 : SUCCESS in 1h 39m 38s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7355204be9ee493e9172547f4c227200 : SUCCESS in 2h 41m 13s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c52b58a2dfde44188b54f6ed0954ab00 : SUCCESS in 1h 04m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dec606ad5d62455d9472ddf8abc43789 : SUCCESS in 11m 35s","accounts_in_message":[],"_revision_number":3},{"id":"065072c04461e52ae945b88dc25b85bb4e686740","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-30 09:37:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fedd55254630e5b9b9dc2abea10955628e1e0792","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 09:37:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"234469b339de2ecb3510aaff6312e8f413759591","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-10-30 09:40:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"84734f1c1707382a2100a1e85432a2c62b1b17eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 11:13:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7b0064e68c4854b1600ea7a1f76c305e5b739b36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 11:13:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ce8f63e41a2495bada63afe940a1a12 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c7c20340fc3c496d940ed4f5eab46255 : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed843f00177241fb9693851874e2e112 : SUCCESS in 8m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/399577e49fa147048ee7fc6033f8ca6c : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44bd6b8b0e8a46fd9b3b5d10e7dc88b3 : SUCCESS in 5m 21s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f10ac8262d0c454b9e505541edfe64f3 : SUCCESS in 41m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/939504597cd9444cb6e73c15c1e92fe2 : SUCCESS in 39m 06s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/98d2199a9beb46c2884ac914538c8f6d : SUCCESS in 29m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0d4208e0875f4141880f3af3667668e4 : SUCCESS in 29m 12s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/22e1d37909b94a43b66d939af5d0c7a4 : SUCCESS in 1h 33m 31s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bf6f3b9d5e0542bb9fe53874db9c5a8f : SUCCESS in 38m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6c91de1fe464144aa36ec572a3c18a9 : SUCCESS in 10m 40s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/b3b3fdc1097843e79c4d384fa344f11c : SUCCESS in 1h 32m 11s","accounts_in_message":[],"_revision_number":3},{"id":"8f708d53243131527a89e41ae7bb2ac9adfdedd1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 11:15:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b0125d7783b47f296c9a664c867712a : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a68136d1e46c2c0193a7d94626a3125bfc32a74c","revisions":{"bbd1b8d8a15d0e8bc17a8ac1c9a69a5802528fd5":{"kind":"REWORK","_number":1,"created":"2019-10-29 12:20:52.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/56/691856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/691856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/691856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/691856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/691856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/691856/1"}}},"commit":{"parents":[{"commit":"33959241933cad0c65d47877dbc8722f4192505e","subject":"Merge \"Prohibit creation workflows with spaces in name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/33959241933cad0c65d47877dbc8722f4192505e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-29 12:20:38.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-29 12:20:38.000000000","tz":120},"subject":"Evaluate input expression should check in context","message":"Evaluate input expression should check in context\n\nThe workflow in the test fails because contextView\ndoes evaluate in_context\n\nChange-Id: I54a4cd38e962d363fd2626476bcae9ec0aa8dad6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bbd1b8d8a15d0e8bc17a8ac1c9a69a5802528fd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bbd1b8d8a15d0e8bc17a8ac1c9a69a5802528fd5"}]},"branch":"refs/heads/master"},"b2e515821a9728fbb656bd2aa87db97436523723":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-29 13:03:22.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/56/691856/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/691856/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/691856/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/691856/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/691856/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/691856/2"}}},"commit":{"parents":[{"commit":"33959241933cad0c65d47877dbc8722f4192505e","subject":"Merge \"Prohibit creation workflows with spaces in name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/33959241933cad0c65d47877dbc8722f4192505e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-29 12:20:38.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-29 13:03:03.000000000","tz":120},"subject":"Evaluate input expression should check in context","message":"Evaluate input expression should check in context\n\nThe workflow in the test fails because contextView\ndoes evaluate in_context\n\nCloses-Bug: #1850315\nChange-Id: I54a4cd38e962d363fd2626476bcae9ec0aa8dad6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2e515821a9728fbb656bd2aa87db97436523723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2e515821a9728fbb656bd2aa87db97436523723"}]},"branch":"refs/heads/master"},"a68136d1e46c2c0193a7d94626a3125bfc32a74c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-30 06:15:56.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/56/691856/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/691856/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/691856/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/691856/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/691856/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/691856/3"}}},"commit":{"parents":[{"commit":"33959241933cad0c65d47877dbc8722f4192505e","subject":"Merge \"Prohibit creation workflows with spaces in name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/33959241933cad0c65d47877dbc8722f4192505e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-29 12:20:38.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-30 06:15:52.000000000","tz":0},"subject":"Evaluate input expression should check the in_context","message":"Evaluate input expression should check the in_context\n\nThe workflow in the test fails because contextView\ndoes not evaluate in_context\n\nCloses-Bug: #1850315\nChange-Id: I54a4cd38e962d363fd2626476bcae9ec0aa8dad6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a68136d1e46c2c0193a7d94626a3125bfc32a74c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a68136d1e46c2c0193a7d94626a3125bfc32a74c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
