)]}'
{"id":"openstack%2Fmistral~692042","triplet_id":"openstack%2Fmistral~stable%2Ftrain~I54a4cd38e962d363fd2626476bcae9ec0aa8dad6","project":"openstack/mistral","branch":"stable/train","topic":"bug/1850315-stable/train","hashtags":[],"change_id":"I54a4cd38e962d363fd2626476bcae9ec0aa8dad6","subject":"Evaluate input expression should check the in_context","status":"MERGED","created":"2019-10-30 06:35:01.000000000","updated":"2019-10-30 19:35:54.000000000","submitted":"2019-10-30 19:34:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692042-1572464085981-7361ecdf","meta_rev_id":"a440ededed8380c106a718df4609e6a62814b2e2","_number":692042,"virtual_id_number":692042,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-30 19:34:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-30 15:57:30.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":1,"date":"2019-10-30 07:00:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-10-30 09:38:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-10-30 09:32:16.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"date":"2019-10-30 09:32:16.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2019-10-30 09:32:16.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2019-10-30 09:32:16.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"date":"2019-10-30 09:32:16.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-10-30 09:38:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-30 07:00:54.000000000","updated_by":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"reviewer":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"state":"REVIEWER"},{"updated":"2019-10-30 09:32:16.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-10-30 09:32:16.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-10-30 09:32:16.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-10-30 09:32:16.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-10-30 09:32:16.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-10-30 09:38:45.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-10-30 19:34:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b97c089eee27474b984542a88f089712ae0fdbe8","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-30 06:35:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4567898164c1f6d9deb3c9b9ae332920ec40219a","author":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"date":"2019-10-30 07:00:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"773d0e9634c0dcb98d0a9159c95004b44aa14b55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 09:18:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb2a72b3ecde4d51b26698a6bbf01d1d : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/963db9ed91c842b782fc755403f0b68a : SUCCESS in 10m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/128eac98538847299dd1644319692611 : SUCCESS in 10m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6af5e1a297cf4a73b5401bf6609ca572 : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b05531a7f5549c6b9a6d2e45510e534 : SUCCESS in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6434454f7ed84bc6bae6209549784801 : SUCCESS in 9m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5ec0e45182d04016b33e25e218148696 : SUCCESS in 39m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6fb1be4a5ad84d3b9a0edd846f195335 : SUCCESS in 42m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/70226e2fc2144525be8918669adf9e12 : SUCCESS in 41m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/33713d9d9efb47159a4c921e03e39367 : SUCCESS in 29m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9b3a9d67b856456684db84511464f0f7 : SUCCESS in 27m 39s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/ed06e00549db4613a7c9718fb16cf05a : SUCCESS in 1h 22m 55s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6d7c7aeec29549b991e2807c92f1ce3e : SUCCESS in 2h 19m 18s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/172b6c9049f444548e4b3b3a113b0aca : SUCCESS in 1h 08m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7eeccd64aaad485d98ea48b6b8ea8ec0 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5fb0f8270c3b771083c41c289aa6cd03679891a0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-30 09:38:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0c119d0bf7b308a8ebeee1f88c4edea05d53193e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-30 09:38:45.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d9941c9ba394fdb45860fd5fc502f26cb2ce5df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 09:38:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36e0ab7bf1f3968f11da895eac4ce5dbe018bed5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 11:13:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd45341048e452c97349b4f7d16d58a : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/55ba3b56056b47a88e0589e5ec16b531 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81b7d068df904fc9ad7d242cc452b89e : SUCCESS in 8m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1af07f3ca724eb4941972314a6123c5 : SUCCESS in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7407bb933c5441fea5507b3a2233dd76 : SUCCESS in 5m 07s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a11068b3ab6d46d881724a675921a503 : SUCCESS in 32m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c0e2f3d103534a85b825301bb8325bc6 : FAILURE in 42m 11s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cefe2dafac5545c9920f4e54c7d44680 : SUCCESS in 28m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/55bc9ec5ef2c4f0bbaff8a1106a380df : SUCCESS in 28m 54s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/46b01f8397e64083b9065076d3abb2ea : SUCCESS in 1h 24m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/20649964e5a24e399cfcf25f7ae724fa : SUCCESS in 39m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44110bc459314cfbbad1872533b29a32 : SUCCESS in 10m 20s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/335a41368f5449afb8a219356eec4d76 : SUCCESS in 1h 30m 32s","accounts_in_message":[],"_revision_number":1},{"id":"fa6e15f445fe529dedd4031ca712d2be048e1512","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-30 11:16:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"be123af5cf8a81111c4e02c7f491bbe0d961b84f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 14:00:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ff426ccd61b4eaea9c62f743409f33d : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/79e445ffe6d5492bbcdfcd61b45cf8be : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/676bf6a5ff01480d81345191158be570 : SUCCESS in 9m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9606a8c259aa4eb19d6954a143c11b0e : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/455ea6d15b804caca061898137df5e0b : SUCCESS in 5m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b41e4b6d808b45ee83c66046e241a29f : SUCCESS in 8m 40s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a55dfbb9b404457d9c62fc4e07a2a4ae : SUCCESS in 39m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6b21a604f791499197b894627068c4d1 : SUCCESS in 37m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b331fccb498e4bd0a43a20cf08128dc5 : SUCCESS in 57m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f0bdaad5c9554243826e1fa14833406f : SUCCESS in 28m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/39adb621beb04d85b9c4719ff9078dff : SUCCESS in 28m 46s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/839f57c8faad47139cdb607d83b245a2 : SUCCESS in 1h 32m 08s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4a15ecb9077441309cda6448ea71f26c : SUCCESS in 2h 36m 49s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d0413ff844bd4858bccf3fba06621b6c : SUCCESS in 58m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87a50fab3190445bb48c7e3860d6d1f3 : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c0d4ec7cdd2b96249c91b9e51ecf3a4b176397c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 14:00:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f39aff9d5187f6d3fa276fcb5a8d3e14793c309d","author":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"date":"2019-10-30 15:10:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a5aaad08b8769d1a9828facb53511391aaceaf64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 15:48:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5803a04e9924cb69d591f60858ddec5 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3bb21a898e64cbfa8d53db53ee82d2d : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b16e71e8333b44d1bb3ec4868c512341 : SUCCESS in 8m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23a78960a5ab49438430d00f092c5394 : SUCCESS in 9m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17a025e73908493fafe6d2668d3c762b : SUCCESS in 4m 28s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b55efe3ac56d487b85e730a6aef95ab7 : SUCCESS in 47m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/246b43332b52435a91f16f238cb77f7e : SUCCESS in 52m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/86a4e660b44742dd970cfa830f187998 : FAILURE in 31m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/da5084effb074b4bb147753d806bd4d5 : SUCCESS in 26m 32s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/bf2ac5b6a93f4e3b91b287db9e62f253 : SUCCESS in 1h 43m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c61dbc12ff0840cba8ee0e88d0aca88b : SUCCESS in 37m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1efac9e3c38c4c6cab2538f299acfe5d : SUCCESS in 10m 15s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/90027cd006514091804b7da32f2c74f6 : SUCCESS in 1h 25m 28s","accounts_in_message":[],"_revision_number":1},{"id":"9a42e4ab32c1aeaa0a77421f1a4b7bc5fa83c1d6","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-10-30 15:57:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e22e05cd6f5c6268f043b6b562d6f50203b4e6b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 17:53:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17197cbae4fa43898c8e2fde8be4a29b : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/03ec32d7971d4875944a50d2ec5fc479 : SUCCESS in 9m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00ac1029747849cd848929c362301e88 : SUCCESS in 7m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7447666ff754829a19ec7e7a8791e72 : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e45cc714f134914aeb5f98b5fa5e383 : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8327e8e711ed4858ac9667ddc2bfb827 : SUCCESS in 8m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d5cf271953ff4370bc8208f053b33297 : SUCCESS in 55m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e2b2f371aeb54d6282d71eb303a54187 : SUCCESS in 51m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/806f529e5ea74d29b83a4f9cabf4cd9a : SUCCESS in 55m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/51383d07dd4740939b6addae3a3dc217 : SUCCESS in 31m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9a4c0e52ed1f48ef86f25e51bcf40ced : SUCCESS in 27m 25s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a474d4f01b164a10a6632e2e7de260aa : SUCCESS in 1h 14m 42s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/57de5858bd594194baf93b253ee4a904 : SUCCESS in 2h 39m 28s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/815fc99c2c9f4f5d914102ae31f9a1cd : FAILURE in 1h 03m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9409b64474334dc0a85aedef68bbf406 : SUCCESS in 12m 02s","accounts_in_message":[],"_revision_number":1},{"id":"147cf7d060bd2c72f8736c5d459161308065fa55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 17:53:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8cee5597d472d3ab4f65f4d15016de8d9a8dc0e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 19:34:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b43987f3db264cd4933d1aa25ecba8a7 : SUCCESS in 7m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/123108c3d9604386a612808f9410d708 : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30ba87cab7ca4e9ab3f4822ef90de472 : SUCCESS in 8m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/765e8abb84cd4714ac3e16256ced2745 : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68e9d49a99d4413e890ebd4e21234d9a : SUCCESS in 6m 19s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/98b7b7204d5a452aad93a3a5cfe28ad0 : SUCCESS in 48m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/16971a0ee30a42f3b520b551cb260134 : SUCCESS in 54m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/df49cb956cca4641b0917d8bdac62765 : SUCCESS in 31m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2ed004d01657447e9c78dc972110a47c : SUCCESS in 27m 04s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/3539e2ad0d35470fba69c7716ed77568 : SUCCESS in 1h 20m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/74f063650c62441fb3f4e6cc8f8cba0c : SUCCESS in 43m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef3d6a11f3d94742a998a4237dccf554 : SUCCESS in 11m 13s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/fdd8fee877014f1f9c35affb54c3272d : SUCCESS in 1h 36m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f6deaa119fd927747bac2cb24c5e7d6f95eabd64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 19:34:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c67a2ad2543a3fe77e6dbe23e409c06796f21156","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 19:35:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d622df86754a7c9eeb1a7f6a98e486 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"090a255a58aa1efa80ea98203509440b3da18446","revisions":{"090a255a58aa1efa80ea98203509440b3da18446":{"kind":"REWORK","_number":1,"created":"2019-10-30 06:35:01.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/42/692042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/42/692042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/42/692042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/42/692042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/42/692042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/42/692042/1"}}},"commit":{"parents":[{"commit":"0a3cbfb5d6dd4f8b926b4d0f2ab5f9a2ea281419","subject":"Using std.ssh without private_key_filename causes TypeError","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0a3cbfb5d6dd4f8b926b4d0f2ab5f9a2ea281419"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-29 12:20:38.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-10-30 06:35:01.000000000","tz":0},"subject":"Evaluate input expression should check the in_context","message":"Evaluate input expression should check the in_context\n\nThe workflow in the test fails because contextView\ndoes not evaluate in_context\n\nCloses-Bug: #1850315\nChange-Id: I54a4cd38e962d363fd2626476bcae9ec0aa8dad6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/090a255a58aa1efa80ea98203509440b3da18446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/090a255a58aa1efa80ea98203509440b3da18446"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
