)]}'
{"id":"openstack%2Fmistral~693019","triplet_id":"openstack%2Fmistral~stable%2Frocky~I2571246947ab5ae06731fc7d35de7a5954325a82","project":"openstack/mistral","branch":"stable/rocky","topic":"releasenote-bug-1843175-stable/rocky","hashtags":[],"change_id":"I2571246947ab5ae06731fc7d35de7a5954325a82","subject":"Add release note for fix error of cron trigger run","status":"MERGED","created":"2019-11-05 14:37:51.000000000","updated":"2019-11-07 14:42:03.000000000","submitted":"2019-11-07 08:38:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693019-1573115881992-8bdcbf74","meta_rev_id":"03539cf85c9a26edbcd3d22273a72bb120bebdff","_number":693019,"virtual_id_number":693019,"owner":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-07 08:38:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":2,"date":"2019-11-07 06:47:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-11-05 14:40:45.000000000","_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"date":"2019-11-05 14:41:17.000000000","_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":1,"date":"2019-11-07 06:47:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-05 14:40:45.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2019-11-05 14:41:17.000000000","updated_by":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"reviewer":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"state":"REVIEWER"},{"updated":"2019-11-07 06:47:29.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-07 08:38:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77400f466810b766a65517f3f4d6b485d6900d84","author":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"date":"2019-11-05 14:37:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"09b141f0c107e99298f4cd1b373f8fa34371807a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 15:09:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bec04ab801a140e2a025db847de6a894 : SUCCESS in 5m 03s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e749c2c0ff9346ec935b4c492ab158d4 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b699ee4c8d4541618495a75f20ee928c : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2822ddc9ca942f6bbe8e0f8dd530d72 : SUCCESS in 8m 10s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/234819a333344f56b4edff629157f079 : SUCCESS in 26m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/77301bfda22d4874a7a6d824ea2e97eb : SUCCESS in 24m 52s","accounts_in_message":[],"_revision_number":1},{"id":"17d3dfdfb1a449e5f5b2f1797b6c2bafb7c3e462","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-06 04:31:37.000000000","message":"Patch Set 1:\n\nHi, do you need it only in stable/rocky? What about stein and train backports?","accounts_in_message":[],"_revision_number":1},{"id":"a01b0c452101d0f8f20d268b3a748509c82713fa","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-07 06:47:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"57767412fb3e2b61baa9b176aa2bce1e0d72e504","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 06:47:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cefb55efd0976e25efa16d16e6717f83bfb318e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 07:20:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4cc8f3141124b9788303d91a3375c75 : SUCCESS in 7m 02s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ed9f44f9ffc84463bd2762c0917acb73 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2bf113b918d41f68e634fe6c3f82ce0 : SUCCESS in 5m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5baec3a5790444d8a1b2ae31aae9df06 : FAILURE in 27m 42s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/42aa8e785c824dafa9f652ecc48038d4 : SUCCESS in 25m 49s","accounts_in_message":[],"_revision_number":1},{"id":"bfe5deeddbb5b69bb5583984236c19bab10e7211","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-07 07:31:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"016cab55b4c8d8a0bb75f1b92af84011882173cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 08:03:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23e66555323542ce971a992ae8287c0a : SUCCESS in 5m 31s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6768ee0a8a094cc888e950f6518ffedb : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/721e4a8fed9e4014ae44750e6b17c006 : SUCCESS in 5m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23ee8bec3c71441f88d029ed2458075d : SUCCESS in 9m 22s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/db955e16f83a46d4ab6f2714b6c68d38 : SUCCESS in 29m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/225e91a5c73c4b078c25403932ef02bc : SUCCESS in 28m 06s","accounts_in_message":[],"_revision_number":1},{"id":"827c8c13e394c2604c61ac6cbf9614bea5b7681b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 08:03:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c5dd287d9e6c5533d72c914dc0c801ce2e766d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 08:38:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d318e968098905a5da7490cbd50b4c259086e6b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 08:38:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79274d4cd0c1442c85f3ccb1535984e7 : SUCCESS in 4m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6d2d02394175428a990e7d263e74e964 : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5aeda2547e0f45929a95b89ae406e378 : SUCCESS in 5m 05s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/53d122325a154c9ab9ef24b9d8c0543e : SUCCESS in 28m 39s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/323eb094c42741e9a09707b85a14b2f0 : SUCCESS in 26m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1ab7539a580f32b4688ed9d5fd06e30dbf82ac96","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 08:40:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b91e869ba5f4c59bad2626135e8ac4b : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6709f56ec5bc1fae6106b12c789f851f241e93c6","author":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"date":"2019-11-07 14:42:03.000000000","message":"Patch Set 1:\n\nThe bug was only backported into the \"stable/rocky\" branch. For now, I need to backport this release note only into the \"stable/rocky\" branch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eab767f3ed1a6dfd4f50b214556142346c830c35","revisions":{"eab767f3ed1a6dfd4f50b214556142346c830c35":{"kind":"REWORK","_number":1,"created":"2019-11-05 14:37:51.000000000","uploader":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"ref":"refs/changes/19/693019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/19/693019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/19/693019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/19/693019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/19/693019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/19/693019/1"}}},"commit":{"parents":[{"commit":"d1b8dd1635aa45d3d644b356f883484997f31ca7","subject":"Fix error validate token when run cron trigger","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d1b8dd1635aa45d3d644b356f883484997f31ca7"}]}],"author":{"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","date":"2019-10-16 15:32:28.000000000","tz":-240},"committer":{"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","date":"2019-11-05 14:37:51.000000000","tz":0},"subject":"Add release note for fix error of cron trigger run","message":"Add release note for fix error of cron trigger run\n\nIt relates to the change I793fbfec03032d9ff7137c11becb6d1c18ec54bc,\ncommit 51b7dd0c6d.\n\nChange-Id: I2571246947ab5ae06731fc7d35de7a5954325a82\n(cherry picked from commit 948ceb150b66b86ba56b11821e1bb9cb8a7209cc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eab767f3ed1a6dfd4f50b214556142346c830c35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eab767f3ed1a6dfd4f50b214556142346c830c35"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
