)]}'
{"id":"openstack%2Fmistral~693189","triplet_id":"openstack%2Fmistral~master~I01bf47f7a83102a1a16b15bf0bbb4021707e11fe","project":"openstack/mistral","branch":"master","topic":"bug/1850843","hashtags":[],"change_id":"I01bf47f7a83102a1a16b15bf0bbb4021707e11fe","subject":"Mask sensitive data when logging action results","status":"MERGED","created":"2019-11-06 14:48:27.000000000","updated":"2019-11-18 09:21:54.000000000","submitted":"2019-11-15 17:37:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693189-1573839435938-de65efff","meta_rev_id":"4d744f4bb4c5c89b01020436eae75ffb2fe9f131","_number":693189,"virtual_id_number":693189,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-15 17:37:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2019-11-15 12:20:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-11-15 12:20:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-11-15 12:20:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":1,"date":"2019-11-15 12:20:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":1,"date":"2019-11-15 12:23:24.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-07 10:08:29.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-11-11 14:21:26.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2019-11-12 07:05:57.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-12 12:46:25.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-11-12 14:48:53.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-11-13 06:43:02.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-11-13 12:35:47.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-11-15 17:37:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78ec83af77b420045acdca8b6ea8ba1fcc390c75","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-06 14:48:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"786de2742d1eb2f82e46b7bdab0ea5b5041adfcf","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-06 15:02:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"25ce160a861c9ed53a950828afd32eb7a1764f5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 16:58:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6225ab6f28de49f9a85b6037e188d819 : FAILURE in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02cc9b71b6614041960aa96ab0e32737 : FAILURE in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85eb6a6ad4954f54b50885ed63071306 : FAILURE in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5cf8837eba3348b6a5278aa262eb6e0c : FAILURE in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be301ec8dbd347a8985ca2c7607ada2b : FAILURE in 3m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4dc1f8fbdbed41e4881567f6144ad963 : FAILURE in 3m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d4cb6177075451f89a8e8ef39bf6059 : FAILURE in 4m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/41c2e6e2dea749a3803cd64289562af5 : SUCCESS in 42m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/83f8cb42ee9a465aab60c963ebc22d3e : SUCCESS in 51m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3e58823500ac4de1bfa4a504b020f319 : SUCCESS in 44m 13s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3234ec9c067c44029e5be6c890801c30 : FAILURE in 4m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3a1b95f5cc8b40fc9aad473409632bbd : FAILURE in 4m 06s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1e437eb9e5d5477d8524ba26c4937415 : SUCCESS in 1h 32m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6e99f8fff5cb4de6b92832a91c97a1a6 : FAILURE in 2h 04m 34s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/330e52775fc2499a91340242ed36a7b0 : TIMED_OUT in 2h 04m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ec6d71ff5224839b7ce943cb328c16f : FAILURE in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"bd9b1cbfcb946a49259b7745c0b7e6ab111d4488","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-11-07 10:06:58.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6c70b489a27ec1262eb0186626fe82308fc3538f","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-07 10:07:33.000000000","message":"Topic set to bug/1850843","accounts_in_message":[],"_revision_number":2},{"id":"686a064c7f6042dfcb85dc66c271cb7c69d2f099","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-07 10:08:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"67cfa0de2e45390f5a35652cf1cf8ca85ba4bd63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 12:16:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bcdccb5781a40f484856aca8d03e7c2 : FAILURE in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ad82ccda421498ba6c9195d11af53f0 : FAILURE in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/771a081be4de4fefa186c754e678c33b : FAILURE in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb9aeeed28cc464397b27de6a92b2a8f : FAILURE in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7d37968e7c344978501d1411eae2ad5 : FAILURE in 3m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e291beb2744540cbb2f0fb5675db9233 : FAILURE in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/477d4bc7936f46d58cf08bd2fdbb8aca : FAILURE in 4m 40s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e35c4cbb99af4d5eb9ea8ecddb930c51 : SUCCESS in 48m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a2fc7beb27d44151971a50762b713610 : SUCCESS in 49m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bdcc9933c9ec4b97b059db0006130a25 : SUCCESS in 44m 40s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/627c0fe2ef2d4c469ed89fee83e5f851 : FAILURE in 3m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/909c2b08968e409491d0d9adda851b0c : FAILURE in 3m 57s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/dd9677839214407f9459622d5168eceb : SUCCESS in 1h 24m 33s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d371e7c307b24f1e8155f7fb92c2e61f : FAILURE in 1h 57m 05s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1648d352c8d94b8bafb447a98983c9a8 : TIMED_OUT in 2h 03m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01089bbcd8434a21ba416df5d797a7ec : FAILURE in 6m 09s","accounts_in_message":[],"_revision_number":2},{"id":"3cb4e321330eb0a5e6efd7f6fa49480eab160074","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-11 12:05:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a7a8059c67d9cfb659a179130998e8bb7ec13755","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-11 14:21:26.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9825245ad2699e3c094eb96ea06419b89e0bd487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 14:34:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6edbc6bc1ee04e629ad2bc60ac495c88 : FAILURE in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4524b2761c944e4d92bce83c711f32f9 : FAILURE in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed61205b5f0b4c9893e9d8a7209629f7 : FAILURE in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bfca9b47b2514c2d95fb751fb77bfa93 : FAILURE in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a40d0c303e94e669cf20c199f748af8 : FAILURE in 4m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d341492c7b45491e9100e1e933394504 : FAILURE in 4m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c747d936ae244e71bf8ae7fa2b1fb945 : FAILURE in 4m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e713d14a356b40339fca5dc354b100a6 : SUCCESS in 41m 53s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c84d76deba36471caec4563ed45b834c : SUCCESS in 43m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/effac64317254c81bf7f58cbb397366e : SUCCESS in 42m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6009fc03c4bb4b3aaaa85b01926b3ef9 : FAILURE in 4m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/47e1648bad614c2797a1c199cd76a9df : FAILURE in 4m 06s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e40b2c748f474f3e8701dbe6af350737 : SUCCESS in 1h 17m 20s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/83110dafe35c40a1a0afd862ea2645b4 : SUCCESS in 2h 23m 53s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/64efd1579e924739a9f3244b325ac799 : SUCCESS in 1h 10m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59c8023abf3d4d688e6b8b88b6576709 : FAILURE in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"52aeeca022c787f90551e65200427bb1a467bda3","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-11 15:05:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"57b8d7529a59f17442b6375ccc67f3bcac172935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 17:28:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59aa9de5d99349c1976ea28f8b2cc829 : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aadee67b5d4a42b0a88bfd12ef0a2dbc : SUCCESS in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/730466b595134caab76a54e3f69c57de : SUCCESS in 11m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea36ec61cb864ad7a8248ed7bd5f9b8b : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3330a021be441b1baf69eda7052c957 : SUCCESS in 6m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8f55330e1eb418dae91db143a16e388 : SUCCESS in 10m 01s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/eefb5a68595e47d392cbc8a303d4c9a0 : SUCCESS in 42m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c257a5574895441682cb91a97d030a1d : SUCCESS in 42m 54s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b595cd9f1652402da99dd477ac66322b : SUCCESS in 58m 28s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c910af0417b7445898428cd925f50e37 : SUCCESS in 29m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1f8790fd57524a229f976a6373d7cd06 : SUCCESS in 28m 00s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/8f1db2c7063d4f2fb960619e32c5ec8a : SUCCESS in 1h 09m 49s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/896823a7001a4d688987f008f4ff25d0 : SUCCESS in 2h 22m 07s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/82ff2bd443fd48458f49ec5196c49da7 : SUCCESS in 1h 19m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63bae3d938344f7088dd1c25af63b028 : SUCCESS in 11m 36s","accounts_in_message":[],"_revision_number":4},{"id":"8f29250b95626f9937bf28c5f7b6dea48e50b04a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-12 07:05:57.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"639e100d4c431149dc112c2843be2319669e1555","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-11-12 12:46:25.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6c902ee4728531a0c760321bd349fdbe6182f6dc","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-11-12 14:48:53.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"630881fe0137ccb49b06884b39e7f643bef66eff","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-13 11:14:02.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"189383d74c67f5ff434a0fce6b2377fbe968c103","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-13 11:15:12.000000000","message":"Patch Set 5: Workflow-1\n\nNeed to verify this manually still and add a release note","accounts_in_message":[],"_revision_number":5},{"id":"febf5214363ae5f167846af93cf4a9debf1a42f0","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-11-13 12:35:47.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d7bdaa77ede568e53d26f52097c4c94327172aaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 13:47:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9803f246bcfa4188be8cdfdca73337a2 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8295043079b4f50ace4c3d4b6640edc : SUCCESS in 8m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d85a8de2095426193abc582e8bc8efa : SUCCESS in 7m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/600e733e232c4bac9fa89a7ede96d242 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19d0c7c157494b63bd069415f640efad : SUCCESS in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d143045c2be045069e6aeb2f078cec78 : SUCCESS in 8m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e948eb0696944ed5bf7b03a19362fed2 : SUCCESS in 43m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/79077ff2ddfa4a21914fde5a7f54b6fc : SUCCESS in 42m 26s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fb619870a46d456d8eaf5f6d48243551 : SUCCESS in 44m 03s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/24e6cbde9c4b4fb690399a58ec75d2a4 : SUCCESS in 27m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e94d2f174e79417892c0d5f375d312fd : SUCCESS in 28m 07s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1142e0f2c3c24a429aaf3df1657efa24 : SUCCESS in 1h 28m 50s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/faec77d1bc3f4f2c92d1ca4d35ac65d0 : SUCCESS in 2h 31m 05s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6aa64e83a5e946eeb3d443d45dedb779 : SUCCESS in 1h 03m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d105f000b3840969b17a02377392e74 : SUCCESS in 10m 12s","accounts_in_message":[],"_revision_number":5},{"id":"d0441545d81fed0dc9bef9a06ecae319614be92b","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-15 12:20:05.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"603c103a07eab68e36374883482eb7722730055e","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-15 12:23:24.000000000","message":"Patch Set 6: Workflow+1\n\nManually verified this. Adding the +W since it has two +2s already.","accounts_in_message":[],"_revision_number":6},{"id":"2724f9b7237fc6dc8619385143663836ff4966e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 15:32:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8c4bd3fe00e452b95adef612c8facca : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/179dcf57642e4d0a9a852207e916b1dc : SUCCESS in 8m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b732bfcfe2d49f6926b1b0c0205223a : SUCCESS in 7m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d8a6432aaf5a4b68a1dba765b5b7ff78 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/232ceb0aab4f4ae38be77da75c1d42df : SUCCESS in 8m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be1947971a8040fc9541bc7989b7864b : SUCCESS in 9m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/13d2e6ef378747d1a776f8c20b0799fa : SUCCESS in 42m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/64c0121c2ea946a692e7ae4274042d51 : SUCCESS in 37m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e78545729dfc4ff8a2fa6b90ee757e84 : SUCCESS in 43m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2719b183ac204fcf8420910391781feb : SUCCESS in 28m 15s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/cc04d0334efc41a4a99bb17ab15efca5 : SUCCESS in 26m 45s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/8720dd2df8f74286bff16830d8e3217c : SUCCESS in 1h 41m 34s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b4f6c6b8df664fbcb7ced342a8a7d72c : SUCCESS in 3h 05m 51s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/174dd51afdb8433eb6dc3e131844ca12 : SUCCESS in 1h 06m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a564d6581214401ea871d3db9876ca88 : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":6},{"id":"0f16aaf771fbeb79fd29d48fed8576feec408ddd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 15:32:32.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"48cd311d1bf0b017272855fe6dade1da835974ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 17:37:15.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bada2b3720494c00b25e031525b44d05 : SUCCESS in 7m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e62e086d1985443caa99ab0f23bd9c7f : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/729c2dbc3f784777a9f084c39b809cfa : SUCCESS in 8m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/702775c986964696ba4503bd476f6709 : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbecd0b337514580a449215faade999c : SUCCESS in 7m 18s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e19010cf669b4b43a4591b3e8a042ff3 : SUCCESS in 45m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/eb4fb4f87f084134b31517aef1b77879 : SUCCESS in 39m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1f667f24ffe14ff79c6f5fb283ae30d5 : SUCCESS in 30m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ad3ed41ac1ab4dc5b466129aeab2eb8c : SUCCESS in 31m 32s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b23770fe18a24ea48965c63e492e65fd : SUCCESS in 1h 35m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6a76e7a5b85e4deb98a6261367f92a96 : SUCCESS in 52m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc7afaaffcc542da92a8a184fb3b9a18 : SUCCESS in 9m 12s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a2c13a055b6748a4bedaad728ced193c : SUCCESS in 2h 01m 57s","accounts_in_message":[],"_revision_number":6},{"id":"b4b4ae999ff4d3b31dea9d7495db36047284d182","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 17:37:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"15c368a29bc2d25672d9d1c0ba6569b6ff4617a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 17:39:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af62c2ebde6044bbb76f6c3138f74abd : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":6},{"id":"3a0b4d4c272adc7afd411cb1e293a3a066eaf43f","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-18 09:21:48.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 9726c1a5938ba9d04d3bb1dd8c3bea5d0c4b10b9","accounts_in_message":[],"_revision_number":6},{"id":"4d744f4bb4c5c89b01020436eae75ffb2fe9f131","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-18 09:21:54.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 6399f9b7e0133202f2d63896b44628668cdc212f","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a25c8fab881decff0e61deffe1895f1d99132f9f","revisions":{"1005b393d7ff3180ffe289795a19f4b002787879":{"kind":"REWORK","_number":1,"created":"2019-11-06 14:48:27.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/89/693189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/89/693189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/89/693189/1"}}},"commit":{"parents":[{"commit":"361debc8118a3c6f9ac3cf2eb850d4e17582ca99","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/361debc8118a3c6f9ac3cf2eb850d4e17582ca99"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-11-06 14:09:46.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-11-06 14:10:30.000000000","tz":0},"subject":"Mask sensitive data when logging action results","message":"Mask sensitive data when logging action results\n\nWhen passwords or other sensible data is returned by a Action, they can\nbe logged by Mistral. This change uses the password masking\nfunctionality used in mistral-lib and privided by oslo-utils.\n\nRelated-Bug: #1850843\nChange-Id: I01bf47f7a83102a1a16b15bf0bbb4021707e11fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1005b393d7ff3180ffe289795a19f4b002787879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1005b393d7ff3180ffe289795a19f4b002787879"}]},"branch":"refs/heads/master"},"716b67656e8faf4b61f30a1253ce67e654eee8b2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-07 10:06:58.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/89/693189/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/89/693189/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/89/693189/2"}}},"commit":{"parents":[{"commit":"361debc8118a3c6f9ac3cf2eb850d4e17582ca99","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/361debc8118a3c6f9ac3cf2eb850d4e17582ca99"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-11-06 14:09:46.000000000","tz":0},"committer":{"name":"Adriano Petrich","email":"apetrich@redhat.com","date":"2019-11-07 10:06:20.000000000","tz":0},"subject":"Mask sensitive data when logging action results","message":"Mask sensitive data when logging action results\n\nWhen passwords or other sensible data is returned by a Action, they can\nbe logged by Mistral. This change uses the password masking\nfunctionality used in mistral-lib and privided by oslo-utils.\n\nRelated-Bug: #1850843\nDepends-on: Iea7280546122199e4ebc603be473b470f2d817fe\nChange-Id: I01bf47f7a83102a1a16b15bf0bbb4021707e11fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/716b67656e8faf4b61f30a1253ce67e654eee8b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/716b67656e8faf4b61f30a1253ce67e654eee8b2"}]},"branch":"refs/heads/master"},"cefc03abd35d70f70a2ef466bc42b962394ae8e2":{"kind":"REWORK","_number":3,"created":"2019-11-11 12:05:10.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/89/693189/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/89/693189/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/89/693189/3"}}},"commit":{"parents":[{"commit":"fd24972befeb4988f09b20bd4a95dd1f2d9a8dbf","subject":"Fix task expression context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fd24972befeb4988f09b20bd4a95dd1f2d9a8dbf"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-11 12:04:53.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-11 12:04:53.000000000","tz":0},"subject":"Mask sensitive data when logging action results","message":"Mask sensitive data when logging action results\n\nWhen passwords or other sensible data is returned by a Action, they can\nbe logged by Mistral. This change uses the password masking\nfunctionality used in mistral-lib and privided by oslo-utils.\n\nThis function uses the standard cut_repr method in mistral-lib, which\nalso means the output is more standardised.\n\nRelated-Bug: #1850843\nChange-Id: I01bf47f7a83102a1a16b15bf0bbb4021707e11fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cefc03abd35d70f70a2ef466bc42b962394ae8e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cefc03abd35d70f70a2ef466bc42b962394ae8e2"}]},"branch":"refs/heads/master"},"ab6f202d8c210c1cd37d0e584b4182a9c66d3c39":{"kind":"REWORK","_number":4,"created":"2019-11-11 15:05:26.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/89/693189/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/89/693189/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/89/693189/4"}}},"commit":{"parents":[{"commit":"fd24972befeb4988f09b20bd4a95dd1f2d9a8dbf","subject":"Fix task expression context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fd24972befeb4988f09b20bd4a95dd1f2d9a8dbf"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-11 12:04:53.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-11 15:05:14.000000000","tz":0},"subject":"Mask sensitive data when logging action results","message":"Mask sensitive data when logging action results\n\nWhen passwords or other sensible data is returned by a Action, they can\nbe logged by Mistral. This change uses the password masking\nfunctionality used in mistral-lib and privided by oslo-utils.\n\nThis function uses the standard cut_repr method in mistral-lib, which\nalso means the output is more standardised.\n\nRelated-Bug: #1850843\nChange-Id: I01bf47f7a83102a1a16b15bf0bbb4021707e11fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ab6f202d8c210c1cd37d0e584b4182a9c66d3c39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ab6f202d8c210c1cd37d0e584b4182a9c66d3c39"}]},"branch":"refs/heads/master"},"6480ee85eeb3eafff1f3b743d206b158dd83283b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-11-13 11:14:02.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/89/693189/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/89/693189/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/89/693189/5"}}},"commit":{"parents":[{"commit":"66d1776f1b992d3b5f593240f4a9bfa75e572f76","subject":"Merge \"Extend capabilities to clean up old executions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66d1776f1b992d3b5f593240f4a9bfa75e572f76"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-11 12:04:53.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-11-13 11:14:02.000000000","tz":0},"subject":"Mask sensitive data when logging action results","message":"Mask sensitive data when logging action results\n\nWhen passwords or other sensible data is returned by a Action, they can\nbe logged by Mistral. This change uses the password masking\nfunctionality used in mistral-lib and privided by oslo-utils.\n\nThis function uses the standard cut_repr method in mistral-lib, which\nalso means the output is more standardised.\n\nRelated-Bug: #1850843\nChange-Id: I01bf47f7a83102a1a16b15bf0bbb4021707e11fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6480ee85eeb3eafff1f3b743d206b158dd83283b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6480ee85eeb3eafff1f3b743d206b158dd83283b"}]},"branch":"refs/heads/master"},"a25c8fab881decff0e61deffe1895f1d99132f9f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-11-15 12:20:05.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/89/693189/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/89/693189/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/89/693189/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/89/693189/6"}}},"commit":{"parents":[{"commit":"a3bb6f08e3585cf240a25cb7ff71b15c8f03cbc2","subject":"Merge \"Make sure minimum amqp is 2.5.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a3bb6f08e3585cf240a25cb7ff71b15c8f03cbc2"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-11 12:04:53.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-11-15 12:20:05.000000000","tz":0},"subject":"Mask sensitive data when logging action results","message":"Mask sensitive data when logging action results\n\nWhen passwords or other sensible data is returned by a Action, they can\nbe logged by Mistral. This change uses the password masking\nfunctionality used in mistral-lib and privided by oslo-utils.\n\nThis function uses the standard cut_repr method in mistral-lib, which\nalso means the output is more standardised.\n\nRelated-Bug: #1850843\nChange-Id: I01bf47f7a83102a1a16b15bf0bbb4021707e11fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a25c8fab881decff0e61deffe1895f1d99132f9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a25c8fab881decff0e61deffe1895f1d99132f9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
