)]}'
{"id":"openstack%2Fmistral~693669","triplet_id":"openstack%2Fmistral~master~If7a054bbf77f9ed761d8f3ac36b6d329544f5ff5","project":"openstack/mistral","branch":"master","topic":"refactor_join_rerun","hashtags":[],"change_id":"If7a054bbf77f9ed761d8f3ac36b6d329544f5ff5","subject":"Refactor rerun of joins","status":"MERGED","created":"2019-11-11 09:42:56.000000000","updated":"2019-11-12 09:34:33.000000000","submitted":"2019-11-12 09:32:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693669-1573551176680-7361ecdf","meta_rev_id":"e0b2317d2ca1726887999b9b6df6f041288d7f15","_number":693669,"virtual_id_number":693669,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2019-11-12 04:36:19.000000000","_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-12 09:32:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2019-11-12 07:56:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":2,"date":"2019-11-11 11:17:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":1,"date":"2019-11-12 07:56:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-11 11:17:08.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-11-12 07:56:38.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-11-12 09:32:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eeed3afb4f9ce9dd2c9137682a0750fdc3aa21f6","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-11 09:42:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5435414d35c7316d56f9493a5db725ed34940d0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-11 09:49:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94b9c594a8a2811cf9c4afb21cbe69dc5130bbe9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-11 10:10:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e727638d087fdaf44640b8785fa25dfd5059064c","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-11-11 11:17:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"89cb1747084317fa0320cf6a653001c445228143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 13:31:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/684ae6504cae4106a21bec77e3ec2864 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c566c52686f4c7c93f18e154504c12e : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f20274827ab1430daf23c17f12cb8daf : SUCCESS in 9m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65689e40cc834b5eaf4d3eb27a157cc7 : SUCCESS in 9m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/849ffa3ea54043528918f127b3abc035 : SUCCESS in 6m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed4557646f964d4984eb805d51ae31f1 : SUCCESS in 10m 58s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d7da86358c5f4c3681b2ea80ab2f4563 : SUCCESS in 39m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1e6a8627ef4446d08b72de74a57886c8 : SUCCESS in 48m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cf2c6b93b8da463ea7fb4a814b092e14 : SUCCESS in 41m 10s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1281ba9be71d4162a1eafa51ac21c27c : FAILURE in 33m 19s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/32ab1586b52a41329cd13353936b0e46 : SUCCESS in 33m 18s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/14ad449e38014e64856cfd5f4c5fc385 : SUCCESS in 1h 16m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e979c066dba4495da8d6bbc7c23eb2b6 : SUCCESS in 3h 05m 50s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8596c6b0744d4379bd5fb81ee5fb5615 : SUCCESS in 58m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbdd84973c7c41c6b2ffe7fa6692a3be : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":3},{"id":"ad06a4ea1fd85427bf11eb3e92aa7e4dd0d27f8b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-12 04:36:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ea7fbc581379f5303f6372d11580b10a96b8027f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 07:20:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f42dd5ccb403415e8849bc991a540d69 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a8d179760d1e4a85bb30e42d9cfd306c : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0893fe678bb4d9bb7f3fb18ef64c392 : SUCCESS in 7m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21e576d35fa24a2b96639c49c1367505 : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0c29c44762246b79a8693b713bf5dcc : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/932f6f505a0d4921a75b16a3f2322c74 : SUCCESS in 8m 16s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/459bc6a6792c452dac5e55e5344f00b3 : SUCCESS in 42m 43s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c1de175f8b7841b09324b67ead96e8dc : SUCCESS in 44m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6187d58a6cbc48da9a6b0d77aa413855 : SUCCESS in 40m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2a15c725c6dd47aabbaeda4f4ad48f4d : SUCCESS in 27m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7df8e452f09d4bf99c757a0e2a042061 : SUCCESS in 26m 46s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c40d695393004b368cb1dd87091e373e : SUCCESS in 1h 12m 58s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9dc0645ba61247349817eb5207e27c58 : SUCCESS in 2h 39m 17s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f96eb6d27edc48729d353c6ede3a8434 : SUCCESS in 1h 05m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55f351d4794f401fbfc1cbf4f94a5d59 : SUCCESS in 10m 28s","accounts_in_message":[],"_revision_number":3},{"id":"b09b0f8f74e5daed69da7cff1f46723afcff5876","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-11-12 07:56:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7699a04cc6245552006999a8a6433dec89155dc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 07:56:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a71b4504eb87869db8c345871835aab23751b33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 09:32:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"10cdcf894b31e1e6cca02d5b2a8399da7495e0ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 09:32:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c1108d34eb8443c9e090da077d6248a : SUCCESS in 9m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b7facb92bcee42fa8f7236df0fb7f13e : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d39e206e92a24452b54f32ab669738ab : SUCCESS in 8m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2d6845c9cb74ce19004d0ba95486f8f : SUCCESS in 14m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ce1c41e9ec84dec8b5671742e267abb : SUCCESS in 9m 14s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/34d23f9063bb468f8563084fe1ff6d37 : SUCCESS in 51m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/58191f3258fc47b6b2063b8221d584c3 : SUCCESS in 40m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/22c77456f0544df3bf643163be138c0d : SUCCESS in 28m 57s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ab9ff3074016490c93913925085e14f9 : SUCCESS in 26m 39s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e48cae18c1e54373a7ed6823110df668 : SUCCESS in 1h 27m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/66b276863ebf4bca805524b0728170a2 : SUCCESS in 44m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b63a259c49d1437f88dabc7a733782a0 : SUCCESS in 9m 14s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/edcdd69059fc40a2a954a222aea81daf : SUCCESS in 1h 34m 48s","accounts_in_message":[],"_revision_number":3},{"id":"68bfbf1a89586b0d08483d2183785c783338f7eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 09:34:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b8596e246ea45f6a7f15e86a297ebd8 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"59bf2509eb7737d7cee84ebdc26cad97e735c7f8","revisions":{"f2b4ef91dc2ca2d0b47260b8929a018c4cdbc59a":{"kind":"REWORK","_number":1,"created":"2019-11-11 09:42:56.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/69/693669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/693669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/693669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/693669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/693669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/693669/1"}}},"commit":{"parents":[{"commit":"bdbfb823010cefe94945ba118199d9ecccdbcab7","subject":"Reformat rerun logic for tasks with join","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bdbfb823010cefe94945ba118199d9ecccdbcab7"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-11 09:41:39.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-11 09:41:39.000000000","tz":420},"subject":"WIP: refactor rerun of joins","message":"WIP: refactor rerun of joins\n\nChange-Id: If7a054bbf77f9ed761d8f3ac36b6d329544f5ff5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f2b4ef91dc2ca2d0b47260b8929a018c4cdbc59a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f2b4ef91dc2ca2d0b47260b8929a018c4cdbc59a"}]},"branch":"refs/heads/master"},"08f8845dbc37efab1d3f92ebf7d4f9449fd8d8fa":{"kind":"REWORK","_number":2,"created":"2019-11-11 09:49:24.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/69/693669/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/693669/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/693669/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/693669/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/693669/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/693669/2"}}},"commit":{"parents":[{"commit":"bdbfb823010cefe94945ba118199d9ecccdbcab7","subject":"Reformat rerun logic for tasks with join","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bdbfb823010cefe94945ba118199d9ecccdbcab7"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-11 09:41:39.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-11 09:49:00.000000000","tz":420},"subject":"WIP: refactor rerun of joins","message":"WIP: refactor rerun of joins\n\nChange-Id: If7a054bbf77f9ed761d8f3ac36b6d329544f5ff5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/08f8845dbc37efab1d3f92ebf7d4f9449fd8d8fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/08f8845dbc37efab1d3f92ebf7d4f9449fd8d8fa"}]},"branch":"refs/heads/master"},"59bf2509eb7737d7cee84ebdc26cad97e735c7f8":{"kind":"REWORK","_number":3,"created":"2019-11-11 10:10:35.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/69/693669/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/693669/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/693669/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/693669/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/693669/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/693669/3"}}},"commit":{"parents":[{"commit":"bdbfb823010cefe94945ba118199d9ecccdbcab7","subject":"Reformat rerun logic for tasks with join","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bdbfb823010cefe94945ba118199d9ecccdbcab7"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-11 09:41:39.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-11 10:10:16.000000000","tz":420},"subject":"Refactor rerun of joins","message":"Refactor rerun of joins\n\n* This patch moves logic that schedules a task state refreshing\n  periodic job in case of rerun from the Task class to\n  task_handler.run_task() so that Task doesn\u0027t have to know any\n  language specific details and call task handler back. It is\n  more architecturally clean.\n\nChange-Id: If7a054bbf77f9ed761d8f3ac36b6d329544f5ff5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/59bf2509eb7737d7cee84ebdc26cad97e735c7f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/59bf2509eb7737d7cee84ebdc26cad97e735c7f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
