)]}'
{"id":"openstack%2Fmistral~693796","triplet_id":"openstack%2Fmistral~master~I9a97c40222e8dc4870c9b6a7c5f5e3c14f37bdd6","project":"openstack/mistral","branch":"master","topic":"refactor_action_hearbeat_mechanism","hashtags":[],"change_id":"I9a97c40222e8dc4870c9b6a7c5f5e3c14f37bdd6","subject":"Refactor action execution reporter","status":"MERGED","created":"2019-11-12 09:48:37.000000000","updated":"2019-11-18 04:59:49.000000000","submitted":"2019-11-18 04:49:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693796-1574052567698-ded4c5e8","meta_rev_id":"e3e46bdac3d41393c15276f4ec11228c87360337","_number":693796,"virtual_id_number":693796,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-18 04:49:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-11-15 16:19:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":1,"date":"2019-11-13 10:30:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":1,"date":"2019-11-13 13:58:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-11-15 10:20:56.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2019-11-15 10:20:56.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2019-11-15 10:20:56.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"date":"2019-11-15 10:20:56.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-11-18 02:43:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-13 10:30:35.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2019-11-13 13:58:58.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2019-11-15 10:20:56.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-11-15 10:20:56.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-11-15 10:20:56.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-11-15 10:20:56.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-11-15 16:19:05.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-11-18 04:49:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ef27c44d59f1cba3cea0ac03fcf69dc7e2f6a08","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-12 09:48:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7536b91c5661557730bcb58afca79308a34df00","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-12 10:47:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dd332fd2db62055c9c3544ae4cc7e0081f287ebf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 14:01:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6213bfdf79d4e7dbabea64aa7b8e4c0 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4071ad5b56994f95a764960adaf9ef33 : SUCCESS in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00debab5b6584874a8fdf7282efe5a28 : SUCCESS in 8m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3085470719d344e6940afd41cda70b29 : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe12ecf989af4b1cb0fcd5ff3e62918a : SUCCESS in 4m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb3d70160e41470ba98820362736c877 : SUCCESS in 8m 41s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9ab200589e004a0286ee3e7b5af7e217 : SUCCESS in 45m 07s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/523f4fbd721e423ea13ca84d4f111ff1 : SUCCESS in 45m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b933ba8dd9df4485986dec6e3afbf45d : SUCCESS in 45m 36s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1553e59325ee482dab11cc22d1835d4a : SUCCESS in 29m 45s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b230bb0fb028446493de787efd0e327d : SUCCESS in 27m 59s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/9f6373e5553248d3857b4b32c3c281b7 : SUCCESS in 1h 25m 09s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ef8d0076a13d4f3cb54d0cd9615b9642 : SUCCESS in 3h 03m 01s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4ea9f002fceb470498c39c7223e9b868 : SUCCESS in 1h 00m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0414893050b41a9b2141ebef3b5dc4a : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":2},{"id":"431aca34325790ca480960864afd2ffb925c525b","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-13 10:30:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4469aaa6d5081ded0367192414e2e53a443be630","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-11-13 13:58:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a661d6bb8c04daa33f2c67e9622b114b32e9040e","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-11-15 16:19:05.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)\n\nLGTM but I\u0027d like the functionality to stay in a class, though.","accounts_in_message":[],"_revision_number":2},{"id":"cb838f9e11e6b064e6fd9524919ec0b39d07e915","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-18 02:43:19.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d2613a8f6f99f3037b8903f011d6134b3be49f01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 02:43:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8a5e607661f96930ba6f7387662fab431cfa8592","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 04:49:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"09f9b06d6d4a7b269d1bbed0a24eb76c059ec9c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 04:49:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/843575fd60c6434ea9642a80fbc14c17 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bdb0cb2231ec44da88cc69d5bf380f7e : SUCCESS in 9m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7be1a34136140a4885e4aaf31d239a4 : SUCCESS in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee1bad6c58d842bc9d0107d4a9ecedf1 : SUCCESS in 9m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24a7cccda17e4fb5990956d896910bbc : SUCCESS in 6m 49s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d2abb34751d840aeb956ac7fcc569938 : SUCCESS in 38m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/543ceaaa12a54ad39323f9433e536b42 : SUCCESS in 39m 41s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3c61a609d87e425c90c064c6443cceb0 : SUCCESS in 28m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3d581414fb7e4f5f9705b2da80f7d0fd : SUCCESS in 27m 42s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/621c52f5053f490588fae3a97e148de1 : SUCCESS in 1h 17m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fe157a2825014ca4873ce7649234f6d8 : SUCCESS in 37m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/051f6698580b4d449fe7ae5255b13b50 : SUCCESS in 9m 05s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7bdbdc3d14dc45628617e738f5936060 : SUCCESS in 1h 32m 03s","accounts_in_message":[],"_revision_number":2},{"id":"aa744870dbbfd6a8690028a93fdacbdd8e32ba17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 04:50:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99efd712098e4c208d0c8559a695197a : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2},{"id":"31b99a031255a311f4d32829dd2d196a24735510","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-18 04:59:49.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 229da9aa1ffe040acc86311822c2a0b231095651","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0e758e16e1abdb2440e28d270457f7329b876708","revisions":{"221d6336fad7f75f2dfe2c72e1db0f00742de7af":{"kind":"REWORK","_number":1,"created":"2019-11-12 09:48:37.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/96/693796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/693796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/693796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/693796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/693796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/693796/1"}}},"commit":{"parents":[{"commit":"1c7e242975a23b6bdf03fa30f94e122df188973f","subject":"Merge \"Reformat rerun logic for tasks with join\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1c7e242975a23b6bdf03fa30f94e122df188973f"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-11 04:36:49.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-12 09:48:07.000000000","tz":420},"subject":"WIP: refactor action execution reporter","message":"WIP: refactor action execution reporter\n\nChange-Id: I9a97c40222e8dc4870c9b6a7c5f5e3c14f37bdd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/221d6336fad7f75f2dfe2c72e1db0f00742de7af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/221d6336fad7f75f2dfe2c72e1db0f00742de7af"}]},"branch":"refs/heads/master"},"0e758e16e1abdb2440e28d270457f7329b876708":{"kind":"REWORK","_number":2,"created":"2019-11-12 10:47:24.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/96/693796/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/96/693796/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/693796/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/693796/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/96/693796/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/96/693796/2"}}},"commit":{"parents":[{"commit":"1c7e242975a23b6bdf03fa30f94e122df188973f","subject":"Merge \"Reformat rerun logic for tasks with join\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1c7e242975a23b6bdf03fa30f94e122df188973f"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-11 04:36:49.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-12 10:42:59.000000000","tz":420},"subject":"Refactor action execution reporter","message":"Refactor action execution reporter\n\n* Moved away from using Oslo periodic tasks in the action execution\n  reporter since in this case they don\u0027t make the code more readable.\n  Also, now it is symmetric with other similar components like action\n  execution checker.\n* Refactored action execution checker w/o using classes since having\n  many instances of it doesn\u0027t make sense.\n* Small style changes\n\nChange-Id: I9a97c40222e8dc4870c9b6a7c5f5e3c14f37bdd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e758e16e1abdb2440e28d270457f7329b876708"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e758e16e1abdb2440e28d270457f7329b876708"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
