)]}'
{"id":"openstack%2Fmistral~694023","triplet_id":"openstack%2Fmistral~master~I4d0fdff54de9bee70aeaf10a4ef483ad7000840b","project":"openstack/mistral","branch":"master","topic":"bug/1852722","hashtags":[],"change_id":"I4d0fdff54de9bee70aeaf10a4ef483ad7000840b","subject":"Make action heartbeats work for all executor types","status":"MERGED","created":"2019-11-13 09:37:08.000000000","updated":"2019-11-18 10:54:21.000000000","submitted":"2019-11-18 04:49:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694023-1574052568723-81677534","meta_rev_id":"cab061b0e1e8252bf7f8b79d921a895fbc10214c","_number":694023,"virtual_id_number":694023,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-18 04:49:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-11-15 16:10:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":2,"date":"2019-11-15 15:28:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"date":"2019-11-15 10:20:47.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-11-15 16:10:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2019-11-15 10:20:47.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2019-11-15 10:20:47.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-15 10:20:47.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-11-15 10:20:47.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-11-15 10:20:47.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-11-15 15:28:47.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-11-15 16:10:02.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-11-18 04:49:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6bfd36532b0be82f002e2c185e583c448a65e222","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-13 09:37:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2490f0daf0f85cebf4b80a6293b4a162d6b0895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 12:36:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f7c1f0d69374c439b5730f81146b9c3 : FAILURE in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fbc3f6c584554b95a9d851c2d0e5dabd : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ccb8e101e8424480ba25b50b9eda7170 : SUCCESS in 7m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33476335538b44d7888166ada3a919b3 : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ce8784430674135b09fee92f4877573 : SUCCESS in 4m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba42bf8bb35645a182d45b1bc5356db0 : SUCCESS in 8m 40s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6f37caad746844bcb60dc6bfb7db1e6f : SUCCESS in 46m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/95c4bbd938024c94bf96aa73efb12ae7 : SUCCESS in 1h 06m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f7ee18905c0b4931b9587664d9862df3 : SUCCESS in 46m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1239cde2acf24ee8b8d59eedf3ced924 : FAILURE in 28m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/523a353100fb4b0cb72706a09f874d41 : FAILURE in 26m 38s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/25e28838a6ba431ebe4b1a7799e2835d : SUCCESS in 1h 31m 47s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/145b99187af24e54836f042d23cb50e1 : SUCCESS in 2h 55m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/219d3e2d33e84b2da22cfe731f826a8f : SUCCESS in 1h 05m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d044e8eaa9449c1a701ab94a1f63580 : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":1},{"id":"766b3636babeb2a154bca8b97328902431678f0c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-15 09:57:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4bf06266e876bb6cee9983544042341577b1838f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 12:55:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/421a00013a094f73a49833c2af612f00 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8a46ee1a2234d47ad50b98ac26e2589 : SUCCESS in 8m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb4db355fa6b4df183f4016558136cb3 : SUCCESS in 7m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47727965e99340559b98750e394330c0 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ad41259edc24fa5b06090ce6b1c4f23 : SUCCESS in 7m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c02fa6377224d43b914c05fd379bae7 : SUCCESS in 9m 12s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/65b7c1bc97054a0fa807a7e23b5ff05b : SUCCESS in 44m 40s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c433b57c69f74068b9b9ea30e6dca250 : SUCCESS in 40m 31s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/391f07eb48f04a0aaa0e4ab8baada6c4 : SUCCESS in 43m 28s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7548788555d04dc7bd78d8ad759b6aa0 : SUCCESS in 28m 39s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3f1ca78a748045398dc621bacff56e80 : SUCCESS in 27m 54s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a3950e23f8dd46a8aad115137b5a67d1 : SUCCESS in 1h 37m 58s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3f2e04cd232a458882b6b04736122b95 : SUCCESS in 2h 44m 08s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bdd159febbd643baba11fd2cf0f34ed7 : SUCCESS in 1h 13m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/340a2ebf2a604ee18a290b594d2e05ae : SUCCESS in 9m 47s","accounts_in_message":[],"_revision_number":2},{"id":"e6bb086eb48767009e644a35b19fcea05c91e613","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-11-15 15:28:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a494813171b427637e215b747fc474278c08abfb","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-11-15 16:10:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3781ff64a1f55c74d86b39858dd31ba78783ab41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 02:43:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a82a5ab9826502f433597cbe430fd7fb411083f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 04:49:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9671f2bd85329c79a44c5685d2c98b1b4db76233","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 04:49:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6928537d494f4e15898db9f628a11257 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9faf601331f9449191ad06bcc4c310a3 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/629cfc2a711d4c68ae326305782ead76 : SUCCESS in 8m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c55030b71a748e7a376709460565288 : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f24e18a8b7024f7fa6d11f99fcdf53e4 : SUCCESS in 7m 01s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/70b69d664bd44eb990ca902b4780aa00 : SUCCESS in 36m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bb39e0b66b89411e90eb41e9378b1682 : SUCCESS in 36m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3af9683a1f4249bc8228612dd4b56314 : SUCCESS in 28m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2f91d1cf84714d9a9986b2c52b8d2151 : SUCCESS in 27m 23s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/2baf19756832400fbd6b5015fefae754 : SUCCESS in 1h 12m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/76991d4757da4e7d876fc515cc8a3d6b : SUCCESS in 37m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e023d586556402cb5988d82f279b391 : SUCCESS in 9m 36s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d6fb3aa2ecc24567abcc69a2c493563e : SUCCESS in 1h 38m 52s","accounts_in_message":[],"_revision_number":2},{"id":"cb9f720ce7e4715bcb60c085bccf1cf82f2c6278","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 04:51:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dec0309b57c40cc802eb5c55103df76 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2},{"id":"09c662c3b2f1b85448c4611b559cb2db61de116f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-18 10:54:21.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit d878d5409f3bd74f6137490d7fac555d4975fee9","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7ec4f26744ac17151adae4c06cd8a17b71f409a7","revisions":{"bd7f89eb479c54056339d9547fce524795d46568":{"kind":"REWORK","_number":1,"created":"2019-11-13 09:37:08.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/23/694023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/23/694023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/694023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/694023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/694023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/23/694023/1"}}},"commit":{"parents":[{"commit":"0e758e16e1abdb2440e28d270457f7329b876708","subject":"Refactor action execution reporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e758e16e1abdb2440e28d270457f7329b876708"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-13 09:34:44.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-13 09:34:44.000000000","tz":420},"subject":"WIP: Make action heartbeats work for all executor types","message":"WIP: Make action heartbeats work for all executor types\n\nChange-Id: I4d0fdff54de9bee70aeaf10a4ef483ad7000840b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bd7f89eb479c54056339d9547fce524795d46568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bd7f89eb479c54056339d9547fce524795d46568"}]},"branch":"refs/heads/master"},"7ec4f26744ac17151adae4c06cd8a17b71f409a7":{"kind":"REWORK","_number":2,"created":"2019-11-15 09:57:11.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/23/694023/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/23/694023/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/694023/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/694023/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/694023/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/23/694023/2"}}},"commit":{"parents":[{"commit":"0e758e16e1abdb2440e28d270457f7329b876708","subject":"Refactor action execution reporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e758e16e1abdb2440e28d270457f7329b876708"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-13 09:34:44.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-15 09:44:40.000000000","tz":420},"subject":"Make action heartbeats work for all executor types","message":"Make action heartbeats work for all executor types\n\n* Previously action hearbeats didn\u0027t work in case of using local\n  executors because the component responsible for sending heartbeats\n  was started by the executor RPC server which doesn\u0027t make sense to\n  initialize for a local executor. This patch refactors the code\n  so that now heartbeats get sent for any type of executors. For\n  local executors it is also useful because a cluster node that\n  runs an engine and a local executor may also crash. With this\n  change, remaining cluster nodes will be able to understand that\n  the action will never complete and one of them will time it out.\n  If all is fine with the node where the local executor is running\n  then heartbeats will be sent normally and the action won\u0027t time\n  out. Before this change, in case of local executors a long running\n  action would always time out after a configured amount of time\n  (by default, 60 mins) just because local executors never sent\n  heartbeats.\n* Made a lot of renamings to clearly see what component is\n  responsible for.\n* Wrote the tests that check the heartbeat sender, both positive\n  and negative scenarios for local and remote executor types.\n\nCloses-Bug: #1852722\n\nChange-Id: I4d0fdff54de9bee70aeaf10a4ef483ad7000840b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7ec4f26744ac17151adae4c06cd8a17b71f409a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7ec4f26744ac17151adae4c06cd8a17b71f409a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
