)]}'
{"id":"openstack%2Fmistral~694233","triplet_id":"openstack%2Fmistral~master~I7d82be4c9174676539d9ea4132504a47ff6e5e0b","project":"openstack/mistral","branch":"master","topic":"eyalb/requirments","hashtags":[],"change_id":"I7d82be4c9174676539d9ea4132504a47ff6e5e0b","subject":"Make sure minimum amqp is 2.5.2","status":"MERGED","created":"2019-11-14 10:29:24.000000000","updated":"2019-11-18 09:04:29.000000000","submitted":"2019-11-15 09:45:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694233-1573811159531-a3a83055","meta_rev_id":"9cf544394873265aa3b5446873001ae3862b8769","_number":694233,"virtual_id_number":694233,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-15 09:45:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":2,"date":"2019-11-15 08:40:06.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-11-15 04:28:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":1,"date":"2019-11-15 08:40:06.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-15 04:28:45.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-15 08:40:06.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-11-15 09:45:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31adbfdb241df639b2d7b16554a1c2904bbddc13","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-14 10:29:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e6162211295a6dd5af6801729cb7ef945377fbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 11:42:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b9170632cde4b0882d4c40c76af40e3 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac25787e5d1a4cd1b7c5b69007e8a22a : SUCCESS in 9m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8b8b8c0da0a4586b5c00239570feb10 : SUCCESS in 10m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3cad902bd7154dc89d9a26c8926de6bd : SUCCESS in 12m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e30a808484bf4847b0b34c5766d907d0 : SUCCESS in 5m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/91b72b0702f840b7869bc1d10bea1a94 : FAILURE in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df3b2bb304104dd19ab77a16a61132ae : SUCCESS in 10m 34s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0b1bb56391ad460eacab92bdeb5c67d1 : SUCCESS in 47m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c18e062cb5f3402493a985b1c51e8b40 : SUCCESS in 1h 05m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ec953fc0709e4146ade043ce452c1fb8 : SUCCESS in 56m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/aae684b2eb9e46bc8eb2acf3f35a0338 : SUCCESS in 28m 07s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/df35b5e96a754f9b992fbf4e02e70537 : SUCCESS in 28m 26s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0b00036b30574a3e985ebe7fe9fbdfca : SUCCESS in 1h 04m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8f3f12bf8984bb5b0aaddefa28200dd : SUCCESS in 9m 52s","accounts_in_message":[],"_revision_number":1},{"id":"17ad08622ef1d0a11d5bda10628efe0d2793f154","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-14 11:43:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2c9b1fdf10a87ce48a91201a08a65519515e0df6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 12:51:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/593cec8655984a87bd0a1094ade79a2e : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4feb810f99ed4720ad7aa4e8cae8a823 : SUCCESS in 11m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/605c324b42594b7b87b0f4f2f0f40f19 : SUCCESS in 8m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3cf4d53bcda34f7ebdd8b07e7e98a773 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eb4e7facd634f2281c512511464453f : SUCCESS in 5m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c9fd5f0b8ec4419f803c3f9edf15b80d : SUCCESS in 2m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8840b612e8c44e793d478230869278b : SUCCESS in 10m 17s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f9b575e1dab64f999fb9c595a927ed8d : SUCCESS in 52m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0a503cf265a94f4f8fbf42d9174f1895 : SUCCESS in 53m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c5c97c3e6f4b4113aa59e80bc27a7270 : SUCCESS in 44m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3dd893b1ce2f4d5680c4e988ef8dc1d0 : SUCCESS in 32m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1eab0e8c7fdf492cb8999d359900803e : SUCCESS in 28m 08s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ae41fab6e70d4d5d9183c86bedf7fb29 : SUCCESS in 1h 01m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8a23b5aa0a0401d8fd5e1e06e00cecb : SUCCESS in 11m 37s","accounts_in_message":[],"_revision_number":2},{"id":"88b2174df88e08d3d7a74159702c18809bd79366","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-15 04:28:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"045fdb2635d17bce3f3abbcd48308a630810538d","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-15 08:40:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"93d5964f54a24c3fb5d3ab631e9eba68e56b412e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 08:40:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6987cbf7a4266b349d5b0a71d8f8fe788a4b7e15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 09:45:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a174a68189eb0486ec1b7ec26102dbe7e6a5bad7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 09:45:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba65bd291976491589beb6fdeac3a5e8 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb948709f92c4919a71081604c8c8c8f : SUCCESS in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9caf3af6c6c42dcbb104a9e837bdab3 : SUCCESS in 11m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dfa433d12d454df1a80c77c5fd664bd2 : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dfea564f33a44d5acff09a3286811cb : SUCCESS in 8m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7e6d8977dc0a49e19f2575af46e4e30d : SUCCESS in 4m 41s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a839aabe621340768a7c562a486b54fd : SUCCESS in 57m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/abd3aa9057d749c0a989124c6eb632fa : SUCCESS in 48m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c4d54f1b014f43da96c71ff1a345c091 : SUCCESS in 28m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/535b64049e09492baa7ffefb61014b34 : SUCCESS in 27m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/81db77e519884544b218f01ffbb74edd : SUCCESS in 59m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b92610cc57c342399549373c3da28b0b : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":2},{"id":"ebd3b86bde6e38d477906ab52654a86a65f1a6c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 09:48:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d59014ae0784a68a69b1f2ce4b954b9 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":2},{"id":"bebaa8f3c76e921a02675b288be41242ccd79962","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-18 09:04:29.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 66f7952692ce8c85f4014c7ba96ccc862558f99e","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"80ff52284e6db62e914ff05cc1b1b44d9d126583","revisions":{"2979dfa078f87d9bee3a9bae8f9c3785dac192fe":{"kind":"REWORK","_number":1,"created":"2019-11-14 10:29:24.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/33/694233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/33/694233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/694233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/694233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/694233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/33/694233/1"}}},"commit":{"parents":[{"commit":"66d1776f1b992d3b5f593240f4a9bfa75e572f76","subject":"Merge \"Extend capabilities to clean up old executions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66d1776f1b992d3b5f593240f4a9bfa75e572f76"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-14 09:45:04.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-14 09:45:04.000000000","tz":120},"subject":"Make sure minimum amqp is 2.5.2","message":"Make sure minimum amqp is 2.5.2\n\namqp fixed a bug in 2.5.2 that is needed\n\nChange-Id: I7d82be4c9174676539d9ea4132504a47ff6e5e0b\nCloses-bug: #1852058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2979dfa078f87d9bee3a9bae8f9c3785dac192fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2979dfa078f87d9bee3a9bae8f9c3785dac192fe"}]},"branch":"refs/heads/master"},"80ff52284e6db62e914ff05cc1b1b44d9d126583":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-14 11:43:48.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/33/694233/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/33/694233/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/694233/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/694233/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/694233/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/33/694233/2"}}},"commit":{"parents":[{"commit":"66d1776f1b992d3b5f593240f4a9bfa75e572f76","subject":"Merge \"Extend capabilities to clean up old executions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66d1776f1b992d3b5f593240f4a9bfa75e572f76"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-14 09:45:04.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-14 11:43:40.000000000","tz":0},"subject":"Make sure minimum amqp is 2.5.2","message":"Make sure minimum amqp is 2.5.2\n\namqp fixed a bug in 2.5.2 that is needed\n\nDepends-On: https://review.opendev.org/#/c/694248/\nCloses-bug: #1852058\n\nChange-Id: I7d82be4c9174676539d9ea4132504a47ff6e5e0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/80ff52284e6db62e914ff05cc1b1b44d9d126583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/80ff52284e6db62e914ff05cc1b1b44d9d126583"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
