)]}'
{"id":"openstack%2Fmistral~694556","triplet_id":"openstack%2Fmistral~master~Ica51b3b36f3784b4b04731406b6bf8320d5de4b8","project":"openstack/mistral","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Ica51b3b36f3784b4b04731406b6bf8320d5de4b8","subject":"[ussuri][goal] Drop python 2.7 support and testing","status":"MERGED","created":"2019-11-15 14:46:28.000000000","updated":"2019-11-18 06:36:57.000000000","submitted":"2019-11-18 06:34:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694556-1574058867072-a4121484","meta_rev_id":"b5bda4e5ad636a981bf4b7ac3711475323cf0d00","_number":694556,"virtual_id_number":694556,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-18 06:34:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-11-18 04:36:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-17 13:13:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-11-18 04:36:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-17 13:13:08.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-11-18 04:36:52.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-18 06:34:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"323fc2b10b75343e582f872e6b5c9c01eceab750","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-15 14:46:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bc2e2f4256d22045f53a937b9c085bff7b0c41d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 17:32:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5f58aef744ad4cc4a8dab0c7ab05f519 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1efd7e1751ad44859bcff79884b476c1 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a159ecbc359345eb9bacd766e5a7f8aa : SUCCESS in 8m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b61c9426a42438f9f32f200dbe72b5b : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2d20a8beb9a4505a283d628a8a85669 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/626c157f15a9454c861c7b0809005888 : SUCCESS in 4m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2c3d33a2b894d2099b5b7a8616751a4 : SUCCESS in 9m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1f4f80fc1c484f57ad2c4b55a876dd6b : SUCCESS in 45m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3a37a5a0bbc04a3e89f0f35999a1114c : SUCCESS in 49m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c07a5e7e3cda41f796e2ed4b8cb34cd1 : FAILURE in 51m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e027e61faa6d4308877011dfb5300347 : SUCCESS in 28m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e712a54984e7471c8bbe1268efe47d0d : SUCCESS in 28m 33s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e86257ed979444c88588bf22bd6cf655 : SUCCESS in 1h 32m 46s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/241b3f9f0c924e01879b1eaf10786195 : SUCCESS in 2h 40m 19s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/86a4774576a24ccca4057543eef8f98d : SUCCESS in 1h 03m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d39b3d42ce1d472ebbdbfcc2ee2c0b41 : SUCCESS in 11m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b3f0642bad024bcb58e5de4fe6534e559c34dc8a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-16 14:08:08.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"231e1473615a58b7f3728bdcb2e3de4506361537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 16:58:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b29749fa8fc54f9a80266739f5f08912 : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4768c2e46cf2482cb805bc44c0bcfb73 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/695741057b084dae92bc2e6fcd89ffed : SUCCESS in 8m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fc6c32f2c4f4dfc983655b5e56d12b7 : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1790b45c6b0943b3ab441a7cfe47772b : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cfc853c00da4f158a2b59f7dd204206 : SUCCESS in 4m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d6cfdc0b3a64df3a1f5f9b6ca28d953 : SUCCESS in 8m 38s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3e10d0a1f1564036b6c2d6f7de5c646f : SUCCESS in 41m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a1182b5d5314481ca44bd3045802dfe3 : SUCCESS in 46m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f0e665f88e9b4fd9b3b13fac080751bf : SUCCESS in 39m 59s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a51e27728c2d4f5e81eca7d709259ac3 : SUCCESS in 27m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/dfa4d83d9a2d4840a780e946b86a490e : SUCCESS in 26m 48s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/4d092aae21114279aa9c88287d9b23dc : SUCCESS in 1h 21m 58s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9690d69f790f49f48bab2badfa6b016a : SUCCESS in 2h 46m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/414f573ad3d84d7aa92041532d92d16c : SUCCESS in 57m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2370cf2422914cef8f3ec7bef22b08d5 : SUCCESS in 9m 40s","accounts_in_message":[],"_revision_number":2},{"id":"4a77d07a404640cb8b684a8ee511926835e7985a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-17 13:13:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ae4c377c6ba94995573b78d17e02dbd78e5ac293","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-18 04:36:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0e864e3276fae85582c8bc7c2b482a63a29ac802","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 04:37:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1ec017462b01595868af9128fe278c76a4079372","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 06:34:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1a36435260284ea8b0eb382692c4e3e4 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26ffe96ffc4a496ea6b00998259898ac : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a733c8a881b4ec49dbe973a53c6d6cb : SUCCESS in 7m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9cee06183c4472e9a6e516a0ebb5f7c : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f6b15f58c704ae8b82de4232c8e56c8 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68793896c05e41dbb95ea93545a71acb : SUCCESS in 3m 54s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3b32d05c6c5a48e6a653068ef96b2bea : SUCCESS in 42m 53s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8e13be72f0dd479e95503200fc9a9b95 : SUCCESS in 44m 26s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6a00b9bbc11a4cc08e88cb8820c0a830 : SUCCESS in 28m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b70ba0ff4d234e9598914c43624ae31b : SUCCESS in 26m 57s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/24842b6f648042f28c1287055cec4076 : SUCCESS in 1h 28m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/65be9d4566f748d095fa105fb4f1d259 : SUCCESS in 42m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a47afed0a714982858cc3df89e93cde : SUCCESS in 9m 13s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4f432f952657417489c137b73c878e23 : SUCCESS in 1h 54m 14s","accounts_in_message":[],"_revision_number":2},{"id":"7705beca967ed5e0c550ec8a1c497109e3d4f99a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 06:34:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"35dd11d6b6a81971a67c4c4a9a93a61890d007b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 06:36:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f149bb67cef46589898777c29793c4a : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/626b9fb6f0934ecea57c2e27b846708f : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fa2042a8213a07c21dcec500e5a290996346afd0","revisions":{"0c8f84b96a6bfd523d4743c599658856fb5306d9":{"kind":"REWORK","_number":1,"created":"2019-11-15 14:46:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/56/694556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/694556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/694556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/694556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/694556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/694556/1"}}},"commit":{"parents":[{"commit":"a3bb6f08e3585cf240a25cb7ff71b15c8f03cbc2","subject":"Merge \"Make sure minimum amqp is 2.5.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a3bb6f08e3585cf240a25cb7ff71b15c8f03cbc2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-15 14:44:36.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-15 14:44:36.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nMistral is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Ica51b3b36f3784b4b04731406b6bf8320d5de4b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0c8f84b96a6bfd523d4743c599658856fb5306d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0c8f84b96a6bfd523d4743c599658856fb5306d9"}]},"branch":"refs/heads/master"},"fa2042a8213a07c21dcec500e5a290996346afd0":{"kind":"REWORK","_number":2,"created":"2019-11-16 14:08:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/56/694556/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/694556/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/694556/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/694556/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/694556/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/694556/2"}}},"commit":{"parents":[{"commit":"a3bb6f08e3585cf240a25cb7ff71b15c8f03cbc2","subject":"Merge \"Make sure minimum amqp is 2.5.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a3bb6f08e3585cf240a25cb7ff71b15c8f03cbc2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-15 14:44:36.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-16 14:07:57.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nMistral is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Ica51b3b36f3784b4b04731406b6bf8320d5de4b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fa2042a8213a07c21dcec500e5a290996346afd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fa2042a8213a07c21dcec500e5a290996346afd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
