)]}'
{"id":"openstack%2Fmistral~694728","triplet_id":"openstack%2Fmistral~stable%2Ftrain~I7d82be4c9174676539d9ea4132504a47ff6e5e0b","project":"openstack/mistral","branch":"stable/train","topic":"eyalb/requirments-stable/train","hashtags":[],"change_id":"I7d82be4c9174676539d9ea4132504a47ff6e5e0b","subject":"Make sure minimum amqp is 2.5.2","status":"MERGED","created":"2019-11-18 09:04:29.000000000","updated":"2019-11-22 09:13:04.000000000","submitted":"2019-11-22 09:11:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694728-1574413911405-c4d0231e","meta_rev_id":"36130e928fbfbb0302228ae3ccacfc64e32f1aac","_number":694728,"virtual_id_number":694728,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-22 09:11:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"all":[{"value":1,"date":"2019-11-19 13:31:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2019-11-22 08:08:38.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2019-11-18 12:57:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":2,"date":"2019-11-22 08:07:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":1,"date":"2019-11-22 08:09:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 12:57:52.000000000","updated_by":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"reviewer":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"state":"REVIEWER"},{"updated":"2019-11-19 13:31:00.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2019-11-22 08:08:38.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-11-22 08:09:47.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-22 09:11:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9fe0075120c2109c1210438d0b7749e047e82237","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-18 09:04:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c7c1dfc589770aa1e6b9ece0b547072a3c294e5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 10:16:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aec700ae9b64544bf0c33939ed1fa7b : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88f22a6ffdf64d34b824a11282ca0d5f : SUCCESS in 11m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d362acad938499fb3a0db0dee9f9d76 : SUCCESS in 8m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/90956fc9b2724be08ca143f346e40003 : SUCCESS in 9m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5bb465322ae417a96d3ffb350f69adb : SUCCESS in 5m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c685873ea2b544379ab9e95c78da4339 : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ca4b557310f4015b4df3921237a77f9 : SUCCESS in 9m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ec6118aff4454df98818cffcee1b2594 : SUCCESS in 45m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f50136f85bb742c288cd501d144c378d : SUCCESS in 41m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ebca8241505f44628fc873982d2f4da9 : SUCCESS in 41m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9e1ae3fd0c7b4e968b80e70cfe3e9da0 : SUCCESS in 28m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/73c76f090fdf4488ae24d1e14b31fbd0 : SUCCESS in 26m 39s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/419c258589ba45aea8f21436dc3d5b84 : SUCCESS in 1h 06m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/934a180ffc8e4df4840f8478c05194a0 : SUCCESS in 11m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9bd4144bc8117ac5e680d50fbef97e7f1afedd7b","author":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"date":"2019-11-18 12:57:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bcf6d70f903ba03c83a25752a62a711467d1586f","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-19 13:31:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"823fc4ca421f3b6e6e951aa196484e9326f05933","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-22 08:07:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"610cfbb6100c615bc3b941170831f90ff5fc701f","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-11-22 08:08:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6cddaa28969a760e5fdb0ab4cdb2dd941ccd07f5","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-22 08:09:47.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d87b994e2865555bc54ad92b0fcd9722a3292ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 08:10:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02e48e6b0bbb24580666196b64ed90eb72c4d2be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 09:11:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"32a023e9933b84638f57cc35107e4b7177bfcfd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 09:11:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/919c1566afd04c0cb721ac78437bb440 : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df19f5cbe9f544ad886e768e231cc8d0 : SUCCESS in 9m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b3c1c7fc7194f4f993aa7e23a3234e8 : SUCCESS in 9m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa7249018d854c64a2d1c39fef0220dc : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c81d634fd1a749e7ae940711d4e733a5 : SUCCESS in 5m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/58789af0cf0a4a5bb3ccc1fd5fa758b7 : SUCCESS in 2m 15s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1acd2fb425404625a6dcde61ed506246 : SUCCESS in 45m 29s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fc12d3462f614093bb3763967fd16262 : SUCCESS in 44m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6c30f85ac4514d1ea179c22c2bac4ad8 : SUCCESS in 30m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f1c0b38722574a9d9c14e21654957210 : SUCCESS in 29m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d5a894503f984c929870ddc35419a77b : SUCCESS in 32m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a98aa595f9a422d9cbb3744bec3e100 : SUCCESS in 9m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ee7a7838ecb75ff2d3e2a1375e586ab56179a0b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 09:13:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71027be525854fb192e361ba52cf3be3 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66f7952692ce8c85f4014c7ba96ccc862558f99e","revisions":{"66f7952692ce8c85f4014c7ba96ccc862558f99e":{"kind":"REWORK","_number":1,"created":"2019-11-18 09:04:29.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/28/694728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/28/694728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/694728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/694728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/28/694728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/28/694728/1"}}},"commit":{"parents":[{"commit":"090a255a58aa1efa80ea98203509440b3da18446","subject":"Evaluate input expression should check the in_context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/090a255a58aa1efa80ea98203509440b3da18446"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-14 09:45:04.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-18 09:04:29.000000000","tz":0},"subject":"Make sure minimum amqp is 2.5.2","message":"Make sure minimum amqp is 2.5.2\n\namqp fixed a bug in 2.5.2 that is needed\n\nDepends-On: https://review.opendev.org/#/c/694248/\nCloses-bug: #1852058\n\nChange-Id: I7d82be4c9174676539d9ea4132504a47ff6e5e0b\n(cherry picked from commit 80ff52284e6db62e914ff05cc1b1b44d9d126583)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66f7952692ce8c85f4014c7ba96ccc862558f99e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66f7952692ce8c85f4014c7ba96ccc862558f99e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
