)]}'
{"id":"openstack%2Fmistral~694731","triplet_id":"openstack%2Fmistral~stable%2Ftrain~I01bf47f7a83102a1a16b15bf0bbb4021707e11fe","project":"openstack/mistral","branch":"stable/train","topic":"bug/1850843","hashtags":[],"change_id":"I01bf47f7a83102a1a16b15bf0bbb4021707e11fe","subject":"Mask sensitive data when logging action results","status":"MERGED","created":"2019-11-18 09:21:48.000000000","updated":"2019-11-18 14:05:50.000000000","submitted":"2019-11-18 14:04:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694731-1574085845464-a4121484","meta_rev_id":"d157f60b28b34870798be235a128497f2c8c921f","_number":694731,"virtual_id_number":694731,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-18 14:04:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-11-18 10:59:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-11-18 10:59:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 10:59:29.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-18 14:04:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0c0cf46895cefd8e4dea4dd191bbca3cc374d06","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-18 09:21:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a09bed4715f7ca1944c0fc2a073964a7b0da8660","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-18 09:21:58.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 077e5eb65bb448055fa4ad0900c59f651963d313","accounts_in_message":[],"_revision_number":1},{"id":"1583a9d310895139ec35cbfc3d6bff291dc368ca","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-18 09:23:00.000000000","message":"Topic changed from bug/1850843-stable/train to bug/1850843","accounts_in_message":[],"_revision_number":1},{"id":"3625bfec47c2460476eba0717acad1ef51cd23b7","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-18 10:59:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f488a491cbb9f3faad5d6dd376316b18180365c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 12:30:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db78efc14d5c488f95c8f3090bcfdd5e : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ee734e8171e4eb981bb8f04d53d1ffb : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e12dcc953cb421b9a08cecf1ebcba5a : SUCCESS in 8m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a506b5b5709e462cac3f1b69aa817a4f : SUCCESS in 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32e113bb659c4310a289bcd609c22837 : SUCCESS in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e1a7e6854ae4b71b46d353ce7769298 : SUCCESS in 11m 10s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/234f991f9b0a497887c6f39e456d466b : SUCCESS in 45m 42s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/37b826cb305e47ad95acc4b5cf98ab47 : SUCCESS in 50m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0431a4e55aef484ab199f84c3acbea16 : SUCCESS in 42m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/df2cb1baffb04e46bfc57cb02c614c98 : SUCCESS in 31m 20s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9828ee5131764ab2a291a0f988723bbd : SUCCESS in 28m 44s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/765143a1c4c54998811df874766fc8ad : SUCCESS in 1h 36m 37s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e9195bcc55db41be8d520bfcc342a3f8 : SUCCESS in 2h 46m 16s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8c38c4f1e9774c1f81b225b12b5f650f : SUCCESS in 1h 13m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d77cfd1db080432cb65e6a279fea2f7b : SUCCESS in 12m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ee8ec7744d17f32b509975d695e10586dac824e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 12:30:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fb7ec9299194f3c9282bcb3452f87c4e5b2dfec3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 14:04:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8438be2349b34bcd9a018c8a98b7fe51 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e0769b9b0abd40d899bd07d7a603c408 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88304213af1348748cde728824a85dfd : SUCCESS in 8m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dfe58d46ae2149ada82e9f2d96be056a : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b5bb77eadf4ef4b4231492a5e34ac9 : SUCCESS in 5m 03s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/76d1d1aa66f143b99d44b28891151a5d : SUCCESS in 37m 42s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a7be0bdd8276450b8fc3e5237ee8376f : SUCCESS in 38m 10s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5c1e42c45f644be388d3287ca882468e : SUCCESS in 28m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e94b18f02df84e53aae6472637491b16 : SUCCESS in 26m 45s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/7640cb3a1ed140b8b555b368ccdc6fc2 : SUCCESS in 1h 21m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7bed731e28ba41d4b0a1622e3de50c00 : SUCCESS in 40m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28bac212742b4a6aa00d1ecd05add14f : SUCCESS in 10m 12s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e162bb1a52bd41a389e60546ce8fdba8 : SUCCESS in 1h 31m 41s","accounts_in_message":[],"_revision_number":1},{"id":"965a87701b74b45d05c4c5ca5a244bba2a0c465e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 14:04:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d157f60b28b34870798be235a128497f2c8c921f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 14:05:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd316d2a6d504af7b3c8a89018cad8ff : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9726c1a5938ba9d04d3bb1dd8c3bea5d0c4b10b9","revisions":{"9726c1a5938ba9d04d3bb1dd8c3bea5d0c4b10b9":{"kind":"REWORK","_number":1,"created":"2019-11-18 09:21:48.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/31/694731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/31/694731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/31/694731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/31/694731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/31/694731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/31/694731/1"}}},"commit":{"parents":[{"commit":"090a255a58aa1efa80ea98203509440b3da18446","subject":"Evaluate input expression should check the in_context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/090a255a58aa1efa80ea98203509440b3da18446"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-11 12:04:53.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-11-18 09:21:48.000000000","tz":0},"subject":"Mask sensitive data when logging action results","message":"Mask sensitive data when logging action results\n\nWhen passwords or other sensible data is returned by a Action, they can\nbe logged by Mistral. This change uses the password masking\nfunctionality used in mistral-lib and privided by oslo-utils.\n\nThis function uses the standard cut_repr method in mistral-lib, which\nalso means the output is more standardised.\n\nRelated-Bug: #1850843\nChange-Id: I01bf47f7a83102a1a16b15bf0bbb4021707e11fe\n(cherry picked from commit a25c8fab881decff0e61deffe1895f1d99132f9f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9726c1a5938ba9d04d3bb1dd8c3bea5d0c4b10b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9726c1a5938ba9d04d3bb1dd8c3bea5d0c4b10b9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
