)]}'
{"id":"openstack%2Fmistral~694732","triplet_id":"openstack%2Fmistral~stable%2Fstein~I01bf47f7a83102a1a16b15bf0bbb4021707e11fe","project":"openstack/mistral","branch":"stable/stein","topic":"bug/1850843","hashtags":[],"change_id":"I01bf47f7a83102a1a16b15bf0bbb4021707e11fe","subject":"Mask sensitive data when logging action results","status":"MERGED","created":"2019-11-18 09:21:54.000000000","updated":"2019-11-18 13:02:08.000000000","submitted":"2019-11-18 13:00:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694732-1574082055051-ded4c5e8","meta_rev_id":"43e0177eb8625499bcc4d3fddec30b15d3eeccb5","_number":694732,"virtual_id_number":694732,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-18 13:00:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-11-18 10:59:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-11-18 10:59:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 10:59:48.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-18 13:00:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ff89d508caae542789f65d41c2497f317fe53f0","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-18 09:21:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e23cf75c2f356803beddc75fee430e1bd2b9ff4a","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-18 09:22:57.000000000","message":"Topic changed from bug/1850843-stable/stein to bug/1850843","accounts_in_message":[],"_revision_number":1},{"id":"358f3105b06e1e53fc93be68c40c2489828e52d1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-18 10:59:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1141e68f54d4df39c334d1dd58fdd42e3d9c1e5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 11:49:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6a07b18334b42c0ab739bca8fd100e7 : SUCCESS in 6m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/943fb078244249cbb640b4acd6d5903b : SUCCESS in 8m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/46851d124e0441b884a769710f5089f7 : SUCCESS in 9m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecd76c07c8a84376a5e1d2f146a429f7 : SUCCESS in 9m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12b4399d09364abf886f77871ee38155 : SUCCESS in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bd2eeeaad464ec09a15b29f688b5825 : SUCCESS in 7m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67666fbe3da0466bae20e1eee21a8f6f : SUCCESS in 9m 15s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/05d7485461674b80a1769f56e7fe7a9f : SUCCESS in 46m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/678fbc34ba4a4a869f083e1080ad3a18 : SUCCESS in 44m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8fc451aa10d448eb846830a84c460d09 : SUCCESS in 48m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5490872e29024b74b9b12e5856fac276 : SUCCESS in 28m 19s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/246f7c4f13d34a4690421551ef8985d4 : SUCCESS in 25m 52s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b05feff829744c299ad1f49817184469 : SUCCESS in 1h 29m 18s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9318b32953914ca0ab535d799d2b0e95 : SUCCESS in 2h 22m 16s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7754029b0d024b8186c6461e15f19a2f : SUCCESS in 1h 01m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80ec1c9e3ac84ee1957d5bd1405ca7a0 : SUCCESS in 10m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6c516c5abbd59761c56e9cc1d9bf4ddff6bec3aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 11:49:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0039eecbde279024b0b454323ac47ecf21e62754","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 13:00:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/719ab08544b340d996a92db9d5b6524a : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b5f1ec588f24bba825e1979066c4b0e : SUCCESS in 8m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f155fa7f2167478794387f5764b09279 : SUCCESS in 9m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5cbee9898c747ecaf02b924ee4f2e71 : SUCCESS in 7m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa50049110574597a98b2814c0a60399 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76c08fc8fbb545b484bba09eefb99208 : SUCCESS in 4m 31s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0d889cabd96a45c49fa3adda48484e67 : SUCCESS in 45m 46s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f5856a178fc84ce9927328f362ce74e3 : SUCCESS in 1h 04m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/07f7f660f42c4a439c4cd5a665b8bcf7 : SUCCESS in 28m 08s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3507868794c04e47bd1dedc994f6bbbe : SUCCESS in 25m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bc61907d76024f98809a096c060df99d : SUCCESS in 43m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e02484aa59043ff89fd479ec90d0f0e : SUCCESS in 10m 20s","accounts_in_message":[],"_revision_number":1},{"id":"dda8b6dd1eb71504fc71451d21ee45e688ddaa7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 13:00:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"43e0177eb8625499bcc4d3fddec30b15d3eeccb5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 13:02:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2915c7155ecc4a168591ee4856323af1 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6399f9b7e0133202f2d63896b44628668cdc212f","revisions":{"6399f9b7e0133202f2d63896b44628668cdc212f":{"kind":"REWORK","_number":1,"created":"2019-11-18 09:21:54.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/32/694732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/32/694732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/694732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/694732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/694732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/32/694732/1"}}},"commit":{"parents":[{"commit":"19ddb052b3f7d78dd22c354682cb9090ce6d2001","subject":"Retry a DB transaction on \"Too many connections\" error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/19ddb052b3f7d78dd22c354682cb9090ce6d2001"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-11 12:04:53.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-11-18 09:21:54.000000000","tz":0},"subject":"Mask sensitive data when logging action results","message":"Mask sensitive data when logging action results\n\nWhen passwords or other sensible data is returned by a Action, they can\nbe logged by Mistral. This change uses the password masking\nfunctionality used in mistral-lib and privided by oslo-utils.\n\nThis function uses the standard cut_repr method in mistral-lib, which\nalso means the output is more standardised.\n\nRelated-Bug: #1850843\nChange-Id: I01bf47f7a83102a1a16b15bf0bbb4021707e11fe\n(cherry picked from commit a25c8fab881decff0e61deffe1895f1d99132f9f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6399f9b7e0133202f2d63896b44628668cdc212f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6399f9b7e0133202f2d63896b44628668cdc212f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
