)]}'
{"id":"openstack%2Fmistral~694744","triplet_id":"openstack%2Fmistral~stable%2Ftrain~I4d0fdff54de9bee70aeaf10a4ef483ad7000840b","project":"openstack/mistral","branch":"stable/train","topic":"bug/1852722-stable/train","hashtags":[],"change_id":"I4d0fdff54de9bee70aeaf10a4ef483ad7000840b","subject":"Make action heartbeats work for all executor types","status":"MERGED","created":"2019-11-18 10:54:21.000000000","updated":"2019-11-18 20:28:36.000000000","submitted":"2019-11-18 20:26:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694744-1574108814533-a4121484","meta_rev_id":"6375b02abe308456982b23a0e1e109f655763ece","_number":694744,"virtual_id_number":694744,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-18 20:26:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"all":[{"value":1,"date":"2019-11-18 16:26:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":2,"date":"2019-11-18 10:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":1,"date":"2019-11-18 10:54:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 16:26:24.000000000","updated_by":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"reviewer":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"state":"REVIEWER"},{"updated":"2019-11-18 20:26:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c00192af237e7bffb9feccd3f650217397004def","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-18 10:54:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"14f6c194cde2dc6337e1fe6fbb30e3bcce9c7347","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-18 10:54:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d323516cfb5ca93efbe1a45d766f45b0eea3f36d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 13:42:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9309b5fe394cf29810213894775151 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9bcb200d2da4bc4a0309b8ee385e5bc : SUCCESS in 9m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00b3fcab7c2f4b65af8e71e8d1632c25 : SUCCESS in 7m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19f814066bdc4377bf54b125c3a5ab07 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd2c5e16a1d4a68bcb3909e39fb1830 : SUCCESS in 5m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c1e1ffcc303403c90d2cd56062d8710 : SUCCESS in 8m 01s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f06792f8516143d1b21d27c2a49bfef3 : SUCCESS in 40m 55s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2f6e46f7a3b64775bb104266821d5a50 : SUCCESS in 49m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2a5e8902eb454235a73dfd2e056a7737 : SUCCESS in 44m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6dcc9e35849e4092998900d3f3fe9d4f : SUCCESS in 28m 57s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2d7cce788f514c489ad5ff3d0af5fd5b : SUCCESS in 27m 08s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a1dd00192be144e280a18981e846b73c : SUCCESS in 1h 27m 34s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6db8f3d5581d406d9643c8c205eb3445 : SUCCESS in 2h 41m 05s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/31e17e06138f4b8fb2b7754b3542f270 : SUCCESS in 53m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/683ed1f4367342ddb70611a5e2bbd4a4 : SUCCESS in 11m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d939c0918a975a0833b181bbfb60c4444a698f7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 13:42:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33ddabde68bdd494eb1292ab1913775dcb7564db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 16:17:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e76ea7c9f53745b29a809f2c8da40001 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a86740b487844924992dcad24d9398d1 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bc619677efb43c6be57db4952e999fe : SUCCESS in 9m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9cb54cd7070e4f088c606d38f973edf3 : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34cfbdde3e4d41268d00858d18fb16bc : SUCCESS in 5m 10s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6ac97d5e6e9d443cbc7a5e53a8191bc4 : SUCCESS in 46m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3f9d5a65a42d4c4187d2a4a56a13adff : FAILURE in 45m 03s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6623539fc1544c40a5f7ac53127d5210 : SUCCESS in 28m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8985fdc0502f45fd91837ac27cff9f07 : SUCCESS in 26m 39s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/30cb7e65877844e697a82041e57fe020 : SUCCESS in 1h 41m 59s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3f318fe19e2f4c97952aa2a545c5844e : SUCCESS in 41m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d4cb34a04b94fcebd4ff43ea7bfc893 : SUCCESS in 9m 43s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/297a04a2f5f640e399f91c02a7a81b3b : SUCCESS in 2h 30m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8e54843e346c332986a837a6a7e061d299138b6a","author":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"date":"2019-11-18 16:26:24.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"343e22f324381bf10b4f3f39036491310aaafc01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 18:57:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22dd3652ab6f4d7a92633f4f3a021e6e : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6edf0038b54c4ccaa2eb5258ded70779 : SUCCESS in 9m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cf6c0c9154d4c27be461c09e5333631 : SUCCESS in 9m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70270745cc0b49fca434bd10ab3490af : SUCCESS in 10m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fc832739760494b8735907002be9a51 : SUCCESS in 6m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbf53254286e49698ddbb0e55ae38445 : SUCCESS in 9m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/935664b4db43496d938aec284b471801 : SUCCESS in 42m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d229b0efc78641d688a0f8655500c672 : SUCCESS in 51m 01s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/71255062786b482aaab961b73644bb4d : SUCCESS in 51m 26s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fc02a24e5c3a4e728becf5ef4f40c53c : SUCCESS in 28m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/78a8f4d7b7c84b58b7aee4494f6e30b7 : SUCCESS in 28m 37s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/bc9adb73d34a4812806368c9a13875d1 : SUCCESS in 1h 32m 25s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ec9c8ce52147401bbf515163e9fc4b8c : SUCCESS in 2h 20m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/34fd1789450942e4b7488c2e256b5da7 : FAILURE in 1h 17m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9318cf40e4cc4a1a913245cd8044d812 : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":1},{"id":"06fb6669bf1303a6ac28ebb8bae3f1ff46fd04de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 18:57:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1742d390712cab38070bd002d55e807eba443903","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 20:26:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"109353c570a4de34d2f6594a5b80ff39843c8f71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 20:26:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/721da60872b04293ad7f648cbfe3de99 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ade33fa73ea4df5a23f2facbbc47340 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/274553a2ff8341ed8b2f8c2d7a879b33 : SUCCESS in 8m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b71f9353b0ee48c98641ddb18b86e7ff : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdf1751a96364cf9a3d40aaf5c512f82 : SUCCESS in 6m 09s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ff54e46b6a1c4288b6e38493ba07df98 : SUCCESS in 45m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/09e452411fa144339ae3a83575e05bcb : SUCCESS in 44m 10s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4069191a566e4930a5745042d7c31bfc : SUCCESS in 29m 26s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/cd7f2911402d4c30ae381aba4ea64178 : SUCCESS in 29m 18s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/438e686751b34cd2b5ef7ba99c9dae44 : SUCCESS in 1h 23m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9dc067def437430e88ef9554367c0074 : SUCCESS in 57m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d5cc5619ccf43d6ad7c63bb7d3fdc75 : SUCCESS in 11m 14s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/823eedcb111f4adaa69b679f72680f3f : SUCCESS in 1h 27m 41s","accounts_in_message":[],"_revision_number":1},{"id":"b74ce87b5c7557ad6a6bc2537cdcdc724d4d1914","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 20:28:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c52cdbdb339e48ad918631657bbab8d2 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d878d5409f3bd74f6137490d7fac555d4975fee9","revisions":{"d878d5409f3bd74f6137490d7fac555d4975fee9":{"kind":"REWORK","_number":1,"created":"2019-11-18 10:54:21.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/44/694744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/694744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/694744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/694744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/694744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/694744/1"}}},"commit":{"parents":[{"commit":"24254685778c028792793774c0f8072a02e09614","subject":"Refactor action execution reporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/24254685778c028792793774c0f8072a02e09614"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-13 09:34:44.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-18 10:54:21.000000000","tz":0},"subject":"Make action heartbeats work for all executor types","message":"Make action heartbeats work for all executor types\n\n* Previously action hearbeats didn\u0027t work in case of using local\n  executors because the component responsible for sending heartbeats\n  was started by the executor RPC server which doesn\u0027t make sense to\n  initialize for a local executor. This patch refactors the code\n  so that now heartbeats get sent for any type of executors. For\n  local executors it is also useful because a cluster node that\n  runs an engine and a local executor may also crash. With this\n  change, remaining cluster nodes will be able to understand that\n  the action will never complete and one of them will time it out.\n  If all is fine with the node where the local executor is running\n  then heartbeats will be sent normally and the action won\u0027t time\n  out. Before this change, in case of local executors a long running\n  action would always time out after a configured amount of time\n  (by default, 60 mins) just because local executors never sent\n  heartbeats.\n* Made a lot of renamings to clearly see what component is\n  responsible for.\n* Wrote the tests that check the heartbeat sender, both positive\n  and negative scenarios for local and remote executor types.\n\nCloses-Bug: #1852722\n\nChange-Id: I4d0fdff54de9bee70aeaf10a4ef483ad7000840b\n(cherry picked from commit 7ec4f26744ac17151adae4c06cd8a17b71f409a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d878d5409f3bd74f6137490d7fac555d4975fee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d878d5409f3bd74f6137490d7fac555d4975fee9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
