)]}'
{"id":"openstack%2Fmistral~695212","triplet_id":"openstack%2Fmistral~master~I26667a82ec768c858f0282124864e377d8cf39f4","project":"openstack/mistral","branch":"master","topic":"bug/1792875","hashtags":[],"change_id":"I26667a82ec768c858f0282124864e377d8cf39f4","subject":"Allow the prefixes like \"eq:\" and \"neq:\" in the custom REST UUID type","status":"MERGED","created":"2019-11-20 14:53:01.000000000","updated":"2019-12-19 08:09:48.000000000","submitted":"2019-12-19 08:08:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695212-1576742902714-d94f54cc","meta_rev_id":"3f86215d1f87c53adb57f2d2996910483e4ce31f","_number":695212,"virtual_id_number":695212,"owner":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-19 08:08:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-12-19 06:11:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-12-19 06:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-12-19 06:11:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-27 12:20:57.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-12-16 15:01:14.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-12-17 08:04:24.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-12-17 08:04:24.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-12-17 08:04:24.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-12-19 06:07:58.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-12-19 06:11:05.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-12-19 08:08:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b8efcf5cec3138e8158e79a29e732bdc62b4bcad","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-20 14:53:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14caaf87dd87349a5f731cd34d9adf6250639c98","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-20 15:53:08.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4dfc3a8c8f0c8eac08b20cd98669c10151fe61ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 18:46:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fe79af9e2bf46eeb12ac73d42341d9b : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e12b2863b8644f8bdbe6f1413aaa925 : SUCCESS in 9m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e91c8ea99abe4bca9a2532911cf2f9aa : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8278389450947c9986407d2d6e3a4df : SUCCESS in 7m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5eddfbc4dd674153b339ef1e6faa5d05 : SUCCESS in 9m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/eca4644c68414bd780b088a6c9029a8b : SUCCESS in 1h 00m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ab0fc66195af4082adc10536bc096bdb : SUCCESS in 52m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d02be5e9797b4ee88b47e2aea8afc0fd : SUCCESS in 54m 14s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d1ebbe3a0f7c434e85b601fc19319c2d : SUCCESS in 29m 36s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1742180ab56d496e9a0873a56032e41b : SUCCESS in 28m 18s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/482798094cff4a0c88fbfbf8b2d84815 : SUCCESS in 1h 25m 53s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/583e2d65bae74c4284dd9f2defffd74f : SUCCESS in 2h 46m 24s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8430143646dc49ba89ded67bc63afd85 : SUCCESS in 1h 06m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6beceae9c4b04ed783351e954cfeb620 : SUCCESS in 9m 35s","accounts_in_message":[],"_revision_number":2},{"id":"68e89f4b7875e96bea1c3e144c9e9d96fa0491f7","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-26 07:15:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"115b01f0c665c29fb60e84f365297d471dc41aa5","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-11-27 12:20:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"69673b92ffc72c614ac77f113ce852db86a1e28f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-06 07:20:40.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f6d3f2cd4a0c13f41303d4eb17333eac5f0bd854","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-06 07:31:36.000000000","message":"Patch Set 3: Code-Review-1\n\nPlease note that I fixed the commit message, fixed typos and made the description of the change clearer. I still vote -1 though because \"eq:\"  and \"neq:\" are not the only possible filter types that can be applied to this type. Most of other filters mentioned in the filter_utils.py module are also applicable. For example, \"in:\" and \"nin:\".\n\nMy suggestion is to allow all the filter types. As the first step, I think we need to refactor filter_utils.py. All filter type values (like \"eq:\" and \"neq:\") need to be defined as public constants of the module and there needs to be the constant ALL \u003d [EQ, NEQ, ...] containing all of them. That will make it much easier to work with filters everywhere throughout the system, including this particular place where we implement the logic of the REST UUID type.","accounts_in_message":[],"_revision_number":3},{"id":"4eb967d1acd344f7523eb23d5a82f1f0ad1d16d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 10:02:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ea8bd27d4e948da92804139c0c3be46 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f95ed24055db4252a9d569084508c899 : SUCCESS in 10m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d9520a7a7644d54bd704be8fb527d5b : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3d0a287edad43d6b606284f9706ddad : SUCCESS in 8m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e431a1b82364bad9e3b253e315ab899 : SUCCESS in 8m 24s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/382ffeed97684acca4a1c5eb13c3d4cc : SUCCESS in 43m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6a0f8e74d98743b4bb44c79e2d5116ca : SUCCESS in 36m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a8f523ffaa2f4aa0a17db3169ae82873 : SUCCESS in 42m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ac05328ef89f44cfae26a76736a525cf : SUCCESS in 28m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6bfa9a61c3264dc593c82e2a3544c51b : SUCCESS in 27m 49s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e749ad86eb194644b9598dbf88c03bb3 : SUCCESS in 1h 16m 26s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4c65ae1c90ba4d4cbee71294b22fe79c : SUCCESS in 2h 39m 39s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/72064de76ab94334baedcf04d0e2bea2 : SUCCESS in 55m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca8f8379737749f0b08aebf03c1e1836 : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":3},{"id":"f7e46001edeb38bd12794652c9425dcb872143d6","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-12-15 12:35:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ec2f8d275b5bbf07087eb573a74dc6be94274cf7","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-12-15 13:59:01.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"65b434ca34d46cb1cdfe2cbf807757b7ac097564","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 14:50:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/068dc9fb29a54453b379625da7cb357b : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d78ac8787f364a9aadf8a4537733c8a4 : SUCCESS in 8m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/89549d02fcc949b880f83b11f78d433a : SUCCESS in 10m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd3f334cfb4d41f99db6a5efb5f38609 : SUCCESS in 8m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30d1b6c5c5ef45b185b32f7e6aad1825 : SUCCESS in 10m 58s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e72bbee1939347f9b655b9651bb91b12 : FAILURE in 15m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/15c16f5bb5cc4cf2a7e0ddfa2e3a9579 : FAILURE in 11m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b1cdfb90dee84788ad273f09a2003f75 : FAILURE in 15m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/37feb61ea5c34cee910c473c51053d18 : SUCCESS in 35m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b72b12840c5d406f9335e108d003328c : FAILURE in 35m 56s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a7dda96bbfc449a2a3dc6819ad05497a : FAILURE in 1h 18m 24s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9df6d84ce7654ab182bfaf4508dd3f86 : FAILURE in 2h 07m 11s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c49d6fb376bf462a9ddec962b0e5c245 : SUCCESS in 57m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7fece02702f45acabf25e9f93adf803 : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":4},{"id":"3695ef37ecf71e815b63841603bb43234722ad5f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-16 04:18:17.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nThanks! It now looks much better with these constants. Just please address minor comments and make the tests pass.","accounts_in_message":[],"_revision_number":4},{"id":"220bc3e1bae87ee2cd49357243132078158dde0a","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-12-16 07:44:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"84a01d102693f9230dc32db2e7b0fded6e0d95da","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-16 07:50:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6479941b2b9fef00e8e1702cbe21a61acad7d0ad","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-12-16 07:54:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"505e875fc323a9e8c3fc1d580760e42e2f143464","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-12-16 08:00:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c5e3420a5088069dad3a053ee7d96c4163c69710","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 09:21:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/152925a786be4b149ec24e708a6d33ed : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c12fe0750614f0ebb5f8048fc31ff8b : SUCCESS in 12m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46105461327f4ecba893c2e545c4302b : SUCCESS in 12m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b34cc133162e4de9840d674450b75c3e : SUCCESS in 8m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48dc1f7b1b7a472e8e4626f51a6ee173 : SUCCESS in 11m 33s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fdd3eb3ee17d48809a5a0d0e212b8085 : FAILURE in 1h 00m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2549569128a44003bb5c1ec44ef333b5 : FAILURE in 46m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/96471759088f4fd4a3b3e20130ebce9c : FAILURE in 49m 39s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/525197493f2942f3a4aa77c975318577 : SUCCESS in 29m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/548bf29af1e64bda85ab248afefb6ea3 : SUCCESS in 28m 57s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c1dc6425c009495ba378c36b2e64ae1b : SUCCESS in 1h 19m 31s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c46ca00e3c72446bbaabefc0ddb33e8e : FAILURE in 49m 01s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9f5f1ed7cd724eb8ac0118f7c76ecd82 : SUCCESS in 1h 06m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ed41143744745e2a6c1d7043d201c49 : SUCCESS in 10m 28s","accounts_in_message":[],"_revision_number":5},{"id":"ad4fd7bd33692e5e4a43edd2bc742886996a27bc","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-16 09:26:22.000000000","message":"Patch Set 5:\n\nLooks good now. Please investigate why CI jobs are failing.","accounts_in_message":[],"_revision_number":5},{"id":"040fa9796317a79e668c8ea1f04342c16094bf6a","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-12-16 12:37:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fb01e9b0d1cbb718d738e05c36c002eff17e3465","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 14:59:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7d5d4b6c78e471a9a20b932613fdf54 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9927b74a3ab1498d98980bfd89e9abab : SUCCESS in 13m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71cf59969caf40f78fb58138da8ec3cf : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39760569d89549ca9cb6ec436e9b1533 : SUCCESS in 18m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/218f96c1b77543588e5346a6aa2adf7d : SUCCESS in 10m 55s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7cd0831d5ea04e2cb255c0498b2cd45c : SUCCESS in 48m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/44f26d3d53a444a491096dca455c3f9f : SUCCESS in 41m 19s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/58e6511962ff4427b1213454fbf109d9 : SUCCESS in 58m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/33baf8739d9141c1a42fa7d17a0bba74 : SUCCESS in 34m 23s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/12d440c196054432b4acb13e7393493d : SUCCESS in 28m 00s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7c77bdd261534ee289d9876de1dcffde : SUCCESS in 2h 18m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/454e3183f6784b879ed296ecde0d32c6 : SUCCESS in 57m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f8baa993802469ab1fb60e3fe4635b4 : SUCCESS in 11m 51s","accounts_in_message":[],"_revision_number":6},{"id":"55deacd97f8e90237504a5d93b5f4a56367bf636","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-16 15:01:14.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"608d83b00ec208591439c493341301e5f316a617","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-17 04:41:11.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"bc84ee12bcd471910e50788b430b871603ee3629","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-12-17 08:43:48.000000000","message":"Patch Set 6: Code-Review-1\n\n(8 comments)\n\nLooks good! Some raking would still be good.","accounts_in_message":[],"_revision_number":6},{"id":"c0b1ff2e3a390041a8718639f257019ff829c5ea","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-18 03:38:26.000000000","message":"Patch Set 6: Code-Review-1\n\nYeah, I agree with the comments of Andras. Please address them.","accounts_in_message":[],"_revision_number":6},{"id":"dcd55bf2f57eab07a7d4a1d13b04d55058693a8d","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-12-18 10:59:42.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3a61f4f9fc3c838b7f566624d0b6a13308d6ed14","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-12-18 10:59:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"efad034ca510f989f96077849c6a46ebcaf0e75c","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-12-18 11:14:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7509bcb89093cec7d2f1379d71e880473c96e52b","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-12-18 12:06:09.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"aa85e7605ecaeaebd6c9ef39e7af60778b4c1228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 15:25:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d57fb24f61eb4bbf942dc6466e049a0b : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac62f57911d842f2acea7b1ba128c197 : SUCCESS in 11m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a4059e16a024f53948bcc38edbae4bb : SUCCESS in 13m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3c4a9488e3c468199df9f4486555923 : SUCCESS in 12m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd22958d3f924923a877e9be3d956fd6 : SUCCESS in 10m 40s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8e2894d9a89240a3b4419495000a3b09 : SUCCESS in 51m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/10dc7722f61d406fa06c8e6627f3cb93 : SUCCESS in 57m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/778264116f304fcb923b2774ecd2d54e : SUCCESS in 46m 20s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/221de995a48e49a79fb39e0234517296 : SUCCESS in 30m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1c92a03b28084a119d6f65320cbc93c5 : SUCCESS in 31m 23s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0905de6d820b4e85b112f05272dd6186 : SUCCESS in 2h 36m 59s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/38daf3982e2f48b09c9330da22b2eb17 : SUCCESS in 1h 04m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4741482fa6942a2a84c52a4967b0474 : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":8},{"id":"e4c892a64574919864c68988b688fba9d1477be2","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-19 06:07:58.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"b39772a67ab23ff0a5456dd7bbcf847369ac3b95","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-12-19 06:11:05.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nThank you!","accounts_in_message":[],"_revision_number":8},{"id":"99554e250c48b11deef32e38be1c58382243f66e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 06:11:19.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"2a935b29794f3c9fcc54a15fd096b3f248f78077","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 08:08:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"1a69d8f039d33ee0c4893328358890f6e87d48ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 08:08:22.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78c9bae68dab405fa3e75c3a520a0464 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91a94797a6f242eea2613356062cc702 : SUCCESS in 13m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/198f1d29bfaf4b469240a04dc44d0969 : SUCCESS in 11m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a33581668d141a08773925b2b85b376 : SUCCESS in 13m 19s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/69f20e3741a14e158cf0a1a9a2285914 : SUCCESS in 54m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ebf773458cfa42c7b0cc0ff089c83c4b : SUCCESS in 50m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/46b58056a22f45de948feb3d031e7868 : SUCCESS in 34m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b46358c7fd2c4386a697c4d4efcf9f0a : SUCCESS in 32m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0480c29d7209473d917458b4f76f4f3e : SUCCESS in 48m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ea07ccb58774a9884740d9f39d2f5da : SUCCESS in 13m 34s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c6c365c08de643b4a1959f9cc6b59c82 : SUCCESS in 1h 50m 23s","accounts_in_message":[],"_revision_number":8},{"id":"a1ffa3009b5ca88c4f1b68f6aa244b199489eb7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 08:09:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b133739a5bb4deb82175913d36d121c : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"6948e50de80559f78988ff34d92969aab61bf012","revisions":{"a18bc4a787b490aa8b7641c12622733facf994cf":{"kind":"REWORK","_number":1,"created":"2019-11-20 14:53:01.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/12/695212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/12/695212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/12/695212/1"}}},"commit":{"parents":[{"commit":"a9b19b5f2502b918163643b453df5bdbc6474ce7","subject":"Merge \"Make it possible to set None to REST API filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a9b19b5f2502b918163643b453df5bdbc6474ce7"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-11-20 14:50:42.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-11-20 14:50:42.000000000","tz":0},"subject":"Extended uuid type to support prefixes eq,neq","message":"Extended uuid type to support prefixes eq,neq\n\n *fields of type uuid with filter prefix are\n  denied because the they dont match the type\n  (ie. eq:6c07a453-c5e1-4bbe-97ed-3cb77b4f55ff\n   will throw an inputException).\n\nChange-Id: I26667a82ec768c858f0282124864e377d8cf39f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a18bc4a787b490aa8b7641c12622733facf994cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a18bc4a787b490aa8b7641c12622733facf994cf"}]},"branch":"refs/heads/master"},"b4e85955f7adbd79f9a2eabed7778654e7b62d13":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-20 15:53:08.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/12/695212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/12/695212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/12/695212/2"}}},"commit":{"parents":[{"commit":"a9b19b5f2502b918163643b453df5bdbc6474ce7","subject":"Merge \"Make it possible to set None to REST API filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a9b19b5f2502b918163643b453df5bdbc6474ce7"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-11-20 14:50:42.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-11-20 15:53:05.000000000","tz":0},"subject":"Extended uuid type to support prefixes eq,neq","message":"Extended uuid type to support prefixes eq,neq\n\n *fields of type uuid with filter prefix are\n  denied because the they dont match the type\n  (ie. eq:6c07a453-c5e1-4bbe-97ed-3cb77b4f55ff\n   will throw an inputException).\n\nCloses-Bug: #1792875\nChange-Id: I26667a82ec768c858f0282124864e377d8cf39f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b4e85955f7adbd79f9a2eabed7778654e7b62d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b4e85955f7adbd79f9a2eabed7778654e7b62d13"}]},"branch":"refs/heads/master"},"084d2cd3c3cc87c7e34ccb46f96a1bb182f6dd85":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-12-06 07:20:40.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/12/695212/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/12/695212/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/12/695212/3"}}},"commit":{"parents":[{"commit":"a9b19b5f2502b918163643b453df5bdbc6474ce7","subject":"Merge \"Make it possible to set None to REST API filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a9b19b5f2502b918163643b453df5bdbc6474ce7"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-11-20 14:50:42.000000000","tz":0},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-06 07:20:36.000000000","tz":0},"subject":"Allow the prefixes \"eq:\" and \"neq:\" in the custom REST UUID type","message":"Allow the prefixes \"eq:\" and \"neq:\" in the custom REST UUID type\n\n* Fields of type UUID with a filter prefix are denied because they\n  don\u0027t match the UUID string format. E.g. \n  \"eq:6c07a453-c5e1-4bbe-97ed-3cb77b4f55ff\" will throw an\n  InputException. This patch makes the change that allows to have \"eq:\"\n  and \"neq:\" prefixes in a value of such fields.\n\nCloses-Bug: #1792875\nChange-Id: I26667a82ec768c858f0282124864e377d8cf39f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/084d2cd3c3cc87c7e34ccb46f96a1bb182f6dd85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/084d2cd3c3cc87c7e34ccb46f96a1bb182f6dd85"}]},"branch":"refs/heads/master"},"8238c645e3e2a275c18efb19b9e42e44a87ac935":{"kind":"REWORK","_number":4,"created":"2019-12-15 12:35:14.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/12/695212/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/12/695212/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/12/695212/4"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-12-15 12:34:58.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-12-15 12:34:58.000000000","tz":0},"subject":"Allow the prefixes like \"eq:\" and \"neq:\" in the custom REST UUID type","message":"Allow the prefixes like \"eq:\" and \"neq:\" in the custom REST UUID type\n\n* Fields of type UUID with a filter prefix are denied because they\n  don\u0027t match the UUID string format. E.g.\n  \"eq:6c07a453-c5e1-4bbe-97ed-3cb77b4f55ff\" will throw an\n  InputException. This patch makes the change that allows to have:\n  \"eq:\" \"neq:\" \"gt:\" \"gte:\" \"lt:\" \"lte:\" \"has:\" \"in:\" \"nin:\" prefixes\n  in a value of such fields.\n\nCloses-Bug: #1792875\nChange-Id: I26667a82ec768c858f0282124864e377d8cf39f4\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8238c645e3e2a275c18efb19b9e42e44a87ac935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8238c645e3e2a275c18efb19b9e42e44a87ac935"}]},"branch":"refs/heads/master"},"e36ce062db6415afe2c6f1dd14647453bc71ca7c":{"kind":"REWORK","_number":5,"created":"2019-12-16 08:00:16.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/12/695212/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/12/695212/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/12/695212/5"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-12-15 12:34:58.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-12-16 07:58:53.000000000","tz":0},"subject":"Allow the prefixes like \"eq:\" and \"neq:\" in the custom REST UUID type","message":"Allow the prefixes like \"eq:\" and \"neq:\" in the custom REST UUID type\n\n* Fields of type UUID with a filter prefix are denied because they\n  don\u0027t match the UUID string format. E.g.\n  \"eq:6c07a453-c5e1-4bbe-97ed-3cb77b4f55ff\" will throw an\n  InputException. This patch makes the change that allows to have:\n  \"eq:\" \"neq:\" \"gt:\" \"gte:\" \"lt:\" \"lte:\" \"has:\" \"in:\" \"nin:\" prefixes\n  in a value of such fields.\n\nCloses-Bug: #1792875\nChange-Id: I26667a82ec768c858f0282124864e377d8cf39f4\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e36ce062db6415afe2c6f1dd14647453bc71ca7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e36ce062db6415afe2c6f1dd14647453bc71ca7c"}]},"branch":"refs/heads/master"},"9af29540e876a267528c26aae6fa91534a10fd48":{"kind":"REWORK","_number":6,"created":"2019-12-16 12:37:26.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/12/695212/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/12/695212/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/12/695212/6"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-12-15 12:34:58.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-12-16 12:37:09.000000000","tz":0},"subject":"Allow the prefixes like \"eq:\" and \"neq:\" in the custom REST UUID type","message":"Allow the prefixes like \"eq:\" and \"neq:\" in the custom REST UUID type\n\n* Fields of type UUID with a filter prefix are denied because they\n  don\u0027t match the UUID string format. E.g.\n  \"eq:6c07a453-c5e1-4bbe-97ed-3cb77b4f55ff\" will throw an\n  InputException. This patch makes the change that allows to have:\n  \"eq:\" \"neq:\" \"gt:\" \"gte:\" \"lt:\" \"lte:\" \"has:\" \"in:\" \"nin:\" prefixes\n  in a value of such fields.\n\nCloses-Bug: #1792875\nChange-Id: I26667a82ec768c858f0282124864e377d8cf39f4\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9af29540e876a267528c26aae6fa91534a10fd48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9af29540e876a267528c26aae6fa91534a10fd48"}]},"branch":"refs/heads/master"},"4afae3b9dd212dfc35ce4d67e2b035ec98dde3fe":{"kind":"REWORK","_number":7,"created":"2019-12-18 10:59:42.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/12/695212/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/12/695212/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/12/695212/7"}}},"commit":{"parents":[{"commit":"bc46b29f01a205d4f5772cfd159e02e5033ca506","subject":"Merge \"Disable the use of anchors when parsing yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bc46b29f01a205d4f5772cfd159e02e5033ca506"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-12-15 12:34:58.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-12-18 10:59:06.000000000","tz":0},"subject":"Allow the prefixes like \"eq:\" and \"neq:\" in the custom REST UUID type","message":"Allow the prefixes like \"eq:\" and \"neq:\" in the custom REST UUID type\n\n* Fields of type UUID with a filter prefix are denied because they\n  don\u0027t match the UUID string format. E.g.\n  \"eq:6c07a453-c5e1-4bbe-97ed-3cb77b4f55ff\" will throw an\n  InputException. This patch makes the change that allows to have:\n  \"eq:\" \"neq:\" \"gt:\" \"gte:\" \"lt:\" \"lte:\" \"has:\" \"in:\" \"nin:\" prefixes\n  in a value of such fields.\n\nCloses-Bug: #1792875\nChange-Id: I26667a82ec768c858f0282124864e377d8cf39f4\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4afae3b9dd212dfc35ce4d67e2b035ec98dde3fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4afae3b9dd212dfc35ce4d67e2b035ec98dde3fe"}]},"branch":"refs/heads/master"},"6948e50de80559f78988ff34d92969aab61bf012":{"kind":"REWORK","_number":8,"created":"2019-12-18 12:06:09.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/12/695212/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/12/695212/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/695212/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/12/695212/8"}}},"commit":{"parents":[{"commit":"bc46b29f01a205d4f5772cfd159e02e5033ca506","subject":"Merge \"Disable the use of anchors when parsing yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bc46b29f01a205d4f5772cfd159e02e5033ca506"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-12-15 12:34:58.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2019-12-18 12:05:57.000000000","tz":0},"subject":"Allow the prefixes like \"eq:\" and \"neq:\" in the custom REST UUID type","message":"Allow the prefixes like \"eq:\" and \"neq:\" in the custom REST UUID type\n\n* Fields of type UUID with a filter prefix are denied because they\n  don\u0027t match the UUID string format. E.g.\n  \"eq:6c07a453-c5e1-4bbe-97ed-3cb77b4f55ff\" will throw an\n  InputException. This patch makes the change that allows to have:\n  \"eq:\" \"neq:\" \"gt:\" \"gte:\" \"lt:\" \"lte:\" \"has:\" \"in:\" \"nin:\" prefixes\n  in a value of such fields.\n\nCloses-Bug: #1792875\nChange-Id: I26667a82ec768c858f0282124864e377d8cf39f4\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6948e50de80559f78988ff34d92969aab61bf012"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6948e50de80559f78988ff34d92969aab61bf012"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
