)]}'
{"id":"openstack%2Fmistral~695240","triplet_id":"openstack%2Fmistral~stable%2Frocky~I26d587241f765b13e135be5c4cf90976903af9ea","project":"openstack/mistral","branch":"stable/rocky","topic":"bug/1850843","hashtags":[],"change_id":"I26d587241f765b13e135be5c4cf90976903af9ea","subject":"Make the mistral devstack jobs non-voting in Rocky","status":"MERGED","created":"2019-11-20 16:36:13.000000000","updated":"2019-11-21 09:16:20.000000000","submitted":"2019-11-21 09:16:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695240-1574327780008-a4121484","meta_rev_id":"71fe35960ef0595b366c8ad60a76ef744d0a6418","_number":695240,"virtual_id_number":695240,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 09:16:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":1,"date":"2019-11-20 20:04:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-11-21 06:28:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-11-21 06:28:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-20 20:04:02.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-11-21 06:28:03.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-21 09:16:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cefd1dbc70c24d7d3dc2eca385b0cc4ae71ba1fa","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-20 16:36:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4f94f0d5e0e04a9ceeaa527c2f1482463113913","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 19:28:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61a4561c8ef14e138b3c2a703d78707b : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b0708d24e0ee4eb7b10dba35162dc88b : SUCCESS in 8m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a0bd7dfa262842d7998d9673345536ec : SUCCESS in 7m 56s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/26a040fa66f842f7a7c1c02b083cff50 : SUCCESS in 6m 49s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/13fa867a88244e28aafa3a6a70d3bad8 : FAILURE in 1h 00m 47s (non-voting)\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d7b6e218f9d140f591ed9392c8d0436b : SUCCESS in 49m 10s (non-voting)\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f7deee707fa24c81a0e48521f8548dfd : FAILURE in 46m 42s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8e1dc59dd22e44d0a94b9602c0c13843 : SUCCESS in 26m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/54195cdfb7374d9c8dbe4189aaa6f8a7 : SUCCESS in 30m 16s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/d0f2498c24244ab08fb383daacc780d3 : SUCCESS in 2h 30m 37s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/db94b68a256f49ef9af8faf6a4260e29 : SUCCESS in 56m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0643a5cac39540faafa4193b4fb8c9a4 : SUCCESS in 10m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0ab0f1db4f09589a26f71e5a534b92e3f53cdf4f","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-11-20 20:04:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a286928020ef3bb66daeec876255e23baeccc7ac","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-21 06:28:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b0a297ea363b57d7c9f71754b541844dfc7256a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 06:28:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"728ba2305079a05cd1fa8c86f77bb396d170dea5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 09:16:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d9e554ce7534f739962e4d101a791f1 : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6eb30138f8749a191d91945a7560c4a : SUCCESS in 8m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/53b40b9f12ba4e0c82cb2f5918571173 : SUCCESS in 9m 20s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f8cd65a39935494e824a966cba2569db : SUCCESS in 5m 24s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/404fc6122165461d85899b2a4e705c0c : SUCCESS in 27m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/cc556ce4d7904419a9132bdc1990bd38 : SUCCESS in 24m 53s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5fb29bb8286d4f66b140a9751e9575bb : SUCCESS in 2h 42m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84aaa750e5274e2aa5678c2ea12acb63 : SUCCESS in 10m 28s","accounts_in_message":[],"_revision_number":1},{"id":"fcea7dfb2768e901154229468d7b44d4413e6091","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 09:16:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4dc6cda9e18e12d297c5eabaac87571ab74f1cb5","revisions":{"4dc6cda9e18e12d297c5eabaac87571ab74f1cb5":{"kind":"REWORK","_number":1,"created":"2019-11-20 16:36:13.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/40/695240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/40/695240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/695240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/695240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/695240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/40/695240/1"}}},"commit":{"parents":[{"commit":"eab767f3ed1a6dfd4f50b214556142346c830c35","subject":"Add release note for fix error of cron trigger run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eab767f3ed1a6dfd4f50b214556142346c830c35"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-20 16:31:19.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-20 16:35:49.000000000","tz":0},"subject":"Make the mistral devstack jobs non-voting in Rocky","message":"Make the mistral devstack jobs non-voting in Rocky\n\nThese jobs are problematic when trying to merge a security fix into\nRocky[1]. This change has passed and merged intro master, Train and\nStein. It even passed CI once in Rocky (on Nov 19th) but then failed\nin the gate.\n\nThis change disables CI to allow this security fix to land and will\nbe reverted afterwards where we can investigate the unstable CI.\n\n[1]: https://review.opendev.org/#/c/694733/\n\nChange-Id: I26d587241f765b13e135be5c4cf90976903af9ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4dc6cda9e18e12d297c5eabaac87571ab74f1cb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4dc6cda9e18e12d297c5eabaac87571ab74f1cb5"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
