)]}'
{"id":"openstack%2Fmistral~695241","triplet_id":"openstack%2Fmistral~stable%2Frocky~I40463bc34006665c0ac8659f0bb445f47a4d747e","project":"openstack/mistral","branch":"stable/rocky","topic":"bug/1850843","hashtags":[],"change_id":"I40463bc34006665c0ac8659f0bb445f47a4d747e","subject":"Revert \"Make the mistral devstack jobs non-voting in Rocky\"","status":"ABANDONED","created":"2019-11-20 16:36:13.000000000","updated":"2019-12-25 08:22:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c1fc0b116c679bef6b92713cfc83c7bb032a6650","_number":695241,"virtual_id_number":695241,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"date":"2019-11-25 03:26:23.000000000","_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-20 19:55:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"recommended":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":-2,"date":"2019-11-26 09:31:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"date":"2019-11-26 08:28:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2019-11-20 20:03:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":-1,"date":"2019-11-26 09:31:52.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-20 19:55:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-20 20:03:31.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-11-25 03:26:23.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-26 08:28:37.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"}],"messages":[{"id":"dbed55cf3a9041062bec34c0b76db252a31fa40b","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-20 16:36:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86df50e7a297695a3c6feb09b4cbb6c7ca3f12e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 19:55:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ec29179d1df4e37a355554d8a6c0434 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fba9a6eb1e074e4abb6d73503c37bcc6 : SUCCESS in 8m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d0ec7d626da94f8b9ab4cfdcc7dcda7f : SUCCESS in 9m 11s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f81adc6712bc45ee8bda83594c14a7c6 : SUCCESS in 5m 21s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d57fa11c99ce4c9283011ae31ef6eb4b : SUCCESS in 44m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/850762d01ade412aa2092bc26bde74e7 : SUCCESS in 50m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1b5d5acd4bde4e7bab8fa2df51e83073 : POST_FAILURE in 52m 48s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bc043d84205140079ca5283c3748c2a4 : SUCCESS in 26m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/816f77b3fd8845bf8eab0362b9d6b54f : SUCCESS in 24m 33s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7b280328c7854a9ca1d158ec8d00b02c : SUCCESS in 2h 42m 13s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5a0e7140bc0d40b88a0bb047336aceb7 : SUCCESS in 1h 16m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3690099035314da296d3b8861ef67407 : SUCCESS in 9m 35s","accounts_in_message":[],"_revision_number":1},{"id":"aab33c2d68de53858f7dde90793a0e33362098e0","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-11-20 20:03:31.000000000","message":"Patch Set 1: Code-Review+1\n\nFrom what I see the devstack failures seems to be random. It really smells of a race condition in our tests in tempest","accounts_in_message":[],"_revision_number":1},{"id":"57bd021ab77c26151be543c7e0a52cef5a261ccf","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-23 13:00:08.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eab4c849af526387ee404248f74b32323fc0573f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 15:39:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/267bc64fd96d4ebda9dc8759c5ccd695 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f0ca4447d554bbcbc3ef4638cf540c9 : SUCCESS in 8m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/74383bc46a9f4fdc8667501dec6f1970 : SUCCESS in 8m 26s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/dc324b26586b44e98d1b67aa701b215b : SUCCESS in 4m 26s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/11adec9f691a46e7b51c50f611cea13a : FAILURE in 49m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d37412afa11e488ea980ba75baabd29d : SUCCESS in 48m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fbae00b7d8f54f3ea9580edf18b5bc05 : FAILURE in 48m 40s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f2ab1d28c35c4e648431f3de61c6ddec : SUCCESS in 26m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/39bd3a7038524d78acbca74a9f9501c3 : SUCCESS in 24m 29s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/fdedb68e6db8450cb505998a8a783411 : SUCCESS in 2h 33m 29s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/3a75d7a3ddcc4971a622e2ee03737ef2 : SUCCESS in 59m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/319f20fa20034247852510ee7f65051a : SUCCESS in 10m 58s","accounts_in_message":[],"_revision_number":1},{"id":"84cfa72512262729ac488ac494065ed06e38c125","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-25 03:26:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f2cfb5f969618b0a2b5dec01dbb796c56e3ad814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 06:16:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/707a90dbd4a14b5499733a0ca51b2c10 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ecd94cba8c24b378986150641c0425f : SUCCESS in 8m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2e0ce1afce68435aa578b827effbd291 : SUCCESS in 8m 26s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1608800f3e1e44989d9ef437a2e20153 : SUCCESS in 5m 19s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/42917171b10d4f4aacfcdaa6df86a12a : FAILURE in 44m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0926a6210381441dbdf85df4cc9af697 : SUCCESS in 42m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/42bdb7c7388442b8bfce010c341a7486 : FAILURE in 44m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/68f48b8b9e334dd8923bac1e4da355d7 : SUCCESS in 26m 16s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2a0f69c33f334ac9bfced02967f55325 : SUCCESS in 25m 05s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/778007db2d0e44bead12e72acc32a041 : SUCCESS in 2h 49m 24s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bd134f0a06cb413dab0a19a5ae4666be : SUCCESS in 55m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92e54162041348dabba93d47b48b2467 : SUCCESS in 9m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e431e0ce0c3816ecd61ed814f1b7a7c404745c7a","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-25 14:03:36.000000000","message":"Patch Set 1: -Code-Review\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1e5da77890bea545ae4d4dd62934ff0fed5df162","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-25 14:03:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cb91809fbabffed394c8b5104e6bf296022abe9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 16:54:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3bee9730b984db9a15e6abc01736adc : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d8616440fc58450399ed5f4bd1059d50 : SUCCESS in 10m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a7bc209cea1d4bb68641195706dbde05 : SUCCESS in 10m 17s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0c8baee1f3f345e8a966800b697481b0 : SUCCESS in 4m 23s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/762f0e0d30294cce8b774697e98f33ff : SUCCESS in 45m 51s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1fdff157b8c8411cbea70e5f4feaff66 : FAILURE in 1h 09m 26s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7ef84c4e2f754af3813306bcbe1933b3 : SUCCESS in 48m 44s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c98f7440e53b4b30ab490569499f15be : SUCCESS in 29m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4356db1c8488469b9ec0eb85cf5cbeb4 : SUCCESS in 26m 39s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2f74a1f9a33f4d66823dd9618d814936 : SUCCESS in 2h 44m 44s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/fc4a60ec1cb44a8b958ffbea10675d99 : SUCCESS in 1h 08m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf61d1a75b794bedbfac55109abba38a : SUCCESS in 10m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c1fbf4acd905d09acfcbc3383bbd8ba4f359f4cf","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-11-26 08:28:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"839f3fff39c9641f20c1319528baaa8b6d9b186e","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-26 09:31:52.000000000","message":"Patch Set 1: Code-Review-2 Workflow-1\n\nPlease don\u0027t recheck. The jobs are unstable and for it to merge they need to pass twice in a row (check and gate). This is unlikely to happen.\n\nIf we want to make these voting again we need to make the jobs stable.","accounts_in_message":[],"_revision_number":1},{"id":"ce92e859417e652704da258b9b3e6c2d443d6bef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 11:02:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36a77de3e6648cda4533c16a4e60cab : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04d9fb5b45c645abadb7f58fef3a5293 : SUCCESS in 8m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c0662abeece14f34ab0013c30a4c21d2 : SUCCESS in 7m 42s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/44fe27c886e94504a2ac3d25f28492e3 : SUCCESS in 4m 33s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1344dbbb07544b019718c31be1eb4cc1 : FAILURE in 51m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d35308c8201b413ba1005729101df958 : SUCCESS in 45m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/28e27361c1d240a1a0de6c9e94dbae11 : SUCCESS in 50m 42s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9dab2c49de514644802ae4dcea15332a : SUCCESS in 25m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/870616e7e95044f5a3d36e09bcdaa576 : SUCCESS in 25m 00s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4d4c53caef4b48d5a46f0a33bf49b690 : SUCCESS in 2h 30m 20s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d4a6c7bce2414ab69109a193f050b2c7 : SUCCESS in 1h 00m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4eeefc7043a145388d08b78f86325b9b : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fe79969372f0f51649de53fcf37a8b72079d12b8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-25 08:22:19.000000000","message":"Abandoned\n\nIf needed, we\u0027ll make these jobs voting again later (given that we fix jobs)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"125215c72c89a009f4c798bfa342b831c5e0094c","revisions":{"125215c72c89a009f4c798bfa342b831c5e0094c":{"kind":"REWORK","_number":1,"created":"2019-11-20 16:36:13.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/41/695241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/41/695241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/41/695241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/41/695241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/41/695241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/41/695241/1"}}},"commit":{"parents":[{"commit":"4d3ef8887fb765c9edd36f62cc66904aff4c1b0e","subject":"Mask sensitive data when logging action results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4d3ef8887fb765c9edd36f62cc66904aff4c1b0e"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-20 16:34:44.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-20 16:35:52.000000000","tz":0},"subject":"Revert \"Make the mistral devstack jobs non-voting in Rocky\"","message":"Revert \"Make the mistral devstack jobs non-voting in Rocky\"\n\nThis reverts commit 3f3b15f7d1c31e8236623e276591f79f2e4fc058.\n\nChange-Id: I40463bc34006665c0ac8659f0bb445f47a4d747e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/125215c72c89a009f4c798bfa342b831c5e0094c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/125215c72c89a009f4c798bfa342b831c5e0094c"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
